)]}'
{"id":"openstack%2Frequirements~658183","triplet_id":"openstack%2Frequirements~master~I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf","project":"openstack/requirements","branch":"master","topic":"u-c","hashtags":[],"change_id":"I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf","subject":"Add upper-constraints enforcement for local reqs","status":"MERGED","created":"2019-05-09 18:50:54.000000000","updated":"2019-05-12 21:28:49.000000000","submitted":"2019-05-12 21:28:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658183-1557696529553-71816a1a","meta_rev_id":"c301777c62da18550b48e44210df89b2f1da4913","_number":658183,"virtual_id_number":658183,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-05-12 21:28:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-12 19:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-12 13:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-12 19:33:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 18:54:03.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-05-09 18:55:33.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-05-09 22:27:44.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-10 06:56:48.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-05-10 18:11:17.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-11 22:38:27.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-05-12 08:04:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-12 13:49:22.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-12 19:33:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-12 21:28:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac6a48935435721ae40cbbe450065ba3240e1c5e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 18:50:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f36ec9f581392863147a73a6269bbb5ae867e6e9","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-05-09 18:54:03.000000000","message":"Patch Set 1: Code-Review+1\n\nI think this makes sense to me.","accounts_in_message":[],"_revision_number":1},{"id":"8f19f1325090b327824d4fd8f93dd89b9261248c","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-09 18:55:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3974c0965d32195bfe1136e950cef458e9f43942","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 18:58:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2c77c931f1064ef2b27e8a0dcc4fc68b5034031","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 19:59:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24f163e9c283eb5f69d87f7074391f22023cc645","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 22:27:44.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a2a1973c1b2eabe4749d63da0fc1cf3ceeb7474","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 23:43:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/658183/2/check/openstack-tox-docs/7e392fa/html/ : SUCCESS in 3m 36s\n- openstack-tox-pypy http://logs.openstack.org/83/658183/2/check/openstack-tox-pypy/fa9660b/ : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/658183/2/check/openstack-tox-pep8/3fed5e2/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/83/658183/2/check/openstack-tox-py27/4d87ebc/ : SUCCESS in 2m 45s\n- openstack-tox-py36 http://logs.openstack.org/83/658183/2/check/openstack-tox-py36/d1a620e/ : SUCCESS in 3m 28s\n- openstack-tox-validate http://logs.openstack.org/83/658183/2/check/openstack-tox-validate/e3820e2/ : SUCCESS in 2m 38s\n- requirements-tox-babel http://logs.openstack.org/83/658183/2/check/requirements-tox-babel/aa6795a/ : SUCCESS in 4m 18s\n- requirements-integration http://logs.openstack.org/83/658183/2/check/requirements-integration/189e760/ : FAILURE in 30m 09s\n- neutron-grenade-multinode http://logs.openstack.org/83/658183/2/check/neutron-grenade-multinode/78f6ecd/ : SUCCESS in 1h 06m 31s\n- tempest-full http://logs.openstack.org/83/658183/2/check/tempest-full/6c6b57e/ : SUCCESS in 1h 50m 13s\n- neutron-grenade http://logs.openstack.org/83/658183/2/check/neutron-grenade/ec9c1a0/ : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":2},{"id":"6338a6c970b9d4720f271526eb5c1473b5a32766","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:10:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e4279e94f0dc192dc16158357fa94bb4a5bf075c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:53:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/658183/2/check/openstack-tox-docs/0d42930/html/ : SUCCESS in 4m 51s\n- openstack-tox-pypy http://logs.openstack.org/83/658183/2/check/openstack-tox-pypy/7ea7b60/ : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/658183/2/check/openstack-tox-pep8/9a4fdcd/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/83/658183/2/check/openstack-tox-py27/978c965/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/83/658183/2/check/openstack-tox-py36/de96574/ : SUCCESS in 4m 03s\n- openstack-tox-validate http://logs.openstack.org/83/658183/2/check/openstack-tox-validate/53dd5f3/ : SUCCESS in 4m 34s\n- requirements-tox-babel http://logs.openstack.org/83/658183/2/check/requirements-tox-babel/421add7/ : SUCCESS in 5m 12s\n- requirements-integration http://logs.openstack.org/83/658183/2/check/requirements-integration/4284ddc/ : FAILURE in 34m 18s\n- neutron-grenade-multinode http://logs.openstack.org/83/658183/2/check/neutron-grenade-multinode/0772e7f/ : SUCCESS in 1h 10m 13s\n- tempest-full http://logs.openstack.org/83/658183/2/check/tempest-full/9dd0437/ : SUCCESS in 1h 36m 19s\n- neutron-grenade http://logs.openstack.org/83/658183/2/check/neutron-grenade/3ddb180/ : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c28bf1e135116d464c00d8c356775ec0bf66a132","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-05-10 06:56:48.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4b50d1c4e626dbb03020a14f99d22b64fbb0c57d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 08:41:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/658183/2/check/openstack-tox-docs/f70c169/html/ : SUCCESS in 2m 44s\n- openstack-tox-pypy http://logs.openstack.org/83/658183/2/check/openstack-tox-pypy/351fe37/ : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/658183/2/check/openstack-tox-pep8/f36174f/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/83/658183/2/check/openstack-tox-py27/26dc5e8/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/83/658183/2/check/openstack-tox-py36/ca70580/ : SUCCESS in 2m 52s\n- openstack-tox-validate http://logs.openstack.org/83/658183/2/check/openstack-tox-validate/f9f005e/ : SUCCESS in 2m 51s\n- requirements-tox-babel http://logs.openstack.org/83/658183/2/check/requirements-tox-babel/4a2f5d5/ : SUCCESS in 3m 07s\n- requirements-integration http://logs.openstack.org/83/658183/2/check/requirements-integration/d4595b3/ : FAILURE in 33m 05s\n- neutron-grenade-multinode http://logs.openstack.org/83/658183/2/check/neutron-grenade-multinode/0c126b3/ : SUCCESS in 1h 10m 38s\n- tempest-full http://logs.openstack.org/83/658183/2/check/tempest-full/2fe88b4/ : SUCCESS in 1h 39m 13s\n- neutron-grenade http://logs.openstack.org/83/658183/2/check/neutron-grenade/9e3d37d/ : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":2},{"id":"7cb0d10096fd3307f955849fd660883527deb22d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 13:26:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b44e59af3af1e756e0841c09452b97fdb47dcd8a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 14:42:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7130010ec5ce34a6489b9c37c79555e5a69ed099","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-10 15:35:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22552299032b72e930d463b0c7a9981f3f64e315","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 15:38:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8f9a6e820d5b74bca0dc85e0e86fb0ca5388c75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:34:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/658183/3/check/openstack-tox-docs/c4f3f4b/html/ : SUCCESS in 2m 42s\n- openstack-tox-pypy http://logs.openstack.org/83/658183/3/check/openstack-tox-pypy/561498a/ : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/658183/3/check/openstack-tox-pep8/b5f7134/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/83/658183/3/check/openstack-tox-py27/322d9eb/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/83/658183/3/check/openstack-tox-py36/c99f930/ : SUCCESS in 5m 53s\n- openstack-tox-validate http://logs.openstack.org/83/658183/3/check/openstack-tox-validate/f4d61e5/ : SUCCESS in 2m 50s\n- requirements-tox-babel http://logs.openstack.org/83/658183/3/check/requirements-tox-babel/3ec091e/ : SUCCESS in 5m 08s\n- requirements-integration http://logs.openstack.org/83/658183/3/check/requirements-integration/55af4f7/ : FAILURE in 31m 11s\n- neutron-grenade-multinode http://logs.openstack.org/83/658183/3/check/neutron-grenade-multinode/bce417a/ : SUCCESS in 1h 06m 48s\n- tempest-full http://logs.openstack.org/83/658183/3/check/tempest-full/75f3ac3/ : SUCCESS in 1h 46m 42s\n- neutron-grenade http://logs.openstack.org/83/658183/3/check/neutron-grenade/d9e217c/ : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":3},{"id":"29b9e2321da98af5ddaaa392898eba391cdee7c8","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 18:11:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd931880844f12c07a40f116806d76c6e4a12139","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 18:20:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b6a703308f22e342af3d76923e4580c803011ab1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 20:41:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/658183/4/check/openstack-tox-docs/84f3898/html/ : SUCCESS in 3m 12s\n- openstack-tox-pypy http://logs.openstack.org/83/658183/4/check/openstack-tox-pypy/f76518b/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/658183/4/check/openstack-tox-pep8/a17ac09/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/83/658183/4/check/openstack-tox-py27/98a957e/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/83/658183/4/check/openstack-tox-py36/34c2242/ : SUCCESS in 3m 06s\n- openstack-tox-validate http://logs.openstack.org/83/658183/4/check/openstack-tox-validate/2090200/ : SUCCESS in 2m 34s\n- requirements-tox-babel http://logs.openstack.org/83/658183/4/check/requirements-tox-babel/b55342d/ : SUCCESS in 2m 53s\n- requirements-integration http://logs.openstack.org/83/658183/4/check/requirements-integration/eddffbe/ : FAILURE in 38m 35s\n- neutron-grenade-multinode http://logs.openstack.org/83/658183/4/check/neutron-grenade-multinode/e1e1646/ : FAILURE in 53m 13s\n- tempest-full http://logs.openstack.org/83/658183/4/check/tempest-full/0982e87/ : SUCCESS in 1h 41m 53s\n- neutron-grenade http://logs.openstack.org/83/658183/4/check/neutron-grenade/82a90dd/ : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":4},{"id":"87a4a14344174bce7709f9840d719ce35f22d184","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 19:37:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2a4e883ac7c693ca7341d9efc6622e7993b6cee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:25:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/83/658183/4/check/openstack-tox-docs/93c84af/html/ : SUCCESS in 3m 03s\n- openstack-tox-pypy http://logs.openstack.org/83/658183/4/check/openstack-tox-pypy/37a3db8/ : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/658183/4/check/openstack-tox-pep8/559f963/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/83/658183/4/check/openstack-tox-py27/9a6566d/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/83/658183/4/check/openstack-tox-py36/a23e4e6/ : SUCCESS in 3m 33s\n- openstack-tox-validate http://logs.openstack.org/83/658183/4/check/openstack-tox-validate/cfbdf15/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/83/658183/4/check/requirements-tox-babel/8adf69d/ : SUCCESS in 2m 51s\n- requirements-integration http://logs.openstack.org/83/658183/4/check/requirements-integration/d9a0b83/ : SUCCESS in 54m 44s\n- neutron-grenade-multinode http://logs.openstack.org/83/658183/4/check/neutron-grenade-multinode/b6f8ea9/ : SUCCESS in 1h 10m 10s\n- tempest-full http://logs.openstack.org/83/658183/4/check/tempest-full/3f4f7ca/ : SUCCESS in 1h 42m 59s\n- neutron-grenade http://logs.openstack.org/83/658183/4/check/neutron-grenade/0c4c179/ : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":4},{"id":"d9da1e30b8fdc20a20725927fecc771c3299e2c3","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-11 22:38:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eecc59b544b2b3125134189a80c4149127c84f9f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-12 08:04:15.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nNatal, please double check...","accounts_in_message":[],"_revision_number":4},{"id":"545d68b06f17aed86f41e66a8236f2d6b2e08808","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:07:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8920d1f237dd9a41bc3f580dfe68ebdaeb697a2f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:07:21.000000000","message":"Patch Set 4:\n\nAJ is correct, this is the URL we want to switch to.","accounts_in_message":[],"_revision_number":4},{"id":"3337550e5b9afb96d96d4ad2a9491e699d991c84","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:10:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nsee inline comment, I believe we need to be different and run against the local copy of the constraints file","accounts_in_message":[],"_revision_number":4},{"id":"10f8a4f6d7581d68a643516ad20dba33e609dcd0","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:31:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"84a185205e8815e8bcf5c405d95023b0ad3005e0","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:49:22.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7a3090d979a134721fa554221d04614897038ae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 15:05:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/83/658183/5/check/openstack-tox-docs/b3d3742/html/ : SUCCESS in 4m 12s\n- openstack-tox-pypy http://logs.openstack.org/83/658183/5/check/openstack-tox-pypy/c60d53d/ : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/658183/5/check/openstack-tox-pep8/433f8b5/ : SUCCESS in 2m 22s\n- openstack-tox-py27 http://logs.openstack.org/83/658183/5/check/openstack-tox-py27/7adbd54/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/83/658183/5/check/openstack-tox-py36/9897117/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/83/658183/5/check/openstack-tox-validate/375cdca/ : SUCCESS in 2m 30s\n- requirements-tox-babel http://logs.openstack.org/83/658183/5/check/requirements-tox-babel/6089aec/ : SUCCESS in 2m 39s\n- requirements-integration http://logs.openstack.org/83/658183/5/check/requirements-integration/09fac53/ : SUCCESS in 56m 53s\n- neutron-grenade-multinode http://logs.openstack.org/83/658183/5/check/neutron-grenade-multinode/efffbd2/ : SUCCESS in 1h 02m 24s\n- tempest-full http://logs.openstack.org/83/658183/5/check/tempest-full/38d75b6/ : SUCCESS in 1h 27m 39s\n- neutron-grenade http://logs.openstack.org/83/658183/5/check/neutron-grenade/e9e831c/ : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":5},{"id":"5c0148146b1aea6f9e22161c8fae9fb5f2e092ee","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 18:52:44.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit aa266cbc630bf853c0bfe9e3333fbf0e82157981","accounts_in_message":[],"_revision_number":5},{"id":"0c70951f7ecc92082208a432ab92c34fef0070dc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-12 19:33:15.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"dd625b79b6edb5f748bfbe15c69665f65a231a18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 19:33:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9c0154b5a71dea4a3b94d4b71453ffe24c0719b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:28:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/83/658183/5/gate/openstack-tox-docs/d61d049/html/ : SUCCESS in 2m 53s\n- openstack-tox-pep8 http://logs.openstack.org/83/658183/5/gate/openstack-tox-pep8/e4f6143/ : SUCCESS in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/83/658183/5/gate/openstack-tox-py27/24afd31/ : SUCCESS in 2m 33s\n- openstack-tox-py36 http://logs.openstack.org/83/658183/5/gate/openstack-tox-py36/1db9411/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/83/658183/5/gate/openstack-tox-validate/d3dd371/ : SUCCESS in 2m 51s\n- requirements-tox-babel http://logs.openstack.org/83/658183/5/gate/requirements-tox-babel/8c6d31a/ : SUCCESS in 2m 35s\n- requirements-integration http://logs.openstack.org/83/658183/5/gate/requirements-integration/7ace916/ : SUCCESS in 56m 28s\n- neutron-grenade-multinode http://logs.openstack.org/83/658183/5/gate/neutron-grenade-multinode/53d595b/ : SUCCESS in 55m 34s\n- tempest-full http://logs.openstack.org/83/658183/5/gate/tempest-full/6889dcc/ : SUCCESS in 1h 52m 43s\n- neutron-grenade http://logs.openstack.org/83/658183/5/gate/neutron-grenade/6335556/ : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":5},{"id":"43799a2d1620bd0ce52dfb8a368369765235caed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:28:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8bad9fe19d09eb2e49fd44b92c57ee5acf85b558","revisions":{"c51fc3991f4541ee38555b8e4947a6daa7c5f6ea":{"kind":"REWORK","_number":1,"created":"2019-05-09 18:50:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/83/658183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/658183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/658183/1"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c51fc3991f4541ee38555b8e4947a6daa7c5f6ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c51fc3991f4541ee38555b8e4947a6daa7c5f6ea"}]},"branch":"refs/heads/master"},"660ed0e3827e0df6d9f6989d5360febf5ec08e7e":{"kind":"REWORK","_number":2,"created":"2019-05-09 19:59:55.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/83/658183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/658183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/658183/2"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 19:59:13.000000000","tz":-300},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/660ed0e3827e0df6d9f6989d5360febf5ec08e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/660ed0e3827e0df6d9f6989d5360febf5ec08e7e"}]},"branch":"refs/heads/master"},"0cd70b2a5104ae367c40658c8067e7b79155eac4":{"kind":"REWORK","_number":3,"created":"2019-05-10 13:26:12.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/83/658183/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/658183/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/658183/3"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 13:18:19.000000000","tz":-300},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nUpdates the requirements-integration job to run using py36 since that\nnow fails with running under py27. This should get the job to pass until\nwe are able to rewrite it to not be a legacy job.\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd70b2a5104ae367c40658c8067e7b79155eac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd70b2a5104ae367c40658c8067e7b79155eac4"}]},"branch":"refs/heads/master"},"f30da41e1275165b50d4431f9b886b9c7b5bf5a3":{"kind":"REWORK","_number":4,"created":"2019-05-10 18:20:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/83/658183/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/658183/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/658183/4"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 18:20:53.000000000","tz":-300},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nUpdates the requirements-integration job to run using py36 since that\nnow fails with running under py27. This should get the job to pass until\nwe are able to rewrite it to not be a legacy job.\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f30da41e1275165b50d4431f9b886b9c7b5bf5a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f30da41e1275165b50d4431f9b886b9c7b5bf5a3"}]},"branch":"refs/heads/master"},"8bad9fe19d09eb2e49fd44b92c57ee5acf85b558":{"kind":"REWORK","_number":5,"created":"2019-05-12 13:31:48.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/83/658183/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/658183/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/658183/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/658183/5"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-12 13:31:10.000000000","tz":120},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nUpdates the requirements-integration job to run using py36 since that\nnow fails with running under py27. This should get the job to pass until\nwe are able to rewrite it to not be a legacy job.\n\nNOTE(dmllr): updated install_command to include local constraints file\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8bad9fe19d09eb2e49fd44b92c57ee5acf85b558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8bad9fe19d09eb2e49fd44b92c57ee5acf85b558"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
