)]}'
{"id":"openstack%2Frequirements~658188","triplet_id":"openstack%2Frequirements~master~Iac612b8bf8d02f58f4677f3e1104d99cacce5b23","project":"openstack/requirements","branch":"master","topic":"u-c","hashtags":[],"change_id":"Iac612b8bf8d02f58f4677f3e1104d99cacce5b23","subject":"Fix docs environment","status":"MERGED","created":"2019-05-09 19:06:06.000000000","updated":"2019-05-12 19:26:16.000000000","submitted":"2019-05-12 19:26:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658188-1557689176929-4a183efd","meta_rev_id":"63d236d8018a2c70722666ca5485ab55f357208a","_number":658188,"virtual_id_number":658188,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-05-11 19:38:56.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-05-12 19:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-11 21:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-12 13:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-05-12 08:05:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-05-11 22:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-12 13:11:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 19:23:04.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-11 21:24:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-11 22:39:56.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-05-12 08:05:59.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-12 13:11:39.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-12 19:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66922686ee4f6a4405bb29d2e02469660676d554","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 19:06:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9c9e25694b9909bcf743bdfd97138a10222fd7d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 19:23:04.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me\n\nThanks Sean","accounts_in_message":[],"_revision_number":1},{"id":"0bfec1cb63992b87d35bdfb3bfd34a62fa3fa965","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-09 19:45:22.000000000","message":"Patch Set 1: Code-Review+1\n\nLgtm weird that this wasn\u0027t a problem so far","accounts_in_message":[],"_revision_number":1},{"id":"34c98a61d15d0bfaf7b8ea4fa429ebf9bac30d47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 23:20:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/658188/1/check/openstack-tox-docs/8f487e7/html/ : SUCCESS in 2m 49s\n- openstack-tox-pypy http://logs.openstack.org/88/658188/1/check/openstack-tox-pypy/6269270/ : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/658188/1/check/openstack-tox-pep8/7bfc3f9/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/88/658188/1/check/openstack-tox-py27/21d697b/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/88/658188/1/check/openstack-tox-py36/4cd1616/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/88/658188/1/check/openstack-tox-validate/aff6338/ : SUCCESS in 4m 26s\n- requirements-tox-babel http://logs.openstack.org/88/658188/1/check/requirements-tox-babel/7ce5ac0/ : SUCCESS in 2m 31s\n- requirements-integration http://logs.openstack.org/88/658188/1/check/requirements-integration/e6fa448/ : FAILURE in 33m 59s\n- neutron-grenade-multinode http://logs.openstack.org/88/658188/1/check/neutron-grenade-multinode/6a7be88/ : SUCCESS in 1h 13m 06s\n- tempest-full http://logs.openstack.org/88/658188/1/check/tempest-full/6a81a1a/ : SUCCESS in 2h 01m 21s\n- neutron-grenade http://logs.openstack.org/88/658188/1/check/neutron-grenade/b86bb07/ : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":1},{"id":"96313750f2c9eb9cdf96e4342d5dc1a1e865615c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:10:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d85339bdbc749e4ffa64998408e7dc0e98665903","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:59:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/658188/1/check/openstack-tox-docs/9069dd8/html/ : SUCCESS in 4m 42s\n- openstack-tox-pypy http://logs.openstack.org/88/658188/1/check/openstack-tox-pypy/1609670/ : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/658188/1/check/openstack-tox-pep8/06b223f/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/88/658188/1/check/openstack-tox-py27/292a570/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/88/658188/1/check/openstack-tox-py36/24ef7b2/ : SUCCESS in 5m 03s\n- openstack-tox-validate http://logs.openstack.org/88/658188/1/check/openstack-tox-validate/f76d60c/ : SUCCESS in 3m 50s\n- requirements-tox-babel http://logs.openstack.org/88/658188/1/check/requirements-tox-babel/4e6ccb1/ : SUCCESS in 4m 18s\n- requirements-integration http://logs.openstack.org/88/658188/1/check/requirements-integration/799c82f/ : FAILURE in 32m 43s\n- neutron-grenade-multinode http://logs.openstack.org/88/658188/1/check/neutron-grenade-multinode/775cd12/ : SUCCESS in 1h 10m 20s\n- tempest-full http://logs.openstack.org/88/658188/1/check/tempest-full/59f1512/ : SUCCESS in 1h 40m 56s\n- neutron-grenade http://logs.openstack.org/88/658188/1/check/neutron-grenade/ccdd1bf/ : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d176decaf28d2058e2974d8f5f107353bc8b4686","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 14:25:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a26d7620642bcdc0b79ae35db9de0f3bec3f8b4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 14:39:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5f74daa030f260352f9c6096f919ee98f5c53ff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 14:40:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a223099b24c4384d4ac96b34f3bc2e12fb211b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 17:55:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/658188/2/check/openstack-tox-docs/f79b999/html/ : SUCCESS in 2m 33s\n- openstack-tox-pypy http://logs.openstack.org/88/658188/2/check/openstack-tox-pypy/6fc1c22/ : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/658188/2/check/openstack-tox-pep8/80bc3a7/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/88/658188/2/check/openstack-tox-py27/953bed9/ : SUCCESS in 2m 43s\n- openstack-tox-py36 http://logs.openstack.org/88/658188/2/check/openstack-tox-py36/0d359a1/ : SUCCESS in 4m 37s\n- openstack-tox-validate http://logs.openstack.org/88/658188/2/check/openstack-tox-validate/4058e0f/ : SUCCESS in 3m 27s\n- requirements-tox-babel http://logs.openstack.org/88/658188/2/check/requirements-tox-babel/e62e729/ : SUCCESS in 2m 35s\n- requirements-integration http://logs.openstack.org/88/658188/2/check/requirements-integration/53b0e24/ : FAILURE in 34m 19s\n- neutron-grenade-multinode http://logs.openstack.org/88/658188/2/check/neutron-grenade-multinode/84ba18a/ : SUCCESS in 1h 03m 11s\n- tempest-full http://logs.openstack.org/88/658188/2/check/tempest-full/c68ac0a/ : SUCCESS in 1h 59m 22s\n- neutron-grenade http://logs.openstack.org/88/658188/2/check/neutron-grenade/7b3c88a/ : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b827676aee9b2f543926f223fbd5ab059de7f697","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-11 18:37:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fcf2fc82f06bf827116fd5c2c7b1dc72a4f97dc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 18:44:32.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nNot sure what you mean Andreas. The doc requirements are being removed from test-requirements. There is nothing doc related in requirements.txt.","accounts_in_message":[],"_revision_number":2},{"id":"2d8d459001bb211b57ff4909143661de28981e37","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 18:45:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6abbc0e15056f6d226ad3c8268ff5b94ecf13f6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-11 18:45:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cca898711d1e7ca07c0fd0095db30237c701574","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 18:45:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03f4f866bfaf2d481b8cebf8c38b0b1d6ed2e966","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-11 18:46:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"168a1c5bc90f7f27f5b9debbafac1114440fd7d5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 18:47:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d05f95f7307b607a0bc2dd2ca8aa207c8f1c8800","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 19:33:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/658188/4/check/openstack-tox-docs/2170223/html/ : SUCCESS in 2m 50s\n- openstack-tox-pypy http://logs.openstack.org/88/658188/4/check/openstack-tox-pypy/e0b1d43/ : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/658188/4/check/openstack-tox-pep8/7911d0e/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/88/658188/4/check/openstack-tox-py27/9f66695/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/88/658188/4/check/openstack-tox-py36/89ec102/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/88/658188/4/check/openstack-tox-validate/be9ef48/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/88/658188/4/check/requirements-tox-babel/8b4228a/ : SUCCESS in 2m 55s\n- requirements-integration http://logs.openstack.org/88/658188/4/check/requirements-integration/6a4645b/ : FAILURE in 40m 41s\n- neutron-grenade-multinode http://logs.openstack.org/88/658188/4/check/neutron-grenade-multinode/3be423f/ : FAILURE in 30m 16s\n- tempest-full http://logs.openstack.org/88/658188/4/check/tempest-full/125e2c0/ : FAILURE in 44m 18s\n- neutron-grenade http://logs.openstack.org/88/658188/4/check/neutron-grenade/53260de/ : FAILURE in 27m 39s","accounts_in_message":[],"_revision_number":4},{"id":"4d6ef9a975ce0d1f1319fa09943b5c431bc04e6d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 19:38:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6e5021b9efeb004b67c93d5764b1308a632f8c98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:20:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/658188/4/check/openstack-tox-docs/8eb11ab/html/ : SUCCESS in 2m 35s\n- openstack-tox-pypy http://logs.openstack.org/88/658188/4/check/openstack-tox-pypy/81cddbd/ : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/658188/4/check/openstack-tox-pep8/a30a5c0/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/88/658188/4/check/openstack-tox-py27/79cb374/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/88/658188/4/check/openstack-tox-py36/2abdbd7/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/88/658188/4/check/openstack-tox-validate/f9f8c14/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/88/658188/4/check/requirements-tox-babel/0a0c87a/ : SUCCESS in 2m 51s\n- requirements-integration http://logs.openstack.org/88/658188/4/check/requirements-integration/46cebc4/ : SUCCESS in 53m 23s\n- neutron-grenade-multinode http://logs.openstack.org/88/658188/4/check/neutron-grenade-multinode/e7d24c6/ : SUCCESS in 57m 47s\n- tempest-full http://logs.openstack.org/88/658188/4/check/tempest-full/5dfd232/ : SUCCESS in 1h 36m 23s\n- neutron-grenade http://logs.openstack.org/88/658188/4/check/neutron-grenade/7dc8b85/ : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":4},{"id":"efee4c348c1a5a4eb347c45e4ccacd98d400bd39","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 21:24:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1188cdcba5350d2fd573511e253c7aa2aceea217","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-11 22:39:56.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a9207f8e02ea93ac0ae6830f0af50deb81f920bd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-12 08:05:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a138e29f8d4cdc5fcc8610eed1b9d11f58822921","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:11:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nWorthwhile cleanup imho.","accounts_in_message":[],"_revision_number":4},{"id":"c404e69c2eaaa624d846df8da8b8c46f95c42173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 13:11:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d1809fb0521ef86907b3c2a575a216ce2cc65df1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 15:29:00.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/658188/4/gate/openstack-tox-docs/76d398b/html/ : SUCCESS in 3m 00s\n- openstack-tox-pep8 http://logs.openstack.org/88/658188/4/gate/openstack-tox-pep8/7b972b7/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/88/658188/4/gate/openstack-tox-py27/531d21d/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/88/658188/4/gate/openstack-tox-py36/54671e2/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/88/658188/4/gate/openstack-tox-validate/79d4b81/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/88/658188/4/gate/requirements-tox-babel/5096c84/ : SUCCESS in 3m 05s\n- requirements-integration http://logs.openstack.org/88/658188/4/gate/requirements-integration/9d1ff18/ : SUCCESS in 53m 21s\n- neutron-grenade-multinode http://logs.openstack.org/88/658188/4/gate/neutron-grenade-multinode/2db2800/ : FAILURE in 30m 02s\n- tempest-full http://logs.openstack.org/88/658188/4/gate/tempest-full/1225040/ : SUCCESS in 1h 21m 01s\n- neutron-grenade http://logs.openstack.org/88/658188/4/gate/neutron-grenade/1dd73ab/ : SUCCESS in 55m 41s","accounts_in_message":[],"_revision_number":4},{"id":"cff3814e3c15338f79d9b26f1060bacbf7fddaf8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-12 15:31:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b5b6ce54f879756f3c9eb075db9d50cbcc3b9417","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:23:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/658188/4/check/openstack-tox-docs/1867263/html/ : SUCCESS in 3m 43s\n- openstack-tox-pypy http://logs.openstack.org/88/658188/4/check/openstack-tox-pypy/4733860/ : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/658188/4/check/openstack-tox-pep8/eb6053c/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/88/658188/4/check/openstack-tox-py27/910cf67/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/88/658188/4/check/openstack-tox-py36/690fa1f/ : SUCCESS in 5m 16s\n- openstack-tox-validate http://logs.openstack.org/88/658188/4/check/openstack-tox-validate/4da9d7e/ : SUCCESS in 4m 32s\n- requirements-tox-babel http://logs.openstack.org/88/658188/4/check/requirements-tox-babel/1f8617e/ : SUCCESS in 3m 40s\n- requirements-integration http://logs.openstack.org/88/658188/4/check/requirements-integration/6e928cf/ : SUCCESS in 59m 26s\n- neutron-grenade-multinode http://logs.openstack.org/88/658188/4/check/neutron-grenade-multinode/8976aaa/ : SUCCESS in 1h 07m 03s\n- tempest-full http://logs.openstack.org/88/658188/4/check/tempest-full/435385b/ : SUCCESS in 1h 50m 12s\n- neutron-grenade http://logs.openstack.org/88/658188/4/check/neutron-grenade/6af29ed/ : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":4},{"id":"bc10667a467c689afc2ccc350e9e8d341188dcaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:23:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c5c3209ee2a399319cad58330509070b297590a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 19:26:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"212ce1b418952a60e2641739eaa8cd9bf829ff47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 19:26:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/658188/4/gate/openstack-tox-docs/0ffa412/html/ : SUCCESS in 2m 20s\n- openstack-tox-pep8 http://logs.openstack.org/88/658188/4/gate/openstack-tox-pep8/7624db9/ : SUCCESS in 2m 21s\n- openstack-tox-py27 http://logs.openstack.org/88/658188/4/gate/openstack-tox-py27/9e07886/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/88/658188/4/gate/openstack-tox-py36/5dbabc9/ : SUCCESS in 3m 58s\n- openstack-tox-validate http://logs.openstack.org/88/658188/4/gate/openstack-tox-validate/35b653b/ : SUCCESS in 4m 01s\n- requirements-tox-babel http://logs.openstack.org/88/658188/4/gate/requirements-tox-babel/4919c04/ : SUCCESS in 3m 33s\n- requirements-integration http://logs.openstack.org/88/658188/4/gate/requirements-integration/3cdcb50/ : SUCCESS in 55m 26s\n- neutron-grenade-multinode http://logs.openstack.org/88/658188/4/gate/neutron-grenade-multinode/fb30902/ : SUCCESS in 1h 09m 54s\n- tempest-full http://logs.openstack.org/88/658188/4/gate/tempest-full/38a9209/ : SUCCESS in 1h 50m 18s\n- neutron-grenade http://logs.openstack.org/88/658188/4/gate/neutron-grenade/2941118/ : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5e75d5d0b484deae581d2c3fc23bf00e69bb82a2","revisions":{"8d73a527ba5a31142a118ea44b5bd850c5c68797":{"kind":"REWORK","_number":1,"created":"2019-05-09 19:06:06.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/88/658188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/658188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/658188/1"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 19:03:15.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 19:05:54.000000000","tz":-300},"subject":"Fix docs environment","message":"Fix docs environment\n\nChange-Id: Iac612b8bf8d02f58f4677f3e1104d99cacce5b23\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d73a527ba5a31142a118ea44b5bd850c5c68797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d73a527ba5a31142a118ea44b5bd850c5c68797"}]},"branch":"refs/heads/master"},"2c449e275b400626165b104e818efb2c26e16740":{"kind":"REWORK","_number":2,"created":"2019-05-10 14:39:25.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/88/658188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/658188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/658188/2"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 19:03:15.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 14:39:19.000000000","tz":-300},"subject":"Fix docs environment","message":"Fix docs environment\n\nChange-Id: Iac612b8bf8d02f58f4677f3e1104d99cacce5b23\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c449e275b400626165b104e818efb2c26e16740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c449e275b400626165b104e818efb2c26e16740"}]},"branch":"refs/heads/master"},"df292a25f74cab7d7442554da78f37980e13e1dc":{"kind":"REWORK","_number":3,"created":"2019-05-11 18:45:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/88/658188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/658188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/658188/3"}}},"commit":{"parents":[{"commit":"adda8f9d091ae5d0df4d2be407f4c5f001b745ce","subject":"Merge \"Cap keyring, scikit-learn, and sphinx on python2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adda8f9d091ae5d0df4d2be407f4c5f001b745ce"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 19:03:15.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-11 18:44:57.000000000","tz":-300},"subject":"Fix docs environment","message":"Fix docs environment\n\nChange-Id: Iac612b8bf8d02f58f4677f3e1104d99cacce5b23\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df292a25f74cab7d7442554da78f37980e13e1dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df292a25f74cab7d7442554da78f37980e13e1dc"}]},"branch":"refs/heads/master"},"5e75d5d0b484deae581d2c3fc23bf00e69bb82a2":{"kind":"REWORK","_number":4,"created":"2019-05-11 18:47:31.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/88/658188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/658188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/658188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/658188/4"}}},"commit":{"parents":[{"commit":"adda8f9d091ae5d0df4d2be407f4c5f001b745ce","subject":"Merge \"Cap keyring, scikit-learn, and sphinx on python2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adda8f9d091ae5d0df4d2be407f4c5f001b745ce"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 19:03:15.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-11 18:46:51.000000000","tz":-300},"subject":"Fix docs environment","message":"Fix docs environment\n\nChange-Id: Iac612b8bf8d02f58f4677f3e1104d99cacce5b23\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e75d5d0b484deae581d2c3fc23bf00e69bb82a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e75d5d0b484deae581d2c3fc23bf00e69bb82a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
