)]}'
{"id":"openstack%2Frequirements~658199","triplet_id":"openstack%2Frequirements~stable%2Fpike~I5ceb922ba16cb10a98570745117596619c423254","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I5ceb922ba16cb10a98570745117596619c423254","subject":"update constraint for python-glanceclient to new release 2.8.1","status":"MERGED","created":"2019-05-09 19:47:37.000000000","updated":"2019-05-14 04:43:37.000000000","submitted":"2019-05-14 04:43:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658199-1557809017543-a4b81bb3","meta_rev_id":"95870fd58e5746ca127b0f666c65e1743033a477","_number":658199,"virtual_id_number":658199,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-14 04:43:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-14 00:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-13 20:38:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-14 00:17:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 20:38:06.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-14 00:17:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-14 04:43:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"551995ff71389e40f0d7d78156147d8554ded5d3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 19:47:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95ee4ec5d39e3b8a0834fda021a5d8901a8ccfb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 23:23:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658199/1/check/openstack-tox-pypy/428a7b8/ : FAILURE in 2m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658199/1/check/openstack-tox-pep8/2b30964/ : FAILURE in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/99/658199/1/check/openstack-tox-py27/b936456/ : FAILURE in 3m 19s\n- openstack-tox-py35 http://logs.openstack.org/99/658199/1/check/openstack-tox-py35/49a1338/ : SUCCESS in 3m 50s\n- openstack-tox-validate http://logs.openstack.org/99/658199/1/check/openstack-tox-validate/5c1f2e6/ : FAILURE in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/99/658199/1/check/requirements-tox-babel/0038d93/ : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658199/1/check/requirements-tox-py27-check-uc/13bfa92/ : SUCCESS in 6m 20s\n- cross-cinder-py27 http://logs.openstack.org/99/658199/1/check/cross-cinder-py27/a997a6f/ : SUCCESS in 8m 01s\n- cross-cinder-py35 http://logs.openstack.org/99/658199/1/check/cross-cinder-py35/2d0b016/ : SUCCESS in 8m 05s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/99/658199/1/check/cross-glance-py27/a8a8e2b/ : SUCCESS in 9m 18s\n- cross-glance-py35 http://logs.openstack.org/99/658199/1/check/cross-glance-py35/2a357e5/ : SUCCESS in 8m 39s\n- cross-horizon-py27 http://logs.openstack.org/99/658199/1/check/cross-horizon-py27/1855e84/ : SUCCESS in 9m 30s\n- cross-horizon-py35 http://logs.openstack.org/99/658199/1/check/cross-horizon-py35/81fff73/ : SUCCESS in 10m 05s\n- cross-keystone-py27 http://logs.openstack.org/99/658199/1/check/cross-keystone-py27/b2a353e/ : SUCCESS in 11m 54s\n- cross-keystone-py35 http://logs.openstack.org/99/658199/1/check/cross-keystone-py35/feb6442/ : SUCCESS in 18m 07s\n- cross-neutron-py27 http://logs.openstack.org/99/658199/1/check/cross-neutron-py27/dcb48e8/ : SUCCESS in 22m 49s\n- cross-neutron-py35 http://logs.openstack.org/99/658199/1/check/cross-neutron-py35/fa5d12b/ : SUCCESS in 22m 52s\n- cross-nova-py27 http://logs.openstack.org/99/658199/1/check/cross-nova-py27/56710fd/ : SUCCESS in 11m 38s\n- cross-nova-py35 http://logs.openstack.org/99/658199/1/check/cross-nova-py35/0387682/ : SUCCESS in 13m 04s\n- cross-nova-functional http://logs.openstack.org/99/658199/1/check/cross-nova-functional/ed6e596/ : SUCCESS in 20m 08s\n- cross-swift-py27 http://logs.openstack.org/99/658199/1/check/cross-swift-py27/031be95/ : SUCCESS in 8m 37s\n- cross-swift-py35 http://logs.openstack.org/99/658199/1/check/cross-swift-py35/498ff96/ : SUCCESS in 3m 35s\n- requirements-integration http://logs.openstack.org/99/658199/1/check/requirements-integration/5ee7a49/ : FAILURE in 29m 20s\n- neutron-grenade-multinode http://logs.openstack.org/99/658199/1/check/neutron-grenade-multinode/024f331/ : SUCCESS in 1h 02m 49s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658199/1/check/legacy-tempest-dsvm-neutron-full/f1ef494/ : SUCCESS in 1h 36m 43s\n- neutron-grenade http://logs.openstack.org/99/658199/1/check/neutron-grenade/a493d10/ : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"522b2f1e6c3725b574a53d4d42c4985cae4d2fd0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:10:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89ed7ab5ecb9fc273504c910cf10c7b35f32a45d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:34:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658199/1/check/openstack-tox-pypy/a5ab8ab/ : FAILURE in 3m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658199/1/check/openstack-tox-pep8/59c78fb/ : FAILURE in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/99/658199/1/check/openstack-tox-py27/9162556/ : FAILURE in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/99/658199/1/check/openstack-tox-py35/409cd4b/ : SUCCESS in 3m 53s\n- openstack-tox-validate http://logs.openstack.org/99/658199/1/check/openstack-tox-validate/3112167/ : FAILURE in 5m 36s\n- requirements-tox-babel http://logs.openstack.org/99/658199/1/check/requirements-tox-babel/fe9a268/ : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658199/1/check/requirements-tox-py27-check-uc/992ef7a/ : SUCCESS in 7m 25s\n- cross-cinder-py27 http://logs.openstack.org/99/658199/1/check/cross-cinder-py27/250602f/ : SUCCESS in 8m 00s\n- cross-cinder-py35 http://logs.openstack.org/99/658199/1/check/cross-cinder-py35/ffd2bb5/ : SUCCESS in 9m 00s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/99/658199/1/check/cross-glance-py27/92d857d/ : SUCCESS in 8m 45s\n- cross-glance-py35 http://logs.openstack.org/99/658199/1/check/cross-glance-py35/fc5afde/ : SUCCESS in 10m 03s\n- cross-horizon-py27 http://logs.openstack.org/99/658199/1/check/cross-horizon-py27/342e303/ : SUCCESS in 12m 12s\n- cross-horizon-py35 http://logs.openstack.org/99/658199/1/check/cross-horizon-py35/5b34852/ : SUCCESS in 11m 10s\n- cross-keystone-py27 http://logs.openstack.org/99/658199/1/check/cross-keystone-py27/811e54b/ : SUCCESS in 12m 11s\n- cross-keystone-py35 http://logs.openstack.org/99/658199/1/check/cross-keystone-py35/892e104/ : SUCCESS in 12m 14s\n- cross-neutron-py27 http://logs.openstack.org/99/658199/1/check/cross-neutron-py27/031d33a/ : SUCCESS in 29m 35s\n- cross-neutron-py35 http://logs.openstack.org/99/658199/1/check/cross-neutron-py35/69fba34/ : SUCCESS in 24m 58s\n- cross-nova-py27 http://logs.openstack.org/99/658199/1/check/cross-nova-py27/b360a07/ : SUCCESS in 15m 02s\n- cross-nova-py35 http://logs.openstack.org/99/658199/1/check/cross-nova-py35/14a4541/ : SUCCESS in 13m 12s\n- cross-nova-functional http://logs.openstack.org/99/658199/1/check/cross-nova-functional/7a04060/ : SUCCESS in 21m 18s\n- cross-swift-py27 http://logs.openstack.org/99/658199/1/check/cross-swift-py27/8307678/ : SUCCESS in 8m 48s\n- cross-swift-py35 http://logs.openstack.org/99/658199/1/check/cross-swift-py35/9644cfc/ : SUCCESS in 5m 16s\n- requirements-integration http://logs.openstack.org/99/658199/1/check/requirements-integration/b5a9725/ : FAILURE in 32m 57s\n- neutron-grenade-multinode http://logs.openstack.org/99/658199/1/check/neutron-grenade-multinode/e3ba325/ : SUCCESS in 58m 27s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658199/1/check/legacy-tempest-dsvm-neutron-full/92b4c75/ : SUCCESS in 1h 18m 00s\n- neutron-grenade http://logs.openstack.org/99/658199/1/check/neutron-grenade/e602875/ : SUCCESS in 48m 31s","accounts_in_message":[],"_revision_number":1},{"id":"db95998261e7f0d698cc7d3f47fb3c6102a37294","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-12 02:32:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"042837404075a24477ec908ae8c79fb2949a8310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:59:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658199/1/check/openstack-tox-pypy/626ec22/ : FAILURE in 4m 00s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658199/1/check/openstack-tox-pep8/d63ef4c/ : FAILURE in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/99/658199/1/check/openstack-tox-py27/6bcd7f0/ : FAILURE in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/99/658199/1/check/openstack-tox-py35/960c240/ : SUCCESS in 7m 31s\n- openstack-tox-validate http://logs.openstack.org/99/658199/1/check/openstack-tox-validate/837fc14/ : FAILURE in 4m 04s\n- requirements-tox-babel http://logs.openstack.org/99/658199/1/check/requirements-tox-babel/585a60d/ : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658199/1/check/requirements-tox-py27-check-uc/3c29567/ : SUCCESS in 6m 12s\n- cross-cinder-py27 http://logs.openstack.org/99/658199/1/check/cross-cinder-py27/4b91ab2/ : SUCCESS in 9m 15s\n- cross-cinder-py35 http://logs.openstack.org/99/658199/1/check/cross-cinder-py35/da509df/ : SUCCESS in 7m 02s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/99/658199/1/check/cross-glance-py27/bfb26eb/ : SUCCESS in 9m 14s\n- cross-glance-py35 http://logs.openstack.org/99/658199/1/check/cross-glance-py35/e4cf8be/ : SUCCESS in 9m 35s\n- cross-horizon-py27 http://logs.openstack.org/99/658199/1/check/cross-horizon-py27/2faf11f/ : SUCCESS in 10m 28s\n- cross-horizon-py35 http://logs.openstack.org/99/658199/1/check/cross-horizon-py35/80acf16/ : SUCCESS in 10m 59s\n- cross-keystone-py27 http://logs.openstack.org/99/658199/1/check/cross-keystone-py27/af54400/ : SUCCESS in 13m 08s\n- cross-keystone-py35 http://logs.openstack.org/99/658199/1/check/cross-keystone-py35/3081dc7/ : SUCCESS in 16m 02s\n- cross-neutron-py27 http://logs.openstack.org/99/658199/1/check/cross-neutron-py27/a293988/ : SUCCESS in 25m 32s\n- cross-neutron-py35 http://logs.openstack.org/99/658199/1/check/cross-neutron-py35/8b5218c/ : SUCCESS in 29m 35s\n- cross-nova-py27 http://logs.openstack.org/99/658199/1/check/cross-nova-py27/0d4d4b8/ : SUCCESS in 12m 34s\n- cross-nova-py35 http://logs.openstack.org/99/658199/1/check/cross-nova-py35/65c62aa/ : SUCCESS in 12m 44s\n- cross-nova-functional http://logs.openstack.org/99/658199/1/check/cross-nova-functional/33b72a5/ : SUCCESS in 20m 39s\n- cross-swift-py27 http://logs.openstack.org/99/658199/1/check/cross-swift-py27/d189b53/ : SUCCESS in 13m 25s\n- cross-swift-py35 http://logs.openstack.org/99/658199/1/check/cross-swift-py35/e12f9d3/ : SUCCESS in 4m 05s\n- requirements-integration http://logs.openstack.org/99/658199/1/check/requirements-integration/d05c08e/ : FAILURE in 30m 47s\n- neutron-grenade-multinode http://logs.openstack.org/99/658199/1/check/neutron-grenade-multinode/5b7c4a4/ : SUCCESS in 1h 07m 50s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658199/1/check/legacy-tempest-dsvm-neutron-full/4d47d53/ : SUCCESS in 1h 23m 55s\n- neutron-grenade http://logs.openstack.org/99/658199/1/check/neutron-grenade/c3a0ef7/ : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3684d34c8f6545cb68b8a5cce6546696f92cf803","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 20:38:06.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b89c576d0e9e954c44a7851b5a331cca715dcc82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:02:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658199/1/check/openstack-tox-pypy/48fd8eb/ : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658199/1/check/openstack-tox-pep8/a9f5deb/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/99/658199/1/check/openstack-tox-py27/8a77b97/ : SUCCESS in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/99/658199/1/check/openstack-tox-py35/2385d4c/ : SUCCESS in 3m 51s\n- openstack-tox-validate http://logs.openstack.org/99/658199/1/check/openstack-tox-validate/fd169d7/ : SUCCESS in 3m 57s\n- requirements-tox-babel http://logs.openstack.org/99/658199/1/check/requirements-tox-babel/5ca4d9b/ : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658199/1/check/requirements-tox-py27-check-uc/84e9ce0/ : SUCCESS in 5m 52s\n- cross-cinder-py27 http://logs.openstack.org/99/658199/1/check/cross-cinder-py27/ca20a68/ : SUCCESS in 8m 06s\n- cross-cinder-py35 http://logs.openstack.org/99/658199/1/check/cross-cinder-py35/2fa42aa/ : SUCCESS in 7m 41s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/99/658199/1/check/cross-glance-py27/66a6b26/ : SUCCESS in 8m 55s\n- cross-glance-py35 http://logs.openstack.org/99/658199/1/check/cross-glance-py35/89d2cb4/ : SUCCESS in 9m 10s\n- cross-horizon-py27 http://logs.openstack.org/99/658199/1/check/cross-horizon-py27/2e8e892/ : SUCCESS in 10m 03s\n- cross-horizon-py35 http://logs.openstack.org/99/658199/1/check/cross-horizon-py35/313c6b1/ : SUCCESS in 10m 05s\n- cross-keystone-py27 http://logs.openstack.org/99/658199/1/check/cross-keystone-py27/aa70b61/ : SUCCESS in 13m 42s\n- cross-keystone-py35 http://logs.openstack.org/99/658199/1/check/cross-keystone-py35/f61929c/ : TIMED_OUT in 40m 45s\n- cross-neutron-py27 http://logs.openstack.org/99/658199/1/check/cross-neutron-py27/94432d2/ : SUCCESS in 22m 30s\n- cross-neutron-py35 http://logs.openstack.org/99/658199/1/check/cross-neutron-py35/5f05aec/ : SUCCESS in 19m 35s\n- cross-nova-py27 http://logs.openstack.org/99/658199/1/check/cross-nova-py27/f5ab086/ : SUCCESS in 12m 57s\n- cross-nova-py35 http://logs.openstack.org/99/658199/1/check/cross-nova-py35/a63ae35/ : SUCCESS in 17m 44s\n- cross-nova-functional http://logs.openstack.org/99/658199/1/check/cross-nova-functional/ec2e794/ : SUCCESS in 23m 56s\n- cross-swift-py27 http://logs.openstack.org/99/658199/1/check/cross-swift-py27/884a178/ : SUCCESS in 7m 08s\n- cross-swift-py35 http://logs.openstack.org/99/658199/1/check/cross-swift-py35/bb256cc/ : SUCCESS in 3m 01s\n- requirements-integration http://logs.openstack.org/99/658199/1/check/requirements-integration/2a678f8/ : SUCCESS in 52m 22s\n- neutron-grenade-multinode http://logs.openstack.org/99/658199/1/check/neutron-grenade-multinode/2e191b2/ : SUCCESS in 1h 04m 21s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658199/1/check/legacy-tempest-dsvm-neutron-full/2fd6af0/ : FAILURE in 1h 19m 30s\n- neutron-grenade http://logs.openstack.org/99/658199/1/check/neutron-grenade/46316dd/ : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a36b061ee6e4e57880e6069d5b1d44db022883c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 00:17:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1be1d08ce0be7bf0461649ff45e5cc87748ef95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:52:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658199/1/check/openstack-tox-pypy/ffff589/ : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658199/1/check/openstack-tox-pep8/5f5456d/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/99/658199/1/check/openstack-tox-py27/2dae4a8/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/99/658199/1/check/openstack-tox-py35/ea5e5f0/ : SUCCESS in 3m 51s\n- openstack-tox-validate http://logs.openstack.org/99/658199/1/check/openstack-tox-validate/2d7c344/ : SUCCESS in 3m 14s\n- requirements-tox-babel http://logs.openstack.org/99/658199/1/check/requirements-tox-babel/ef424d9/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658199/1/check/requirements-tox-py27-check-uc/b5a6c28/ : SUCCESS in 6m 07s\n- cross-cinder-py27 http://logs.openstack.org/99/658199/1/check/cross-cinder-py27/839ecab/ : SUCCESS in 7m 59s\n- cross-cinder-py35 http://logs.openstack.org/99/658199/1/check/cross-cinder-py35/0bcd214/ : SUCCESS in 7m 24s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/99/658199/1/check/cross-glance-py27/97419ee/ : SUCCESS in 9m 54s\n- cross-glance-py35 http://logs.openstack.org/99/658199/1/check/cross-glance-py35/0d13b82/ : SUCCESS in 10m 22s\n- cross-horizon-py27 http://logs.openstack.org/99/658199/1/check/cross-horizon-py27/73ca348/ : SUCCESS in 10m 59s\n- cross-horizon-py35 http://logs.openstack.org/99/658199/1/check/cross-horizon-py35/318c83f/ : SUCCESS in 9m 28s\n- cross-keystone-py27 http://logs.openstack.org/99/658199/1/check/cross-keystone-py27/cf0da9e/ : SUCCESS in 11m 51s\n- cross-keystone-py35 http://logs.openstack.org/99/658199/1/check/cross-keystone-py35/2338cb6/ : SUCCESS in 10m 57s\n- cross-neutron-py27 http://logs.openstack.org/99/658199/1/check/cross-neutron-py27/4a32615/ : SUCCESS in 21m 12s\n- cross-neutron-py35 http://logs.openstack.org/99/658199/1/check/cross-neutron-py35/66f7f17/ : SUCCESS in 22m 32s\n- cross-nova-py27 http://logs.openstack.org/99/658199/1/check/cross-nova-py27/792dc24/ : SUCCESS in 12m 13s\n- cross-nova-py35 http://logs.openstack.org/99/658199/1/check/cross-nova-py35/ae48194/ : SUCCESS in 11m 49s\n- cross-nova-functional http://logs.openstack.org/99/658199/1/check/cross-nova-functional/bca24ac/ : SUCCESS in 19m 36s\n- cross-swift-py27 http://logs.openstack.org/99/658199/1/check/cross-swift-py27/14699a0/ : SUCCESS in 7m 27s\n- cross-swift-py35 http://logs.openstack.org/99/658199/1/check/cross-swift-py35/ecde764/ : SUCCESS in 3m 15s\n- requirements-integration http://logs.openstack.org/99/658199/1/check/requirements-integration/ad965e8/ : SUCCESS in 51m 18s\n- neutron-grenade-multinode http://logs.openstack.org/99/658199/1/check/neutron-grenade-multinode/36ea746/ : SUCCESS in 59m 51s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658199/1/check/legacy-tempest-dsvm-neutron-full/348828d/ : FAILURE in 1h 25m 19s\n- neutron-grenade http://logs.openstack.org/99/658199/1/check/neutron-grenade/433561a/ : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7d58a1621caf1c5afb0dfd797bc722f2ec37b601","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 01:57:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"112fd1741fcabd9a624ab1f2ef90d338ca5b4468","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 03:18:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/99/658199/1/check/openstack-tox-pypy/d3b61c4/ : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658199/1/check/openstack-tox-pep8/7dabb35/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/99/658199/1/check/openstack-tox-py27/08a8dd6/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/99/658199/1/check/openstack-tox-py35/52afbb8/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/99/658199/1/check/openstack-tox-validate/148963e/ : SUCCESS in 3m 04s\n- requirements-tox-babel http://logs.openstack.org/99/658199/1/check/requirements-tox-babel/18c9e1e/ : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658199/1/check/requirements-tox-py27-check-uc/7dc6e84/ : SUCCESS in 5m 43s\n- cross-cinder-py27 http://logs.openstack.org/99/658199/1/check/cross-cinder-py27/6c6c1b5/ : SUCCESS in 7m 19s\n- cross-cinder-py35 http://logs.openstack.org/99/658199/1/check/cross-cinder-py35/c4e52dd/ : SUCCESS in 7m 47s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/99/658199/1/check/cross-glance-py27/a6a7436/ : SUCCESS in 9m 10s\n- cross-glance-py35 http://logs.openstack.org/99/658199/1/check/cross-glance-py35/d67f059/ : SUCCESS in 8m 13s\n- cross-horizon-py27 http://logs.openstack.org/99/658199/1/check/cross-horizon-py27/0572c13/ : SUCCESS in 10m 27s\n- cross-horizon-py35 http://logs.openstack.org/99/658199/1/check/cross-horizon-py35/fb978a2/ : SUCCESS in 8m 26s\n- cross-keystone-py27 http://logs.openstack.org/99/658199/1/check/cross-keystone-py27/b71f4a4/ : SUCCESS in 12m 04s\n- cross-keystone-py35 http://logs.openstack.org/99/658199/1/check/cross-keystone-py35/184bf0c/ : SUCCESS in 10m 32s\n- cross-neutron-py27 http://logs.openstack.org/99/658199/1/check/cross-neutron-py27/87b5923/ : SUCCESS in 22m 23s\n- cross-neutron-py35 http://logs.openstack.org/99/658199/1/check/cross-neutron-py35/9717d03/ : SUCCESS in 22m 57s\n- cross-nova-py27 http://logs.openstack.org/99/658199/1/check/cross-nova-py27/e44b8d4/ : SUCCESS in 15m 21s\n- cross-nova-py35 http://logs.openstack.org/99/658199/1/check/cross-nova-py35/ca99e0d/ : SUCCESS in 12m 17s\n- cross-nova-functional http://logs.openstack.org/99/658199/1/check/cross-nova-functional/f1a18e7/ : SUCCESS in 18m 22s\n- cross-swift-py27 http://logs.openstack.org/99/658199/1/check/cross-swift-py27/135a383/ : SUCCESS in 6m 47s\n- cross-swift-py35 http://logs.openstack.org/99/658199/1/check/cross-swift-py35/00e361b/ : SUCCESS in 3m 11s\n- requirements-integration http://logs.openstack.org/99/658199/1/check/requirements-integration/54b127f/ : SUCCESS in 49m 19s\n- neutron-grenade-multinode http://logs.openstack.org/99/658199/1/check/neutron-grenade-multinode/3720d38/ : SUCCESS in 58m 36s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658199/1/check/legacy-tempest-dsvm-neutron-full/309aaeb/ : SUCCESS in 1h 20m 05s\n- neutron-grenade http://logs.openstack.org/99/658199/1/check/neutron-grenade/2027ac8/ : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4d5cc47c7485e646cf824c68fb3d7b63e4994928","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 03:18:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"717261ca2e39cf1c101f3ad6cdc60d4f7fd5dd3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 04:43:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e14f1ee92ec91c5b61b3ba122b1819f49acec9af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 04:43:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/658199/1/gate/openstack-tox-pep8/bfb8053/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/99/658199/1/gate/openstack-tox-py27/f897dab/ : SUCCESS in 3m 20s\n- openstack-tox-py35 http://logs.openstack.org/99/658199/1/gate/openstack-tox-py35/6bb58f1/ : SUCCESS in 3m 40s\n- openstack-tox-validate http://logs.openstack.org/99/658199/1/gate/openstack-tox-validate/a20ed15/ : SUCCESS in 3m 12s\n- requirements-tox-babel http://logs.openstack.org/99/658199/1/gate/requirements-tox-babel/c6f3599/ : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658199/1/gate/requirements-tox-py27-check-uc/65815a4/ : SUCCESS in 5m 49s\n- cross-cinder-py27 http://logs.openstack.org/99/658199/1/gate/cross-cinder-py27/90d133b/ : SUCCESS in 9m 23s\n- cross-glance-py27 http://logs.openstack.org/99/658199/1/gate/cross-glance-py27/4f18cbf/ : SUCCESS in 8m 53s\n- cross-glance-py35 http://logs.openstack.org/99/658199/1/gate/cross-glance-py35/66ff7a5/ : SUCCESS in 9m 04s\n- cross-horizon-py27 http://logs.openstack.org/99/658199/1/gate/cross-horizon-py27/b0260b2/ : SUCCESS in 9m 25s\n- cross-horizon-py35 http://logs.openstack.org/99/658199/1/gate/cross-horizon-py35/545e037/ : SUCCESS in 8m 42s\n- cross-keystone-py27 http://logs.openstack.org/99/658199/1/gate/cross-keystone-py27/ca5cee8/ : SUCCESS in 10m 47s\n- cross-keystone-py35 http://logs.openstack.org/99/658199/1/gate/cross-keystone-py35/db73d62/ : SUCCESS in 19m 25s\n- cross-neutron-py27 http://logs.openstack.org/99/658199/1/gate/cross-neutron-py27/f59c115/ : SUCCESS in 20m 59s\n- cross-neutron-py35 http://logs.openstack.org/99/658199/1/gate/cross-neutron-py35/8d3a7db/ : SUCCESS in 21m 07s\n- cross-nova-py27 http://logs.openstack.org/99/658199/1/gate/cross-nova-py27/4f45cf1/ : SUCCESS in 14m 00s\n- cross-nova-py35 http://logs.openstack.org/99/658199/1/gate/cross-nova-py35/ede5c37/ : SUCCESS in 13m 03s\n- cross-nova-functional http://logs.openstack.org/99/658199/1/gate/cross-nova-functional/e09b21c/ : SUCCESS in 20m 26s\n- cross-swift-py27 http://logs.openstack.org/99/658199/1/gate/cross-swift-py27/f3d5a8f/ : SUCCESS in 7m 15s\n- cross-swift-py35 http://logs.openstack.org/99/658199/1/gate/cross-swift-py35/cd99ede/ : SUCCESS in 3m 07s\n- requirements-integration http://logs.openstack.org/99/658199/1/gate/requirements-integration/fa1453f/ : SUCCESS in 51m 50s\n- neutron-grenade-multinode http://logs.openstack.org/99/658199/1/gate/neutron-grenade-multinode/44a7b1d/ : SUCCESS in 1h 00m 06s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658199/1/gate/legacy-tempest-dsvm-neutron-full/1ab73ef/ : SUCCESS in 1h 20m 09s\n- neutron-grenade http://logs.openstack.org/99/658199/1/gate/neutron-grenade/800282d/ : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41100124993d13b12000863e0d008f1db65fbd58","revisions":{"41100124993d13b12000863e0d008f1db65fbd58":{"kind":"REWORK","_number":1,"created":"2019-05-09 19:47:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/658199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/658199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/658199/1"}}},"commit":{"parents":[{"commit":"3f49011b4145023021627b2738c328c09e4ec1fc","subject":"Merge \"update constraint for ironic-lib to new release 2.10.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f49011b4145023021627b2738c328c09e4ec1fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 19:47:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 19:47:36.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 2.8.1","message":"update constraint for python-glanceclient to new release 2.8.1\n\nChange-Id: I5ceb922ba16cb10a98570745117596619c423254\nmeta:version: 2.8.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I1f8f0e14828e560412947b69b9d76ed6c5a41e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41100124993d13b12000863e0d008f1db65fbd58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41100124993d13b12000863e0d008f1db65fbd58"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
