)]}'
{"id":"openstack%2Frequirements~658465","triplet_id":"openstack%2Frequirements~stable%2Fpike~I13fd7933b4754e92daab4e5843922030f614b249","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I13fd7933b4754e92daab4e5843922030f614b249","subject":"update constraint for neutron-lib to new release 1.9.2","status":"MERGED","created":"2019-05-10 20:09:32.000000000","updated":"2019-05-14 03:23:11.000000000","submitted":"2019-05-14 03:23:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658465-1557804191436-88673ecb","meta_rev_id":"cea85dc53db341dd2ac6fa97bbf166edecd1ad9b","_number":658465,"virtual_id_number":658465,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-14 03:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-14 00:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-13 20:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-14 00:20:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 20:37:47.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-14 00:20:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-14 03:23:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd8f7ef2fdf307e2930e991d6fd62124e4adf04b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-10 20:09:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55fb29e7026c031bb0d4b3278d84a8a749b1fc35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:31:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/65/658465/1/check/openstack-tox-pypy/024f211/ : FAILURE in 3m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/658465/1/check/openstack-tox-pep8/bb15bc9/ : FAILURE in 2m 32s\n- openstack-tox-py27 http://logs.openstack.org/65/658465/1/check/openstack-tox-py27/ca8932e/ : FAILURE in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/65/658465/1/check/openstack-tox-py35/f994fa8/ : SUCCESS in 5m 06s\n- openstack-tox-validate http://logs.openstack.org/65/658465/1/check/openstack-tox-validate/3ed4334/ : FAILURE in 3m 50s\n- requirements-tox-babel http://logs.openstack.org/65/658465/1/check/requirements-tox-babel/426b775/ : SUCCESS in 5m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/658465/1/check/requirements-tox-py27-check-uc/db09793/ : SUCCESS in 5m 51s\n- cross-cinder-py27 http://logs.openstack.org/65/658465/1/check/cross-cinder-py27/485fc3b/ : FAILURE in 8m 19s\n- cross-cinder-py35 http://logs.openstack.org/65/658465/1/check/cross-cinder-py35/4a597a3/ : SUCCESS in 8m 47s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/65/658465/1/check/cross-glance-py27/ca5e9f8/ : SUCCESS in 8m 51s\n- cross-glance-py35 http://logs.openstack.org/65/658465/1/check/cross-glance-py35/762d4b9/ : SUCCESS in 9m 44s\n- cross-horizon-py27 http://logs.openstack.org/65/658465/1/check/cross-horizon-py27/f8eb15c/ : SUCCESS in 11m 04s\n- cross-horizon-py35 http://logs.openstack.org/65/658465/1/check/cross-horizon-py35/f4a2f15/ : SUCCESS in 10m 12s\n- cross-keystone-py27 http://logs.openstack.org/65/658465/1/check/cross-keystone-py27/b7c1d49/ : SUCCESS in 10m 37s\n- cross-keystone-py35 http://logs.openstack.org/65/658465/1/check/cross-keystone-py35/cf0dc1d/ : SUCCESS in 11m 26s\n- cross-neutron-py27 http://logs.openstack.org/65/658465/1/check/cross-neutron-py27/e9e4650/ : SUCCESS in 23m 47s\n- cross-neutron-py35 http://logs.openstack.org/65/658465/1/check/cross-neutron-py35/3de0a67/ : SUCCESS in 22m 57s\n- cross-nova-py27 http://logs.openstack.org/65/658465/1/check/cross-nova-py27/3225a89/ : SUCCESS in 13m 43s\n- cross-nova-py35 http://logs.openstack.org/65/658465/1/check/cross-nova-py35/98d589a/ : SUCCESS in 13m 47s\n- cross-nova-functional http://logs.openstack.org/65/658465/1/check/cross-nova-functional/3feeb8b/ : SUCCESS in 26m 06s\n- cross-swift-py27 http://logs.openstack.org/65/658465/1/check/cross-swift-py27/4769152/ : SUCCESS in 7m 11s\n- cross-swift-py35 http://logs.openstack.org/65/658465/1/check/cross-swift-py35/275750d/ : SUCCESS in 3m 20s\n- requirements-integration http://logs.openstack.org/65/658465/1/check/requirements-integration/092f230/ : FAILURE in 21m 36s\n- neutron-grenade-multinode http://logs.openstack.org/65/658465/1/check/neutron-grenade-multinode/19af5b5/ : SUCCESS in 58m 06s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/65/658465/1/check/legacy-tempest-dsvm-neutron-full/0db854a/ : SUCCESS in 1h 18m 41s\n- neutron-grenade http://logs.openstack.org/65/658465/1/check/neutron-grenade/3e842f3/ : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9199abc1db9cd5d0c4226eee31b5c1c963227c85","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-12 02:32:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee1ebd3f83f9689b3ba376db8a52d757a20a9fa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 04:03:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/65/658465/1/check/openstack-tox-pypy/1f6e7d0/ : FAILURE in 3m 17s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/658465/1/check/openstack-tox-pep8/23dc9ee/ : FAILURE in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/65/658465/1/check/openstack-tox-py27/a9005c1/ : FAILURE in 2m 58s\n- openstack-tox-py35 http://logs.openstack.org/65/658465/1/check/openstack-tox-py35/2024d80/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/65/658465/1/check/openstack-tox-validate/4e11122/ : FAILURE in 4m 52s\n- requirements-tox-babel http://logs.openstack.org/65/658465/1/check/requirements-tox-babel/81cf059/ : SUCCESS in 6m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/658465/1/check/requirements-tox-py27-check-uc/91bc3df/ : SUCCESS in 7m 39s\n- cross-cinder-py27 http://logs.openstack.org/65/658465/1/check/cross-cinder-py27/918cbab/ : SUCCESS in 10m 20s\n- cross-cinder-py35 http://logs.openstack.org/65/658465/1/check/cross-cinder-py35/2d1be82/ : SUCCESS in 11m 02s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/65/658465/1/check/cross-glance-py27/f4759c0/ : SUCCESS in 9m 11s\n- cross-glance-py35 http://logs.openstack.org/65/658465/1/check/cross-glance-py35/c7147f4/ : SUCCESS in 9m 24s\n- cross-horizon-py27 http://logs.openstack.org/65/658465/1/check/cross-horizon-py27/edaafc3/ : SUCCESS in 10m 09s\n- cross-horizon-py35 http://logs.openstack.org/65/658465/1/check/cross-horizon-py35/9a7347f/ : SUCCESS in 10m 52s\n- cross-keystone-py27 http://logs.openstack.org/65/658465/1/check/cross-keystone-py27/d7160e4/ : SUCCESS in 16m 34s\n- cross-keystone-py35 http://logs.openstack.org/65/658465/1/check/cross-keystone-py35/e1cfc72/ : SUCCESS in 16m 31s\n- cross-neutron-py27 http://logs.openstack.org/65/658465/1/check/cross-neutron-py27/91bd827/ : SUCCESS in 22m 44s\n- cross-neutron-py35 http://logs.openstack.org/65/658465/1/check/cross-neutron-py35/e3c4d51/ : SUCCESS in 26m 21s\n- cross-nova-py27 http://logs.openstack.org/65/658465/1/check/cross-nova-py27/fdddce1/ : SUCCESS in 11m 45s\n- cross-nova-py35 http://logs.openstack.org/65/658465/1/check/cross-nova-py35/5a4f3ce/ : SUCCESS in 20m 11s\n- cross-nova-functional http://logs.openstack.org/65/658465/1/check/cross-nova-functional/df8e60a/ : SUCCESS in 19m 41s\n- cross-swift-py27 http://logs.openstack.org/65/658465/1/check/cross-swift-py27/c2fdda4/ : SUCCESS in 7m 57s\n- cross-swift-py35 http://logs.openstack.org/65/658465/1/check/cross-swift-py35/968ae30/ : SUCCESS in 3m 07s\n- requirements-integration http://logs.openstack.org/65/658465/1/check/requirements-integration/c6da458/ : FAILURE in 34m 20s\n- neutron-grenade-multinode http://logs.openstack.org/65/658465/1/check/neutron-grenade-multinode/20b8338/ : SUCCESS in 59m 34s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/65/658465/1/check/legacy-tempest-dsvm-neutron-full/c6a86cb/ : SUCCESS in 1h 30m 14s\n- neutron-grenade http://logs.openstack.org/65/658465/1/check/neutron-grenade/dc6c3a2/ : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f933c0d7ce8aa9d966507718ed796aba314c7ff3","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 20:37:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4448368f225f67547513abb4dfe21319179e6001","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 20:37:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab4c69a19592c5aae3c63710e2c07ddb072b1e52","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-13 20:38:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ebbc515dd6fb4ac2af870a2023f8ef88ea5dfe24","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 00:15:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13094201e0f2a3db7cc88bbe144ec3a38e8c6d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:17:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/65/658465/1/check/openstack-tox-pypy/e2ae010/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/658465/1/check/openstack-tox-pep8/e9938bb/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/65/658465/1/check/openstack-tox-py27/106efc6/ : SUCCESS in 3m 43s\n- openstack-tox-py35 http://logs.openstack.org/65/658465/1/check/openstack-tox-py35/ef34ff1/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/65/658465/1/check/openstack-tox-validate/527a524/ : SUCCESS in 3m 44s\n- requirements-tox-babel http://logs.openstack.org/65/658465/1/check/requirements-tox-babel/5d79a8a/ : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/658465/1/check/requirements-tox-py27-check-uc/3c77267/ : SUCCESS in 6m 03s\n- cross-cinder-py27 http://logs.openstack.org/65/658465/1/check/cross-cinder-py27/2a934d1/ : SUCCESS in 7m 56s\n- cross-cinder-py35 http://logs.openstack.org/65/658465/1/check/cross-cinder-py35/41157ec/ : SUCCESS in 7m 07s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/65/658465/1/check/cross-glance-py27/134721c/ : SUCCESS in 8m 27s\n- cross-glance-py35 http://logs.openstack.org/65/658465/1/check/cross-glance-py35/418757d/ : SUCCESS in 8m 50s\n- cross-horizon-py27 http://logs.openstack.org/65/658465/1/check/cross-horizon-py27/3b614c4/ : SUCCESS in 10m 25s\n- cross-horizon-py35 http://logs.openstack.org/65/658465/1/check/cross-horizon-py35/77fa928/ : SUCCESS in 10m 26s\n- cross-keystone-py27 http://logs.openstack.org/65/658465/1/check/cross-keystone-py27/03c7b14/ : SUCCESS in 13m 30s\n- cross-keystone-py35 http://logs.openstack.org/65/658465/1/check/cross-keystone-py35/c58f260/ : TIMED_OUT in 40m 47s\n- cross-neutron-py27 http://logs.openstack.org/65/658465/1/check/cross-neutron-py27/d0f48dc/ : SUCCESS in 22m 03s\n- cross-neutron-py35 http://logs.openstack.org/65/658465/1/check/cross-neutron-py35/5527138/ : SUCCESS in 23m 42s\n- cross-nova-py27 http://logs.openstack.org/65/658465/1/check/cross-nova-py27/478ab2a/ : SUCCESS in 12m 25s\n- cross-nova-py35 http://logs.openstack.org/65/658465/1/check/cross-nova-py35/64d4b2a/ : SUCCESS in 11m 58s\n- cross-nova-functional http://logs.openstack.org/65/658465/1/check/cross-nova-functional/051633c/ : SUCCESS in 18m 21s\n- cross-swift-py27 http://logs.openstack.org/65/658465/1/check/cross-swift-py27/66ce19f/ : SUCCESS in 8m 14s\n- cross-swift-py35 http://logs.openstack.org/65/658465/1/check/cross-swift-py35/791ee2a/ : SUCCESS in 3m 13s\n- requirements-integration http://logs.openstack.org/65/658465/1/check/requirements-integration/9b9eabe/ : SUCCESS in 1h 01m 30s\n- neutron-grenade-multinode http://logs.openstack.org/65/658465/1/check/neutron-grenade-multinode/84aa7e3/ : SUCCESS in 58m 00s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/65/658465/1/check/legacy-tempest-dsvm-neutron-full/6d41a3c/ : SUCCESS in 1h 33m 09s\n- neutron-grenade http://logs.openstack.org/65/658465/1/check/neutron-grenade/eabe694/ : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cbdcc67e9006daa72f24b82a3f23ce986738ab77","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 00:20:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b4b5ffbc47d42bbab086a1d22c163db12d52a2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:54:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/65/658465/1/check/openstack-tox-pypy/805154a/ : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/658465/1/check/openstack-tox-pep8/30908d8/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/65/658465/1/check/openstack-tox-py27/67ec286/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/65/658465/1/check/openstack-tox-py35/554a3aa/ : SUCCESS in 3m 34s\n- openstack-tox-validate http://logs.openstack.org/65/658465/1/check/openstack-tox-validate/9feb282/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/65/658465/1/check/requirements-tox-babel/63cf823/ : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/658465/1/check/requirements-tox-py27-check-uc/3a614f5/ : SUCCESS in 5m 10s\n- cross-cinder-py27 http://logs.openstack.org/65/658465/1/check/cross-cinder-py27/87f7153/ : SUCCESS in 7m 20s\n- cross-cinder-py35 http://logs.openstack.org/65/658465/1/check/cross-cinder-py35/36bd5f7/ : SUCCESS in 7m 41s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/65/658465/1/check/cross-glance-py27/9d2ec93/ : SUCCESS in 8m 13s\n- cross-glance-py35 http://logs.openstack.org/65/658465/1/check/cross-glance-py35/5994f43/ : SUCCESS in 9m 09s\n- cross-horizon-py27 http://logs.openstack.org/65/658465/1/check/cross-horizon-py27/2ccc621/ : SUCCESS in 10m 26s\n- cross-horizon-py35 http://logs.openstack.org/65/658465/1/check/cross-horizon-py35/94ad1ba/ : SUCCESS in 10m 41s\n- cross-keystone-py27 http://logs.openstack.org/65/658465/1/check/cross-keystone-py27/6cd2460/ : SUCCESS in 10m 12s\n- cross-keystone-py35 http://logs.openstack.org/65/658465/1/check/cross-keystone-py35/185e6ab/ : SUCCESS in 11m 00s\n- cross-neutron-py27 http://logs.openstack.org/65/658465/1/check/cross-neutron-py27/07209ce/ : SUCCESS in 21m 56s\n- cross-neutron-py35 http://logs.openstack.org/65/658465/1/check/cross-neutron-py35/64d8b0e/ : SUCCESS in 24m 35s\n- cross-nova-py27 http://logs.openstack.org/65/658465/1/check/cross-nova-py27/c88b75f/ : SUCCESS in 12m 31s\n- cross-nova-py35 http://logs.openstack.org/65/658465/1/check/cross-nova-py35/4729d97/ : SUCCESS in 15m 42s\n- cross-nova-functional http://logs.openstack.org/65/658465/1/check/cross-nova-functional/e0fd0fa/ : SUCCESS in 22m 27s\n- cross-swift-py27 http://logs.openstack.org/65/658465/1/check/cross-swift-py27/91eea99/ : SUCCESS in 7m 57s\n- cross-swift-py35 http://logs.openstack.org/65/658465/1/check/cross-swift-py35/ec01427/ : SUCCESS in 5m 19s\n- requirements-integration http://logs.openstack.org/65/658465/1/check/requirements-integration/ae979c2/ : SUCCESS in 45m 52s\n- neutron-grenade-multinode http://logs.openstack.org/65/658465/1/check/neutron-grenade-multinode/ee9f0e8/ : SUCCESS in 50m 20s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/65/658465/1/check/legacy-tempest-dsvm-neutron-full/1591ca6/ : SUCCESS in 1h 31m 10s\n- neutron-grenade http://logs.openstack.org/65/658465/1/check/neutron-grenade/232f376/ : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":1},{"id":"db0467d39ac87c1bd7505cac931bedae842dae67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:54:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20f049d9b3f45cf85753a8b83b4b1956afab29fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 03:23:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dc50f2e0dcc727f3683e3d1eadc7495033d80ff6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 03:23:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/658465/1/gate/openstack-tox-pep8/c85b72f/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/65/658465/1/gate/openstack-tox-py27/add836c/ : SUCCESS in 5m 20s\n- openstack-tox-py35 http://logs.openstack.org/65/658465/1/gate/openstack-tox-py35/97c7121/ : SUCCESS in 4m 21s\n- openstack-tox-validate http://logs.openstack.org/65/658465/1/gate/openstack-tox-validate/71451fc/ : SUCCESS in 3m 35s\n- requirements-tox-babel http://logs.openstack.org/65/658465/1/gate/requirements-tox-babel/0b5a6be/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/658465/1/gate/requirements-tox-py27-check-uc/d0f536c/ : SUCCESS in 5m 59s\n- cross-cinder-py27 http://logs.openstack.org/65/658465/1/gate/cross-cinder-py27/bcec500/ : SUCCESS in 7m 37s\n- cross-glance-py27 http://logs.openstack.org/65/658465/1/gate/cross-glance-py27/ed1e3d7/ : SUCCESS in 9m 39s\n- cross-glance-py35 http://logs.openstack.org/65/658465/1/gate/cross-glance-py35/849276a/ : SUCCESS in 10m 32s\n- cross-horizon-py27 http://logs.openstack.org/65/658465/1/gate/cross-horizon-py27/24eac56/ : SUCCESS in 10m 07s\n- cross-horizon-py35 http://logs.openstack.org/65/658465/1/gate/cross-horizon-py35/0a62a9d/ : SUCCESS in 11m 05s\n- cross-keystone-py27 http://logs.openstack.org/65/658465/1/gate/cross-keystone-py27/5b6a8b6/ : SUCCESS in 19m 01s\n- cross-keystone-py35 http://logs.openstack.org/65/658465/1/gate/cross-keystone-py35/28e4cd8/ : SUCCESS in 10m 47s\n- cross-neutron-py27 http://logs.openstack.org/65/658465/1/gate/cross-neutron-py27/8cf2208/ : SUCCESS in 21m 37s\n- cross-neutron-py35 http://logs.openstack.org/65/658465/1/gate/cross-neutron-py35/b366bbb/ : SUCCESS in 28m 53s\n- cross-nova-py27 http://logs.openstack.org/65/658465/1/gate/cross-nova-py27/93b5c9a/ : SUCCESS in 12m 17s\n- cross-nova-py35 http://logs.openstack.org/65/658465/1/gate/cross-nova-py35/9aa280a/ : SUCCESS in 14m 04s\n- cross-nova-functional http://logs.openstack.org/65/658465/1/gate/cross-nova-functional/f336127/ : SUCCESS in 18m 48s\n- cross-swift-py27 http://logs.openstack.org/65/658465/1/gate/cross-swift-py27/bed7c09/ : SUCCESS in 6m 26s\n- cross-swift-py35 http://logs.openstack.org/65/658465/1/gate/cross-swift-py35/21ae847/ : SUCCESS in 3m 00s\n- requirements-integration http://logs.openstack.org/65/658465/1/gate/requirements-integration/65835a8/ : SUCCESS in 54m 09s\n- neutron-grenade-multinode http://logs.openstack.org/65/658465/1/gate/neutron-grenade-multinode/240c848/ : SUCCESS in 53m 23s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/65/658465/1/gate/legacy-tempest-dsvm-neutron-full/982c045/ : SUCCESS in 1h 27m 12s\n- neutron-grenade http://logs.openstack.org/65/658465/1/gate/neutron-grenade/bb165ac/ : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8227d3f6416fea165e8af43904bfeb9ac52fd00a","revisions":{"8227d3f6416fea165e8af43904bfeb9ac52fd00a":{"kind":"REWORK","_number":1,"created":"2019-05-10 20:09:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/658465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/658465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/658465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/658465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/658465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/658465/1"}}},"commit":{"parents":[{"commit":"3f49011b4145023021627b2738c328c09e4ec1fc","subject":"Merge \"update constraint for ironic-lib to new release 2.10.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f49011b4145023021627b2738c328c09e4ec1fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-10 20:09:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-10 20:09:30.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 1.9.2","message":"update constraint for neutron-lib to new release 1.9.2\n\nChange-Id: I13fd7933b4754e92daab4e5843922030f614b249\nmeta:version: 1.9.2\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I4a12cf4068b2c13a1ef448dfc1083f52f38baaf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8227d3f6416fea165e8af43904bfeb9ac52fd00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8227d3f6416fea165e8af43904bfeb9ac52fd00a"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
