)]}'
{"id":"openstack%2Frequirements~658691","triplet_id":"openstack%2Frequirements~stable%2Fstein~I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf","project":"openstack/requirements","branch":"stable/stein","topic":"u-c-stable/stein","hashtags":[],"change_id":"I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf","subject":"Add upper-constraints enforcement for local reqs","status":"MERGED","created":"2019-05-12 18:52:44.000000000","updated":"2019-05-13 08:58:42.000000000","submitted":"2019-05-13 08:58:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658691-1557737922479-7f1a7d1e","meta_rev_id":"e4728593eeb9493a22550cd48d9159dae9fbddba","_number":658691,"virtual_id_number":658691,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-13 08:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-12 20:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-13 07:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-13 07:20:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 20:57:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-13 08:58:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eea4d6ba804f8c470565c5819b485c8773993d75","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 18:52:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b523bf792344b08614363b200014a416589f3a67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 20:29:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/91/658691/1/check/openstack-tox-pypy/044199d/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/658691/1/check/openstack-tox-pep8/9fc66a9/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/91/658691/1/check/openstack-tox-py27/4891009/ : SUCCESS in 3m 16s\n- openstack-tox-py35 http://logs.openstack.org/91/658691/1/check/openstack-tox-py35/2e9a60f/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/91/658691/1/check/openstack-tox-py36/0a44a54/ : SUCCESS in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/91/658691/1/check/openstack-tox-validate/270b43b/ : SUCCESS in 3m 06s\n- requirements-tox-babel http://logs.openstack.org/91/658691/1/check/requirements-tox-babel/51d3af5/ : SUCCESS in 3m 08s\n- requirements-integration http://logs.openstack.org/91/658691/1/check/requirements-integration/81cafc6/ : SUCCESS in 42m 34s\n- neutron-grenade-multinode http://logs.openstack.org/91/658691/1/check/neutron-grenade-multinode/27407e8/ : SUCCESS in 56m 58s\n- tempest-full http://logs.openstack.org/91/658691/1/check/tempest-full/5785e16/ : SUCCESS in 1h 35m 42s\n- neutron-grenade http://logs.openstack.org/91/658691/1/check/neutron-grenade/c2ad633/ : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7e60322fd64a5aa1481f1022237896f5bc3cb8b5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-12 20:57:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ab266e8d48915237faa8c7e391d8788f23fe3cf","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 07:20:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8737e5edd62aa4f2a0661301814947d681cfa052","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 07:20:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8d6c6fec395ac004ab43c65d7d4f03a696bbbfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 08:58:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3eae69a9038124c55169f02df325f3bd8a6b022d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 08:58:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/91/658691/1/gate/openstack-tox-pep8/0c6f198/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/91/658691/1/gate/openstack-tox-py27/421c7b1/ : SUCCESS in 2m 59s\n- openstack-tox-py35 http://logs.openstack.org/91/658691/1/gate/openstack-tox-py35/7122890/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/91/658691/1/gate/openstack-tox-py36/b902362/ : SUCCESS in 3m 10s\n- openstack-tox-validate http://logs.openstack.org/91/658691/1/gate/openstack-tox-validate/3b6d31d/ : SUCCESS in 2m 51s\n- requirements-tox-babel http://logs.openstack.org/91/658691/1/gate/requirements-tox-babel/f50edfd/ : SUCCESS in 2m 45s\n- requirements-integration http://logs.openstack.org/91/658691/1/gate/requirements-integration/9127fbe/ : SUCCESS in 50m 24s\n- neutron-grenade-multinode http://logs.openstack.org/91/658691/1/gate/neutron-grenade-multinode/8919ea0/ : SUCCESS in 1h 06m 42s\n- tempest-full http://logs.openstack.org/91/658691/1/gate/tempest-full/9edef68/ : SUCCESS in 1h 33m 14s\n- neutron-grenade http://logs.openstack.org/91/658691/1/gate/neutron-grenade/39cd8a4/ : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa266cbc630bf853c0bfe9e3333fbf0e82157981","revisions":{"aa266cbc630bf853c0bfe9e3333fbf0e82157981":{"kind":"REWORK","_number":1,"created":"2019-05-12 18:52:44.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/91/658691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/658691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/658691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/658691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/658691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/658691/1"}}},"commit":{"parents":[{"commit":"c59a15adbf9f6c0df161f74f0b38ba0d47fe2185","subject":"Cap networkx, keyring, scikit-learn, and sphinx on python2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c59a15adbf9f6c0df161f74f0b38ba0d47fe2185"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-12 18:52:44.000000000","tz":0},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nUpdates the requirements-integration job to run using py36 since that\nnow fails with running under py27. This should get the job to pass until\nwe are able to rewrite it to not be a legacy job.\n\nNOTE(dmllr): updated install_command to include local constraints file\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa266cbc630bf853c0bfe9e3333fbf0e82157981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa266cbc630bf853c0bfe9e3333fbf0e82157981"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
