)]}'
{"id":"openstack%2Frequirements~658692","triplet_id":"openstack%2Frequirements~stable%2Frocky~I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf","project":"openstack/requirements","branch":"stable/rocky","hashtags":[],"change_id":"I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf","subject":"Add upper-constraints enforcement for local reqs","status":"MERGED","created":"2019-05-12 18:53:48.000000000","updated":"2019-05-13 15:26:35.000000000","submitted":"2019-05-13 15:26:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658692-1557761195710-96febb6b","meta_rev_id":"4935cc29360024d093e04d46620e99fdfdc079f0","_number":658692,"virtual_id_number":658692,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-05-13 15:26:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-13 13:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-13 08:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-05-13 13:48:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-13 13:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 13:28:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-13 13:48:06.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-05-13 15:26:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3433bc200080fbd7ef682ffe33b103830d693d0","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 18:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bf5591620dc9af1f8b33fc260d4c09ef9c673d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 20:26:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/92/658692/1/check/openstack-tox-pypy/b67bd11/ : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/92/658692/1/check/openstack-tox-pep8/64a5fa0/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/92/658692/1/check/openstack-tox-py27/e8e50c9/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/92/658692/1/check/openstack-tox-py35/fb0c2a6/ : SUCCESS in 3m 36s\n- openstack-tox-validate http://logs.openstack.org/92/658692/1/check/openstack-tox-validate/af7c14c/ : SUCCESS in 3m 22s\n- requirements-tox-babel http://logs.openstack.org/92/658692/1/check/requirements-tox-babel/a6e66a7/ : SUCCESS in 2m 54s\n- openstack-tox-py36 http://logs.openstack.org/92/658692/1/check/openstack-tox-py36/7f0d695/ : SUCCESS in 3m 05s\n- requirements-integration http://logs.openstack.org/92/658692/1/check/requirements-integration/f317b4b/ : FAILURE in 28m 25s\n- neutron-grenade-multinode http://logs.openstack.org/92/658692/1/check/neutron-grenade-multinode/7248bb2/ : SUCCESS in 1h 06m 14s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/92/658692/1/check/legacy-tempest-dsvm-neutron-full/da94786/ : SUCCESS in 1h 28m 29s\n- neutron-grenade http://logs.openstack.org/92/658692/1/check/neutron-grenade/822cb84/ : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":1},{"id":"111f019e11c80aa8e532bdd505656d6462475b28","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 06:28:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40c1e262bcd85a6fac7e4513f6b3cd16f0d1976e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 08:03:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/92/658692/1/check/openstack-tox-pypy/34f7e22/ : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/92/658692/1/check/openstack-tox-pep8/40dee2d/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/92/658692/1/check/openstack-tox-py27/7ed909c/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/92/658692/1/check/openstack-tox-py35/3abbf9f/ : SUCCESS in 4m 12s\n- openstack-tox-validate http://logs.openstack.org/92/658692/1/check/openstack-tox-validate/7672e3c/ : SUCCESS in 3m 04s\n- requirements-tox-babel http://logs.openstack.org/92/658692/1/check/requirements-tox-babel/7b5ba45/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/92/658692/1/check/openstack-tox-py36/d1b1919/ : SUCCESS in 4m 54s\n- requirements-integration http://logs.openstack.org/92/658692/1/check/requirements-integration/b54e4c5/ : SUCCESS in 1h 04m 06s\n- neutron-grenade-multinode http://logs.openstack.org/92/658692/1/check/neutron-grenade-multinode/8cd215e/ : SUCCESS in 1h 08m 34s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/92/658692/1/check/legacy-tempest-dsvm-neutron-full/dc0f352/ : SUCCESS in 1h 33m 08s\n- neutron-grenade http://logs.openstack.org/92/658692/1/check/neutron-grenade/d925647/ : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":1},{"id":"25719c15dc3300e3b25fff88f1cd3844a602654b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 08:18:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa068127c1f5bc02eb6869fe6cb97536e2f93e7b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-13 13:28:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2810fffe7bcb1a44f4fc90207175d33f89f1ce2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 13:29:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"970ee725bbd4d632ee9fcf4574e63638c1e580f8","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-05-13 13:48:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"36cb395c1adcca4dabee7f9b8965376ac43e9db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 15:26:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4935cc29360024d093e04d46620e99fdfdc079f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 15:26:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/658692/1/gate/openstack-tox-pep8/1df0a5c/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/92/658692/1/gate/openstack-tox-py27/75efa25/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/92/658692/1/gate/openstack-tox-py35/c357b29/ : SUCCESS in 6m 57s\n- openstack-tox-validate http://logs.openstack.org/92/658692/1/gate/openstack-tox-validate/d231d16/ : SUCCESS in 7m 04s\n- requirements-tox-babel http://logs.openstack.org/92/658692/1/gate/requirements-tox-babel/a036ace/ : SUCCESS in 2m 48s\n- openstack-tox-py36 http://logs.openstack.org/92/658692/1/gate/openstack-tox-py36/3becc7f/ : SUCCESS in 7m 31s\n- requirements-integration http://logs.openstack.org/92/658692/1/gate/requirements-integration/2d397fb/ : SUCCESS in 1h 02m 39s\n- neutron-grenade-multinode http://logs.openstack.org/92/658692/1/gate/neutron-grenade-multinode/da3ed7e/ : SUCCESS in 1h 21m 54s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/92/658692/1/gate/legacy-tempest-dsvm-neutron-full/7151874/ : SUCCESS in 1h 19m 52s\n- neutron-grenade http://logs.openstack.org/92/658692/1/gate/neutron-grenade/4ba84cc/ : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab0875127bd23cf18c549440b9e0130d5162b73d","revisions":{"ab0875127bd23cf18c549440b9e0130d5162b73d":{"kind":"REWORK","_number":1,"created":"2019-05-12 18:53:48.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/92/658692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/658692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/658692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/658692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/658692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/658692/1"}}},"commit":{"parents":[{"commit":"964f79d543aae22babac1b93ffbd8694f4a64df7","subject":"update git.openstack in _DEFAULT_REQS_DIR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/964f79d543aae22babac1b93ffbd8694f4a64df7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-12 18:53:34.000000000","tz":120},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nUpdates the requirements-integration job to run using py36 since that\nnow fails with running under py27. This should get the job to pass until\nwe are able to rewrite it to not be a legacy job.\n\nNOTE(dmllr): updated install_command to include local constraints file\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 8bad9fe19d09eb2e49fd44b92c57ee5acf85b558)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab0875127bd23cf18c549440b9e0130d5162b73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab0875127bd23cf18c549440b9e0130d5162b73d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
