)]}'
{"id":"openstack%2Frequirements~658699","triplet_id":"openstack%2Frequirements~stable%2Fqueens~If558f184c959e4b63b56dec3ca1571d1034cfe5c","project":"openstack/requirements","branch":"stable/queens","topic":"py2-caps","hashtags":[],"change_id":"If558f184c959e4b63b56dec3ca1571d1034cfe5c","subject":"Cap keyring, scikit-learn, and sphinx on python2","status":"MERGED","created":"2019-05-12 20:19:53.000000000","updated":"2019-05-13 14:16:15.000000000","submitted":"2019-05-13 14:16:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658699-1557756975975-f41f2753","meta_rev_id":"3937f966636018a915f987cb8e2b04f916611b25","_number":658699,"virtual_id_number":658699,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-13 14:16:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-13 12:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-13 06:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-13 12:52:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 00:43:55.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-13 06:26:26.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-13 12:52:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-13 14:16:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe8479a69e1c1e2d0d539688eb47ee052a356769","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-12 20:19:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61d3459880ddcecba5f57493adec0d01114774ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:56:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658699/1/check/openstack-tox-pypy/d2acb6c/ : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658699/1/check/openstack-tox-pep8/afb396a/ : SUCCESS in 2m 21s\n- openstack-tox-py27 http://logs.openstack.org/99/658699/1/check/openstack-tox-py27/7c2789a/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/99/658699/1/check/openstack-tox-py35/938995c/ : SUCCESS in 3m 18s\n- openstack-tox-validate http://logs.openstack.org/99/658699/1/check/openstack-tox-validate/fb4e716/ : SUCCESS in 2m 39s\n- requirements-integration http://logs.openstack.org/99/658699/1/check/requirements-integration/74c522b/ : FAILURE in 28m 06s\n- neutron-grenade-multinode http://logs.openstack.org/99/658699/1/check/neutron-grenade-multinode/091a1ed/ : SUCCESS in 1h 06m 05s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658699/1/check/legacy-tempest-dsvm-neutron-full/f4e3795/ : SUCCESS in 1h 31m 27s\n- neutron-grenade http://logs.openstack.org/99/658699/1/check/neutron-grenade/5312764/ : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":1},{"id":"980cd98d0beeafd3285fa4dab559a3cb663893f6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-12 22:05:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d0b1434682fcf650bfb41ca2cddc4a9ace464ef9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 23:33:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658699/2/check/openstack-tox-pypy/b894c3a/ : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658699/2/check/openstack-tox-pep8/18fe43a/ : SUCCESS in 2m 33s\n- openstack-tox-py27 http://logs.openstack.org/99/658699/2/check/openstack-tox-py27/16e304b/ : SUCCESS in 2m 50s\n- openstack-tox-py35 http://logs.openstack.org/99/658699/2/check/openstack-tox-py35/ea541f9/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/99/658699/2/check/openstack-tox-validate/390d8be/ : SUCCESS in 3m 08s\n- requirements-integration http://logs.openstack.org/99/658699/2/check/requirements-integration/9ea324d/ : FAILURE in 30m 45s\n- neutron-grenade-multinode http://logs.openstack.org/99/658699/2/check/neutron-grenade-multinode/4260a44/ : SUCCESS in 1h 00m 17s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658699/2/check/legacy-tempest-dsvm-neutron-full/3672d78/ : SUCCESS in 1h 27m 44s\n- neutron-grenade http://logs.openstack.org/99/658699/2/check/neutron-grenade/c68c3dc/ : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5be9f52d0c75d4317229316b655d3f7de1c5ada8","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-13 00:43:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"404f81a56f2a5c65d124332bc17998623d82abda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 02:10:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658699/3/check/openstack-tox-pypy/9979fa6/ : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658699/3/check/openstack-tox-pep8/aa81c41/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/99/658699/3/check/openstack-tox-py27/dcabb7c/ : SUCCESS in 3m 07s\n- openstack-tox-py35 http://logs.openstack.org/99/658699/3/check/openstack-tox-py35/8ff8349/ : SUCCESS in 2m 57s\n- openstack-tox-validate http://logs.openstack.org/99/658699/3/check/openstack-tox-validate/9b80254/ : SUCCESS in 2m 40s\n- requirements-integration http://logs.openstack.org/99/658699/3/check/requirements-integration/51e7c25/ : FAILURE in 31m 30s\n- neutron-grenade-multinode http://logs.openstack.org/99/658699/3/check/neutron-grenade-multinode/72f9f57/ : SUCCESS in 59m 45s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658699/3/check/legacy-tempest-dsvm-neutron-full/b1141ff/ : SUCCESS in 1h 21m 10s\n- neutron-grenade http://logs.openstack.org/99/658699/3/check/neutron-grenade/c3edef7/ : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":3},{"id":"d0494848467f3b91e2fcf6a42fe563d342018d16","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 06:26:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7d20095a66efae3c2e587163788cadbba03a11de","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 06:26:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2d08c83b8cfbd6b5351579bdf43f97a2296d1a50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 08:00:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/99/658699/4/check/openstack-tox-pypy/86f26a5/ : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658699/4/check/openstack-tox-pep8/127f730/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/99/658699/4/check/openstack-tox-py27/e127cb8/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/99/658699/4/check/openstack-tox-py35/127647a/ : SUCCESS in 3m 18s\n- openstack-tox-validate http://logs.openstack.org/99/658699/4/check/openstack-tox-validate/905542d/ : SUCCESS in 3m 01s\n- requirements-integration http://logs.openstack.org/99/658699/4/check/requirements-integration/858dcf6/ : SUCCESS in 55m 39s\n- neutron-grenade-multinode http://logs.openstack.org/99/658699/4/check/neutron-grenade-multinode/2d56198/ : SUCCESS in 1h 02m 11s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658699/4/check/legacy-tempest-dsvm-neutron-full/635438a/ : SUCCESS in 1h 27m 51s\n- neutron-grenade http://logs.openstack.org/99/658699/4/check/neutron-grenade/4e3a698/ : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":4},{"id":"118acd52a192c14a49dca853a6c0b84e9dca39d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-13 12:52:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d9fc443c772230b5c1dbc0175f4c81623be2f5dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 12:53:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d4931629a5b417baf9e3b3b47bc442ceb6229aad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 14:16:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/658699/4/gate/openstack-tox-pep8/d02a6cb/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/99/658699/4/gate/openstack-tox-py27/c2fd0d0/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/99/658699/4/gate/openstack-tox-py35/00d92f4/ : SUCCESS in 4m 02s\n- openstack-tox-validate http://logs.openstack.org/99/658699/4/gate/openstack-tox-validate/d4575e9/ : SUCCESS in 4m 09s\n- requirements-integration http://logs.openstack.org/99/658699/4/gate/requirements-integration/208380f/ : SUCCESS in 52m 10s\n- neutron-grenade-multinode http://logs.openstack.org/99/658699/4/gate/neutron-grenade-multinode/baf461d/ : SUCCESS in 58m 30s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658699/4/gate/legacy-tempest-dsvm-neutron-full/2000dd3/ : SUCCESS in 1h 17m 01s\n- neutron-grenade http://logs.openstack.org/99/658699/4/gate/neutron-grenade/835c353/ : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":4},{"id":"bee06bcf940122ccb0964fc55b48394e33c1a65a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 14:16:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2282e3df88079d73863d3abe50864167d9fa6366","revisions":{"2eb5e70154bd6b561931635e978fd0589af82e82":{"kind":"REWORK","_number":1,"created":"2019-05-12 20:19:53.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/99/658699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/658699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/658699/1"}}},"commit":{"parents":[{"commit":"7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9","subject":"Bump pylxd to 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9"}]}],"author":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2019-04-12 14:51:41.000000000","tz":120},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-12 20:19:21.000000000","tz":-300},"subject":"Cap keyring, scikit-learn, and sphinx on python2","message":"Cap keyring, scikit-learn, and sphinx on python2\n\nkeyring drops py2 support so lets ensure our requirements wont allow us\nto install 19.X releases on py2.\n\nscikit-learn drops support for 3.4 or lower with release 0.21.0.\n\nSetting the minimum supported python for both to 3.4 to be consistent\nwith u-c. When we remove the 3.4 constraints we\u0027ll also need to fix g-r\n\nSimilarly sphinx 2.0.0 drop py27 support do express that in\nglobal-requirements.\n\nIncludes squash of commit:\nChange-Id: I0a9700926c9a0db93e782c853c33f1aaee3d4876\n(cherry picked from commit d78bf07af26b6ff485722838b1c11e43796a115b)\n\nChange-Id: If558f184c959e4b63b56dec3ca1571d1034cfe5c\n(cherry picked from commit 00b2bcf7d664b1526b4eefe157c33113206d6251)\n(cherry picked from commit c59a15adbf9f6c0df161f74f0b38ba0d47fe2185)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit e91a7887ba83f0906960188e3c1173f38e5760c8)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2eb5e70154bd6b561931635e978fd0589af82e82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2eb5e70154bd6b561931635e978fd0589af82e82"}]},"branch":"refs/heads/stable/queens"},"61fddef2f1eb774623b71e8cc27e17f1caa98e66":{"kind":"REWORK","_number":2,"created":"2019-05-12 22:05:03.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/99/658699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/658699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/658699/2"}}},"commit":{"parents":[{"commit":"7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9","subject":"Bump pylxd to 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9"}]}],"author":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2019-04-12 14:51:41.000000000","tz":120},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-12 22:04:52.000000000","tz":-300},"subject":"Cap keyring, scikit-learn, and sphinx on python2","message":"Cap keyring, scikit-learn, and sphinx on python2\n\nkeyring drops py2 support so lets ensure our requirements wont allow us\nto install 19.X releases on py2.\n\nscikit-learn drops support for 3.4 or lower with release 0.21.0.\n\nSetting the minimum supported python for both to 3.4 to be consistent\nwith u-c. When we remove the 3.4 constraints we\u0027ll also need to fix g-r\n\nSimilarly sphinx 2.0.0 drop py27 support do express that in\nglobal-requirements.\n\nIncludes squash of commit:\nChange-Id: I0a9700926c9a0db93e782c853c33f1aaee3d4876\n(cherry picked from commit d78bf07af26b6ff485722838b1c11e43796a115b)\n\nChange-Id: If558f184c959e4b63b56dec3ca1571d1034cfe5c\n(cherry picked from commit 00b2bcf7d664b1526b4eefe157c33113206d6251)\n(cherry picked from commit c59a15adbf9f6c0df161f74f0b38ba0d47fe2185)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit e91a7887ba83f0906960188e3c1173f38e5760c8)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61fddef2f1eb774623b71e8cc27e17f1caa98e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61fddef2f1eb774623b71e8cc27e17f1caa98e66"}]},"branch":"refs/heads/stable/queens"},"34b5c2e8de65dc729866f14af1a79e92656b0406":{"kind":"REWORK","_number":3,"created":"2019-05-13 00:43:55.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/99/658699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/658699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/658699/3"}}},"commit":{"parents":[{"commit":"7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9","subject":"Bump pylxd to 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9"}]}],"author":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2019-04-12 14:51:41.000000000","tz":120},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-13 00:42:07.000000000","tz":-240},"subject":"Cap keyring, scikit-learn, and sphinx on python2","message":"Cap keyring, scikit-learn, and sphinx on python2\n\nkeyring drops py2 support so lets ensure our requirements wont allow us\nto install 19.X releases on py2.\n\nscikit-learn drops support for 3.4 or lower with release 0.21.0.\n\nSetting the minimum supported python for both to 3.4 to be consistent\nwith u-c. When we remove the 3.4 constraints we\u0027ll also need to fix g-r\n\nSimilarly sphinx 2.0.0 drop py27 support do express that in\nglobal-requirements.\n\nNOTE(tonyb): The more modern branches allow networkx \u003e\u003d 2.0.  These\nversion are explictly blocked on queens so we no longer squash\nI0a9700926c9a0db93e782c853c33f1aaee3d4876\n\nChange-Id: If558f184c959e4b63b56dec3ca1571d1034cfe5c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit 00b2bcf7d664b1526b4eefe157c33113206d6251)\n(cherry picked from commit c59a15adbf9f6c0df161f74f0b38ba0d47fe2185)\n(cherry picked from commit e91a7887ba83f0906960188e3c1173f38e5760c8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34b5c2e8de65dc729866f14af1a79e92656b0406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34b5c2e8de65dc729866f14af1a79e92656b0406"}]},"branch":"refs/heads/stable/queens"},"2282e3df88079d73863d3abe50864167d9fa6366":{"kind":"REWORK","_number":4,"created":"2019-05-13 06:26:12.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/99/658699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/658699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/658699/4"}}},"commit":{"parents":[{"commit":"7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9","subject":"Bump pylxd to 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9"}]}],"author":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2019-04-12 14:51:41.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-13 06:25:50.000000000","tz":120},"subject":"Cap keyring, scikit-learn, and sphinx on python2","message":"Cap keyring, scikit-learn, and sphinx on python2\n\nkeyring drops py2 support so lets ensure our requirements wont allow us\nto install 19.X releases on py2.\n\nscikit-learn drops support for 3.4 or lower with release 0.21.0.\n\nSetting the minimum supported python for both to 3.4 to be consistent\nwith u-c. When we remove the 3.4 constraints we\u0027ll also need to fix g-r\n\nSimilarly sphinx 2.0.0 drop py27 support do express that in\nglobal-requirements.\n\nNOTE(tonyb): The more modern branches allow networkx \u003e\u003d 2.0.  These\nversion are explictly blocked on queens so we no longer squash\nI0a9700926c9a0db93e782c853c33f1aaee3d4876\n\nNOTE(dmllr): Fix typo.\n\nChange-Id: If558f184c959e4b63b56dec3ca1571d1034cfe5c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit 00b2bcf7d664b1526b4eefe157c33113206d6251)\n(cherry picked from commit c59a15adbf9f6c0df161f74f0b38ba0d47fe2185)\n(cherry picked from commit e91a7887ba83f0906960188e3c1173f38e5760c8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2282e3df88079d73863d3abe50864167d9fa6366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2282e3df88079d73863d3abe50864167d9fa6366"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
