)]}'
{"id":"openstack%2Frequirements~658728","triplet_id":"openstack%2Frequirements~stable%2Fpike~If558f184c959e4b63b56dec3ca1571d1034cfe5c","project":"openstack/requirements","branch":"stable/pike","hashtags":[],"change_id":"If558f184c959e4b63b56dec3ca1571d1034cfe5c","subject":"Cap keyring, scikit-learn, and sphinx on python2","status":"MERGED","created":"2019-05-13 06:37:06.000000000","updated":"2019-05-13 15:26:34.000000000","submitted":"2019-05-13 15:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658728-1557761194736-72f24b91","meta_rev_id":"4d6549f5f86a38693b3af73a6ba1f83839c486a3","_number":658728,"virtual_id_number":658728,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-05-13 15:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-13 11:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-13 13:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-13 13:48:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-13 13:26:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 11:01:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-13 13:26:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-13 13:48:28.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-05-13 15:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bce9aa71841d3240edb31c30c962f56706e072ff","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 06:37:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0165b0654630d43f129cdf7042036c0a9edd9c02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 08:08:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/28/658728/1/check/openstack-tox-pypy/0078b8b/ : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/658728/1/check/openstack-tox-pep8/4abc62f/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/28/658728/1/check/openstack-tox-py27/a8bccc8/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/28/658728/1/check/openstack-tox-py35/ae719fc/ : SUCCESS in 4m 40s\n- openstack-tox-validate http://logs.openstack.org/28/658728/1/check/openstack-tox-validate/1cfc286/ : SUCCESS in 3m 47s\n- requirements-integration http://logs.openstack.org/28/658728/1/check/requirements-integration/fc90801/ : SUCCESS in 51m 54s\n- neutron-grenade-multinode http://logs.openstack.org/28/658728/1/check/neutron-grenade-multinode/6179005/ : SUCCESS in 58m 53s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/28/658728/1/check/legacy-tempest-dsvm-neutron-full/eeb1b18/ : SUCCESS in 1h 28m 39s\n- neutron-grenade http://logs.openstack.org/28/658728/1/check/neutron-grenade/02ddc92/ : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":1},{"id":"693c312641ff033cde4e063aff581e97bb6cddcf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-13 11:01:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9aee73fed6ec245d9a441a8a9c86cf67cc4e702e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-13 13:26:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60372c1186cd6bf385a98157585f145cd3a055a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 13:26:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08054777ea4e8c6113d11fc3d342bad239e1115e","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-05-13 13:48:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd6c80f9db0420f0362cd869ce2cbddaeec5fe6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 15:26:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/28/658728/1/gate/openstack-tox-pep8/b996fbf/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/28/658728/1/gate/openstack-tox-py27/2396ea5/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/28/658728/1/gate/openstack-tox-py35/582ba23/ : SUCCESS in 4m 01s\n- openstack-tox-validate http://logs.openstack.org/28/658728/1/gate/openstack-tox-validate/59b0723/ : SUCCESS in 4m 22s\n- requirements-integration http://logs.openstack.org/28/658728/1/gate/requirements-integration/683d57a/ : SUCCESS in 1h 02m 34s\n- neutron-grenade-multinode http://logs.openstack.org/28/658728/1/gate/neutron-grenade-multinode/3c13fd9/ : SUCCESS in 1h 09m 46s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/28/658728/1/gate/legacy-tempest-dsvm-neutron-full/b8e51eb/ : SUCCESS in 1h 55m 00s\n- neutron-grenade http://logs.openstack.org/28/658728/1/gate/neutron-grenade/cb87cd4/ : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4d6549f5f86a38693b3af73a6ba1f83839c486a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 15:26:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6cc1abbd9a68854c5f9eb2fd1677845d597ce395","revisions":{"6cc1abbd9a68854c5f9eb2fd1677845d597ce395":{"kind":"REWORK","_number":1,"created":"2019-05-13 06:37:06.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/28/658728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/658728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/658728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/658728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/658728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/658728/1"}}},"commit":{"parents":[{"commit":"3f49011b4145023021627b2738c328c09e4ec1fc","subject":"Merge \"update constraint for ironic-lib to new release 2.10.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f49011b4145023021627b2738c328c09e4ec1fc"}]}],"author":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2019-04-12 14:51:41.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-13 06:36:28.000000000","tz":120},"subject":"Cap keyring, scikit-learn, and sphinx on python2","message":"Cap keyring, scikit-learn, and sphinx on python2\n\nkeyring drops py2 support so lets ensure our requirements wont allow us\nto install 19.X releases on py2.\n\nscikit-learn drops support for 3.4 or lower with release 0.21.0.\n\nSetting the minimum supported python for both to 3.4 to be consistent\nwith u-c. When we remove the 3.4 constraints we\u0027ll also need to fix g-r\n\nSimilarly sphinx 2.0.0 drop py27 support do express that in\nglobal-requirements.\n\nNOTE(tonyb): The more modern branches allow networkx \u003e\u003d 2.0.  These\nversion are explictly blocked on queens so we no longer squash\nI0a9700926c9a0db93e782c853c33f1aaee3d4876\n\nNOTE(dmllr): Fix typo.\n\nChange-Id: If558f184c959e4b63b56dec3ca1571d1034cfe5c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit 00b2bcf7d664b1526b4eefe157c33113206d6251)\n(cherry picked from commit c59a15adbf9f6c0df161f74f0b38ba0d47fe2185)\n(cherry picked from commit e91a7887ba83f0906960188e3c1173f38e5760c8)\n(cherry picked from commit 2282e3df88079d73863d3abe50864167d9fa6366)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cc1abbd9a68854c5f9eb2fd1677845d597ce395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cc1abbd9a68854c5f9eb2fd1677845d597ce395"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
