)]}'
{"id":"openstack%2Frequirements~658735","triplet_id":"openstack%2Frequirements~stable%2Focata~If394af569790b38106fd4cb18266ef0300703e87","project":"openstack/requirements","branch":"stable/ocata","topic":"bug/1718576","hashtags":[],"change_id":"If394af569790b38106fd4cb18266ef0300703e87","subject":"Avoid networkx 2.0","status":"ABANDONED","created":"2019-05-13 06:49:01.000000000","updated":"2019-05-24 23:17:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f89f2ae99a65fefa890f160772d1acaca08dc4d1","_number":658735,"virtual_id_number":658735,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-23 06:18:03.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":-1,"date":"2019-05-13 08:15:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-13 06:49:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 06:49:01.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-05-13 08:15:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-23 06:18:03.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"af76db77fdb6581f576c758fdf364355d47429d1","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 06:49:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce8211ae8e8f64220500e5ca26d7b360d3d12afe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 08:15:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/35/658735/1/check/openstack-tox-pypy/567be1e/ : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/35/658735/1/check/openstack-tox-pep8/bd495f0/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/35/658735/1/check/openstack-tox-py27/0a899e8/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/35/658735/1/check/openstack-tox-py35/07780d6/ : SUCCESS in 3m 33s\n- openstack-tox-validate http://logs.openstack.org/35/658735/1/check/openstack-tox-validate/da88254/ : SUCCESS in 2m 43s\n- requirements-integration http://logs.openstack.org/35/658735/1/check/requirements-integration/b21e74a/ : FAILURE in 26m 18s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/35/658735/1/check/legacy-tempest-dsvm-neutron-full/6721165/ : SUCCESS in 1h 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"35fa35ca2fa4997bccaf30b438c4f9f78263da86","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:18:03.000000000","message":"Patch Set 1:\n\nLooks like we can abandon this as it\u0027s been squashed into: https://review.opendev.org/660860","accounts_in_message":[],"_revision_number":1},{"id":"664945c1139320d3f2a3f652b0b8655cb38d233a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-24 23:17:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b1578bc9f00301e2160055fdf6e05e6d39ccae0","revisions":{"8b1578bc9f00301e2160055fdf6e05e6d39ccae0":{"kind":"REWORK","_number":1,"created":"2019-05-13 06:49:01.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/35/658735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/658735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/658735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/658735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/658735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/658735/1"}}},"commit":{"parents":[{"commit":"45b6c90d6a6cf66dbaa2b88437f323cd05ed68e1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45b6c90d6a6cf66dbaa2b88437f323cd05ed68e1"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2017-09-21 01:37:57.000000000","tz":600},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-13 06:48:45.000000000","tz":120},"subject":"Avoid networkx 2.0","message":"Avoid networkx 2.0\n\nnetworkx 2.0 has released but is not backwards compatible with 1.0.\nAvoid it for now.\n\nPartial-Bug: #1718576\nChange-Id: If394af569790b38106fd4cb18266ef0300703e87\n(cherry picked from commit 80393d7d673e2ed08a604310b0c242fd31fb4132)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b1578bc9f00301e2160055fdf6e05e6d39ccae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b1578bc9f00301e2160055fdf6e05e6d39ccae0"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
