)]}'
{"id":"openstack%2Frequirements~658748","triplet_id":"openstack%2Frequirements~stable%2Focata~I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf","project":"openstack/requirements","branch":"stable/ocata","hashtags":[],"change_id":"I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf","subject":"Add upper-constraints enforcement for local reqs","status":"MERGED","created":"2019-05-13 08:22:31.000000000","updated":"2019-05-29 06:17:57.000000000","submitted":"2019-05-29 06:17:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658748-1559110677535-1d743c1f","meta_rev_id":"d47a8098f299e8f2cfc9ff691137c77428ce366d","_number":658748,"virtual_id_number":658748,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-24 02:07:25.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-29 06:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-24 00:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-24 00:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-28 18:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 06:22:03.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-24 02:07:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-29 06:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37236f032d944893a31398616e1c8f8c600def61","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 08:22:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed990f5e7fe73e56f99a975a0f0c6b64020265b1","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 08:22:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9cc3e1751a2125cd36a63626a5f39709843c9283","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 09:49:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/48/658748/1/check/openstack-tox-pypy/3d61fa0/ : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/1/check/openstack-tox-pep8/dc7b8b7/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/1/check/openstack-tox-py27/a4a6501/ : SUCCESS in 3m 22s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/1/check/openstack-tox-py35/be387ab/ : SUCCESS in 3m 39s\n- openstack-tox-validate http://logs.openstack.org/48/658748/1/check/openstack-tox-validate/6740399/ : SUCCESS in 3m 15s\n- requirements-tox-babel http://logs.openstack.org/48/658748/1/check/requirements-tox-babel/a4bef77/ : SUCCESS in 3m 00s\n- requirements-integration http://logs.openstack.org/48/658748/1/check/requirements-integration/64e962d/ : FAILURE in 30m 00s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/1/check/legacy-tempest-dsvm-neutron-full/6ce164b/ : FAILURE in 1h 22m 03s","accounts_in_message":[],"_revision_number":1},{"id":"af7e87e78d55903978fc5c93dafe2bd681b5afb6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 01:14:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"083a4f0b56ff0f44730d947d8a382c01458d7d87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 02:57:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/48/658748/1/check/openstack-tox-pypy/d5a761b/ : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/1/check/openstack-tox-pep8/906d68d/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/1/check/openstack-tox-py27/79c512f/ : SUCCESS in 3m 15s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/1/check/openstack-tox-py35/7f0ca85/ : SUCCESS in 3m 20s\n- openstack-tox-validate http://logs.openstack.org/48/658748/1/check/openstack-tox-validate/f34d645/ : SUCCESS in 2m 53s\n- requirements-tox-babel http://logs.openstack.org/48/658748/1/check/requirements-tox-babel/a258c7f/ : SUCCESS in 2m 46s\n- requirements-integration http://logs.openstack.org/48/658748/1/check/requirements-integration/79d5d2f/ : FAILURE in 29m 20s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/1/check/legacy-tempest-dsvm-neutron-full/a0b617f/ : SUCCESS in 1h 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"99f7191207959d5dbaf5484c549ed7efc0322042","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:21:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"41bf1a05cb1a65df727e191538326456c7241a6f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:22:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0ef0265cb4f2097057f7282f50e72cd122c98421","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:49:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/48/658748/2/check/openstack-tox-pypy/e0e7cb9/ : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/2/check/openstack-tox-pep8/806d94b/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/2/check/openstack-tox-py27/bac0ebb/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/2/check/openstack-tox-py35/28438e1/ : SUCCESS in 3m 32s\n- openstack-tox-validate http://logs.openstack.org/48/658748/2/check/openstack-tox-validate/1ee93bf/ : SUCCESS in 3m 13s\n- requirements-tox-babel http://logs.openstack.org/48/658748/2/check/requirements-tox-babel/d7d5ad2/ : SUCCESS in 3m 13s\n- requirements-integration http://logs.openstack.org/48/658748/2/check/requirements-integration/f1131fd/ : FAILURE in 28m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/2/check/legacy-tempest-dsvm-neutron-full/ed0c219/ : SUCCESS in 1h 30m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f5c0f5961c08213a832b0141605d6a8f3bf83563","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-24 00:10:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8da067a9ca64b79700fc43584c813049afb78c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 01:25:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/48/658748/3/check/openstack-tox-pypy/9c4f8ee/ : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/3/check/openstack-tox-pep8/ee4177f/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/3/check/openstack-tox-py27/1a82e64/ : SUCCESS in 5m 44s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/3/check/openstack-tox-py35/0ad4cff/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/48/658748/3/check/openstack-tox-validate/af506f1/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/48/658748/3/check/requirements-tox-babel/5cd6c36/ : SUCCESS in 6m 43s\n- requirements-integration http://logs.openstack.org/48/658748/3/check/requirements-integration/cbd9472/ : SUCCESS in 54m 07s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/3/check/legacy-tempest-dsvm-neutron-full/08ffa9f/ : FAILURE in 1h 11m 44s","accounts_in_message":[],"_revision_number":3},{"id":"9dbb58eea654acdd231a0c1051126193f2103754","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 02:07:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"92b47b319b8d1c868dca2c40b5c54b711e0c6d6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 03:52:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/48/658748/3/check/openstack-tox-pypy/31dc15b/ : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/3/check/openstack-tox-pep8/e63a2a2/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/3/check/openstack-tox-py27/84f9aca/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/3/check/openstack-tox-py35/4a85fb2/ : SUCCESS in 4m 14s\n- openstack-tox-validate http://logs.openstack.org/48/658748/3/check/openstack-tox-validate/f9775f8/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/48/658748/3/check/requirements-tox-babel/4906cd2/ : SUCCESS in 3m 15s\n- requirements-integration http://logs.openstack.org/48/658748/3/check/requirements-integration/a543a82/ : SUCCESS in 45m 38s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/3/check/legacy-tempest-dsvm-neutron-full/2a475ab/ : FAILURE in 1h 25m 47s","accounts_in_message":[],"_revision_number":3},{"id":"aea679fe73966e522c8c2d246bccf1569179ff22","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-24 23:17:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0711b796d781f1f7f654a72cfb24132676e2cd52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 00:59:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/48/658748/3/check/openstack-tox-pypy/8a04a18/ : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/3/check/openstack-tox-pep8/f36c812/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/3/check/openstack-tox-py27/e550cbf/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/3/check/openstack-tox-py35/3601832/ : SUCCESS in 4m 59s\n- openstack-tox-validate http://logs.openstack.org/48/658748/3/check/openstack-tox-validate/20f384f/ : SUCCESS in 3m 02s\n- requirements-tox-babel http://logs.openstack.org/48/658748/3/check/requirements-tox-babel/c949dbc/ : SUCCESS in 4m 27s\n- requirements-integration http://logs.openstack.org/48/658748/3/check/requirements-integration/0f50145/ : SUCCESS in 39m 25s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/3/check/legacy-tempest-dsvm-neutron-full/8c576d2/ : SUCCESS in 1h 36m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1f1b203403780121de5c85f53945076b9f3c4361","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-28 18:43:49.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a0586b46a8e94ced21b3955ce2010554964592d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:44:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0485620981d03f0ed484d5d8320d282445cb65aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 01:22:59.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/3/gate/openstack-tox-pep8/cae3103/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/3/gate/openstack-tox-py27/abf4575/ : SUCCESS in 5m 36s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/3/gate/openstack-tox-py35/4f1a34d/ : SUCCESS in 7m 45s\n- openstack-tox-validate http://logs.openstack.org/48/658748/3/gate/openstack-tox-validate/d252061/ : SUCCESS in 3m 41s\n- requirements-tox-babel http://logs.openstack.org/48/658748/3/gate/requirements-tox-babel/67fca25/ : SUCCESS in 5m 43s\n- requirements-integration http://logs.openstack.org/48/658748/3/gate/requirements-integration/516926d/ : SUCCESS in 57m 08s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/3/gate/legacy-tempest-dsvm-neutron-full/80d3297/ : FAILURE in 1h 17m 14s","accounts_in_message":[],"_revision_number":3},{"id":"6eb35cb3831ebdc54d22ebe410675d35e054a5b2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-29 01:25:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ecabaaecbddcc340ec241fd4eb97c1408171d4cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 04:45:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/48/658748/3/check/openstack-tox-pypy/262801e/ : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/3/check/openstack-tox-pep8/ee41f3f/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/3/check/openstack-tox-py27/687fc02/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/3/check/openstack-tox-py35/887425b/ : SUCCESS in 3m 36s\n- openstack-tox-validate http://logs.openstack.org/48/658748/3/check/openstack-tox-validate/5f4cc9a/ : SUCCESS in 3m 31s\n- requirements-tox-babel http://logs.openstack.org/48/658748/3/check/requirements-tox-babel/494f496/ : SUCCESS in 3m 28s\n- requirements-integration http://logs.openstack.org/48/658748/3/check/requirements-integration/ecf3197/ : SUCCESS in 58m 27s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/3/check/legacy-tempest-dsvm-neutron-full/84c9dc5/ : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":3},{"id":"be1d50f94d3852404335962cc927bcb110b2691d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 04:45:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1b770c6b173381514aa66b09619fbbc40f133d27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 06:17:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/658748/3/gate/openstack-tox-pep8/14e545a/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/48/658748/3/gate/openstack-tox-py27/04b003e/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/48/658748/3/gate/openstack-tox-py35/40c9ba6/ : SUCCESS in 3m 15s\n- openstack-tox-validate http://logs.openstack.org/48/658748/3/gate/openstack-tox-validate/4884dca/ : SUCCESS in 2m 47s\n- requirements-tox-babel http://logs.openstack.org/48/658748/3/gate/requirements-tox-babel/14aa73c/ : SUCCESS in 3m 05s\n- requirements-integration http://logs.openstack.org/48/658748/3/gate/requirements-integration/bd2faf4/ : SUCCESS in 57m 57s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/658748/3/gate/legacy-tempest-dsvm-neutron-full/ec2c508/ : SUCCESS in 1h 27m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d47a8098f299e8f2cfc9ff691137c77428ce366d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 06:17:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a82dc23de95429aa0bd96f2e74b831d87bbaa908","revisions":{"c61592fd1c98b830defbd30d5791139b795c3394":{"kind":"REWORK","_number":1,"created":"2019-05-13 08:22:31.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/48/658748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/658748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/658748/1"}}},"commit":{"parents":[{"commit":"45b6c90d6a6cf66dbaa2b88437f323cd05ed68e1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45b6c90d6a6cf66dbaa2b88437f323cd05ed68e1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-13 08:22:21.000000000","tz":120},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nNOTE(dmllr): updated install_command to include local constraints file\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 8bad9fe19d09eb2e49fd44b92c57ee5acf85b558)\n(cherry picked from commit ab0875127bd23cf18c549440b9e0130d5162b73d)\n(cherry picked from commit f707d12d81e68a8a7c568a3bda0779b1db8839ba)\n(cherry picked from commit 31237956f2512035d992315c58caa9d1df2ca6d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c61592fd1c98b830defbd30d5791139b795c3394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c61592fd1c98b830defbd30d5791139b795c3394"}]},"branch":"refs/heads/stable/ocata"},"9016762201bf2e2d974421d84cb4440f94ccb01e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-23 06:21:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/48/658748/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/658748/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/658748/2"}}},"commit":{"parents":[{"commit":"2484a9266c9f434af04654020fbab64781f5ed74","subject":"Cap keyring and scikit-learn on python2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2484a9266c9f434af04654020fbab64781f5ed74"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-23 06:21:51.000000000","tz":0},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nNOTE(dmllr): updated install_command to include local constraints file\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 8bad9fe19d09eb2e49fd44b92c57ee5acf85b558)\n(cherry picked from commit ab0875127bd23cf18c549440b9e0130d5162b73d)\n(cherry picked from commit f707d12d81e68a8a7c568a3bda0779b1db8839ba)\n(cherry picked from commit 31237956f2512035d992315c58caa9d1df2ca6d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9016762201bf2e2d974421d84cb4440f94ccb01e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9016762201bf2e2d974421d84cb4440f94ccb01e"}]},"branch":"refs/heads/stable/ocata"},"a82dc23de95429aa0bd96f2e74b831d87bbaa908":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-24 00:10:59.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/48/658748/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/658748/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/658748/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/658748/3"}}},"commit":{"parents":[{"commit":"d0e1c0a6ffacb50cb00d1bdb62fd66d1b4ee5fdf","subject":"Cap keyring and scikit-learn on python2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0e1c0a6ffacb50cb00d1bdb62fd66d1b4ee5fdf"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 18:49:16.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-24 00:10:59.000000000","tz":0},"subject":"Add upper-constraints enforcement for local reqs","message":"Add upper-constraints enforcement for local reqs\n\nThis adds the use of upper constraints in our tox environments. Jobs\nlike openstack-tox-py27 have started failing due to trying to install\nlib versions that only support Python 3.\n\nThe \u0027generate\u0027 target keeps the unconstrained deps since we actually do\nneed the latest installed for that.\n\nNOTE(dmllr): updated install_command to include local constraints file\n\nChange-Id: I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 8bad9fe19d09eb2e49fd44b92c57ee5acf85b558)\n(cherry picked from commit ab0875127bd23cf18c549440b9e0130d5162b73d)\n(cherry picked from commit f707d12d81e68a8a7c568a3bda0779b1db8839ba)\n(cherry picked from commit 31237956f2512035d992315c58caa9d1df2ca6d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a82dc23de95429aa0bd96f2e74b831d87bbaa908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a82dc23de95429aa0bd96f2e74b831d87bbaa908"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
