)]}'
{"id":"openstack%2Frequirements~658767","triplet_id":"openstack%2Frequirements~master~Id1dbff3b54aa5acdd2454319c3fc25404059bca4","project":"openstack/requirements","branch":"master","topic":"update_constraints_bandit","hashtags":[],"change_id":"Id1dbff3b54aa5acdd2454319c3fc25404059bca4","subject":"Update constraint for bandit.","status":"ABANDONED","created":"2019-05-13 09:20:19.000000000","updated":"2019-05-15 17:48:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c25c85f7cbfbe8c99a2b6758a35721e1eb82877","_number":658767,"virtual_id_number":658767,"owner":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2019-05-15 11:43:46.000000000","_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2019-05-13 10:53:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-05-13 09:51:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2019-05-13 09:25:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-05-15 15:49:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 09:25:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-13 09:51:12.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-05-13 10:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-15 15:49:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9ac469ae4d80fe7057d56db2b4d16755ce212ed4","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-13 09:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"615dd87dcd19a2572a915c5aec7692edd94e4810","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-13 09:25:52.000000000","message":"Patch Set 1: Code-Review+1\n\nExample of possible issues related to this \u003d\u003e https://review.opendev.org/#/c/658319/","accounts_in_message":[],"_revision_number":1},{"id":"6781d8198d964ada8724a4f47737661c7a51abe9","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2019-05-13 09:51:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f65a156c39bc6cc4b03949669e5365325a6385fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 10:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/658767/1/check/openstack-tox-docs/8ee041c/html/ : SUCCESS in 2m 39s\n- openstack-tox-pypy http://logs.openstack.org/67/658767/1/check/openstack-tox-pypy/81814a8/ : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/658767/1/check/openstack-tox-pep8/7dddb5a/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/67/658767/1/check/openstack-tox-py27/cf7196b/ : SUCCESS in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/67/658767/1/check/openstack-tox-py36/da335ba/ : SUCCESS in 3m 31s\n- openstack-tox-validate http://logs.openstack.org/67/658767/1/check/openstack-tox-validate/9d209a2/ : SUCCESS in 2m 57s\n- requirements-integration http://logs.openstack.org/67/658767/1/check/requirements-integration/82c0b48/ : SUCCESS in 1h 20m 15s\n- neutron-grenade-multinode http://logs.openstack.org/67/658767/1/check/neutron-grenade-multinode/2dcea37/ : SUCCESS in 58m 08s\n- tempest-full http://logs.openstack.org/67/658767/1/check/tempest-full/3d41241/ : SUCCESS in 1h 31m 31s\n- neutron-grenade http://logs.openstack.org/67/658767/1/check/neutron-grenade/3ac2a9c/ : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":1},{"id":"01482d28141973498e06954db52dbede1938d82c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-13 14:49:44.000000000","message":"Patch Set 1:\n\nI\u0027ve pasted the list of projects that I found using -x into the email thread, it only checked tox though.  I do think that we should move forward rather than not using this.","accounts_in_message":[],"_revision_number":1},{"id":"1425eb514e48dbe89055f173141bf7fdfbd85d95","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-15 11:43:46.000000000","message":"Patch Set 1:\n\nWait the next version.","accounts_in_message":[],"_revision_number":1},{"id":"f3fefcc923777a52ed804df89e15d99e91ffe0da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-15 15:49:17.000000000","message":"Patch Set 1: Workflow-1\n\nwaiting","accounts_in_message":[],"_revision_number":1},{"id":"1e5534e0b2662449b1ca8ea2dd36d5a8b106a7c8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-15 16:21:52.000000000","message":"Patch Set 1:\n\nthis only updates the test file as well, may want to fix that","accounts_in_message":[],"_revision_number":1},{"id":"d54df275e02619912cd22fc7d1e7f390f1b3150b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-15 16:23:57.000000000","message":"Patch Set 1:\n\nactually, this is in blacklist.txt so we don\u0027t track or control this in requirements, I\u0027d suggest abandoning this","accounts_in_message":[],"_revision_number":1},{"id":"72f25ebd4b7f9a89125c0e06760c475a0f8bb270","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-15 17:48:31.000000000","message":"Abandoned\n\nThe team disagree.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e673c1382ea7d02c88b7eaaf98c3aa55c008269c","revisions":{"e673c1382ea7d02c88b7eaaf98c3aa55c008269c":{"kind":"REWORK","_number":1,"created":"2019-05-13 09:20:19.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/67/658767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/658767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/658767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/658767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/658767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/658767/1"}}},"commit":{"parents":[{"commit":"6723ed02684772181deab850831ac8a20c6d565e","subject":"Merge \"update constraint for os-service-types to new release 1.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6723ed02684772181deab850831ac8a20c6d565e"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-13 09:20:01.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-13 09:20:01.000000000","tz":120},"subject":"Update constraint for bandit.","message":"Update constraint for bandit.\n\nUpdate the constraint for bandit to new release 1.6.0. This\nrelease change the behavior of the -x options so that it now supports\nglob patterns.\n\nChange-Id: Id1dbff3b54aa5acdd2454319c3fc25404059bca4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e673c1382ea7d02c88b7eaaf98c3aa55c008269c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e673c1382ea7d02c88b7eaaf98c3aa55c008269c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
