)]}'
{"id":"openstack%2Frequirements~659087","triplet_id":"openstack%2Frequirements~stable%2Fpike~I64fa2e2e1149798aa80624d7be6ab48600e297de","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I64fa2e2e1149798aa80624d7be6ab48600e297de","subject":"update constraint for os-apply-config to new release 7.2.4","status":"MERGED","created":"2019-05-14 14:49:52.000000000","updated":"2019-05-14 22:34:15.000000000","submitted":"2019-05-14 22:34:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659087-1557873255424-6085ff31","meta_rev_id":"042d59d9dca02baf54ebc044ad0c2a94f0e6be37","_number":659087,"virtual_id_number":659087,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-14 22:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-14 17:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-14 19:50:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-14 19:50:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 17:31:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-14 19:50:33.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-14 22:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"472598410564ea076c3b92c301f9c1e9ba50a2b8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-14 14:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"306b2e32b6399d474439a53ef27ee858f7eac4d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:31:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/87/659087/1/check/openstack-tox-pypy/7f1b60c/ : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/659087/1/check/openstack-tox-pep8/3b3e141/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/87/659087/1/check/openstack-tox-py27/f7f7e39/ : SUCCESS in 3m 52s\n- openstack-tox-py35 http://logs.openstack.org/87/659087/1/check/openstack-tox-py35/ccb607c/ : SUCCESS in 3m 53s\n- openstack-tox-validate http://logs.openstack.org/87/659087/1/check/openstack-tox-validate/13aa3a7/ : SUCCESS in 3m 45s\n- requirements-tox-babel http://logs.openstack.org/87/659087/1/check/requirements-tox-babel/4492b52/ : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/659087/1/check/requirements-tox-py27-check-uc/0e9cb27/ : SUCCESS in 6m 43s\n- cross-cinder-py27 http://logs.openstack.org/87/659087/1/check/cross-cinder-py27/fb25726/ : SUCCESS in 7m 38s\n- cross-cinder-py35 http://logs.openstack.org/87/659087/1/check/cross-cinder-py35/948f110/ : SUCCESS in 8m 09s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/87/659087/1/check/cross-glance-py27/9abe727/ : SUCCESS in 9m 14s\n- cross-glance-py35 http://logs.openstack.org/87/659087/1/check/cross-glance-py35/6900c30/ : SUCCESS in 9m 02s\n- cross-horizon-py27 http://logs.openstack.org/87/659087/1/check/cross-horizon-py27/fd2e139/ : SUCCESS in 10m 30s\n- cross-horizon-py35 http://logs.openstack.org/87/659087/1/check/cross-horizon-py35/298892e/ : SUCCESS in 10m 03s\n- cross-keystone-py27 http://logs.openstack.org/87/659087/1/check/cross-keystone-py27/f4e7138/ : SUCCESS in 11m 52s\n- cross-keystone-py35 http://logs.openstack.org/87/659087/1/check/cross-keystone-py35/9da5d21/ : SUCCESS in 11m 31s\n- cross-neutron-py27 http://logs.openstack.org/87/659087/1/check/cross-neutron-py27/9d85e2c/ : SUCCESS in 22m 15s\n- cross-neutron-py35 http://logs.openstack.org/87/659087/1/check/cross-neutron-py35/1fd9847/ : SUCCESS in 25m 36s\n- cross-nova-py27 http://logs.openstack.org/87/659087/1/check/cross-nova-py27/c8acb6b/ : SUCCESS in 13m 49s\n- cross-nova-py35 http://logs.openstack.org/87/659087/1/check/cross-nova-py35/f6f6979/ : SUCCESS in 13m 00s\n- cross-nova-functional http://logs.openstack.org/87/659087/1/check/cross-nova-functional/01c072b/ : SUCCESS in 20m 14s\n- cross-swift-py27 http://logs.openstack.org/87/659087/1/check/cross-swift-py27/74ef722/ : SUCCESS in 6m 46s\n- cross-swift-py35 http://logs.openstack.org/87/659087/1/check/cross-swift-py35/2d74ee7/ : SUCCESS in 3m 35s\n- requirements-integration http://logs.openstack.org/87/659087/1/check/requirements-integration/9b00bbb/ : SUCCESS in 1h 02m 36s\n- neutron-grenade-multinode http://logs.openstack.org/87/659087/1/check/neutron-grenade-multinode/a817b95/ : SUCCESS in 1h 02m 35s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/87/659087/1/check/legacy-tempest-dsvm-neutron-full/32f245b/ : SUCCESS in 1h 36m 07s\n- neutron-grenade http://logs.openstack.org/87/659087/1/check/neutron-grenade/5848b91/ : SUCCESS in 42m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2834bac75d39abd12955020f6b57a0d983e8f135","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 17:31:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ca4b484abfb2b5f3d38d3d84291522dd130a710","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-14 19:50:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1d4603c2b83b43a90007708575fb72fc617c66c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 19:50:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74998c3c3a8a233a62db2c950364a4c5c396e3b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 22:34:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"06c3139dfd85e700046d7cf6be9a9992fdf3be89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 22:34:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/659087/1/gate/openstack-tox-pep8/1e61030/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/87/659087/1/gate/openstack-tox-py27/93f742f/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/87/659087/1/gate/openstack-tox-py35/f62b20e/ : SUCCESS in 3m 57s\n- openstack-tox-validate http://logs.openstack.org/87/659087/1/gate/openstack-tox-validate/2bffca3/ : SUCCESS in 3m 27s\n- requirements-tox-babel http://logs.openstack.org/87/659087/1/gate/requirements-tox-babel/4ece70a/ : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/659087/1/gate/requirements-tox-py27-check-uc/858a858/ : SUCCESS in 6m 13s\n- cross-cinder-py27 http://logs.openstack.org/87/659087/1/gate/cross-cinder-py27/13a2552/ : SUCCESS in 7m 57s\n- cross-glance-py27 http://logs.openstack.org/87/659087/1/gate/cross-glance-py27/c0e882a/ : SUCCESS in 9m 12s\n- cross-glance-py35 http://logs.openstack.org/87/659087/1/gate/cross-glance-py35/ab39a03/ : SUCCESS in 8m 42s\n- cross-horizon-py27 http://logs.openstack.org/87/659087/1/gate/cross-horizon-py27/7ce970c/ : SUCCESS in 9m 34s\n- cross-horizon-py35 http://logs.openstack.org/87/659087/1/gate/cross-horizon-py35/c1bae97/ : SUCCESS in 9m 04s\n- cross-keystone-py27 http://logs.openstack.org/87/659087/1/gate/cross-keystone-py27/517c9df/ : SUCCESS in 12m 59s\n- cross-keystone-py35 http://logs.openstack.org/87/659087/1/gate/cross-keystone-py35/e90bf14/ : SUCCESS in 15m 51s\n- cross-neutron-py27 http://logs.openstack.org/87/659087/1/gate/cross-neutron-py27/9225b95/ : SUCCESS in 21m 06s\n- cross-neutron-py35 http://logs.openstack.org/87/659087/1/gate/cross-neutron-py35/9628efb/ : SUCCESS in 23m 23s\n- cross-nova-py27 http://logs.openstack.org/87/659087/1/gate/cross-nova-py27/b2aa72f/ : SUCCESS in 20m 25s\n- cross-nova-py35 http://logs.openstack.org/87/659087/1/gate/cross-nova-py35/c659c89/ : SUCCESS in 11m 49s\n- cross-nova-functional http://logs.openstack.org/87/659087/1/gate/cross-nova-functional/985c602/ : SUCCESS in 18m 43s\n- cross-swift-py27 http://logs.openstack.org/87/659087/1/gate/cross-swift-py27/13debe9/ : SUCCESS in 8m 06s\n- cross-swift-py35 http://logs.openstack.org/87/659087/1/gate/cross-swift-py35/f4bd394/ : SUCCESS in 3m 36s\n- requirements-integration http://logs.openstack.org/87/659087/1/gate/requirements-integration/2ac5b31/ : SUCCESS in 51m 21s\n- neutron-grenade-multinode http://logs.openstack.org/87/659087/1/gate/neutron-grenade-multinode/a03103d/ : SUCCESS in 51m 03s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/87/659087/1/gate/legacy-tempest-dsvm-neutron-full/ee3d748/ : SUCCESS in 1h 15m 32s\n- neutron-grenade http://logs.openstack.org/87/659087/1/gate/neutron-grenade/c0f74ab/ : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a0c613123f4cf38aaab2c144d0d909550e06d44","revisions":{"6a0c613123f4cf38aaab2c144d0d909550e06d44":{"kind":"REWORK","_number":1,"created":"2019-05-14 14:49:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/659087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/659087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/659087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/659087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/659087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/659087/1"}}},"commit":{"parents":[{"commit":"36780a9026f95f268bddec14f17a33c7004e4c21","subject":"Merge \"update constraint for python-troveclient to new release 2.12.1\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36780a9026f95f268bddec14f17a33c7004e4c21"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-14 14:49:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-14 14:49:50.000000000","tz":0},"subject":"update constraint for os-apply-config to new release 7.2.4","message":"update constraint for os-apply-config to new release 7.2.4\n\nChange-Id: I64fa2e2e1149798aa80624d7be6ab48600e297de\nmeta:version: 7.2.4\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I9216306aef6974b9275efeebf32c31631ffa63a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a0c613123f4cf38aaab2c144d0d909550e06d44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a0c613123f4cf38aaab2c144d0d909550e06d44"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
