)]}'
{"id":"openstack%2Frequirements~659088","triplet_id":"openstack%2Frequirements~stable%2Fpike~I9bf4483ae59a207057ed3425e59505451743e36c","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I9bf4483ae59a207057ed3425e59505451743e36c","subject":"update constraint for os-net-config to new release 7.3.9","status":"MERGED","created":"2019-05-14 14:51:10.000000000","updated":"2019-05-15 03:54:26.000000000","submitted":"2019-05-15 03:54:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659088-1557892466329-88673ecb","meta_rev_id":"7aa5dc3e00ca63563561f36c348718d805241f49","_number":659088,"virtual_id_number":659088,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-15 03:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-14 17:31:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-14 19:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-14 19:50:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 17:31:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-14 19:50:22.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-15 03:54:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d4daaecbc518d52f2ee2821a66ae37e3f31b530","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-14 14:51:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"488196292238f3d4c95e15f660ffd83eda6a5965","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:23:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/88/659088/1/check/openstack-tox-pypy/2cbe780/ : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/659088/1/check/openstack-tox-pep8/db8e627/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/88/659088/1/check/openstack-tox-py27/4e55665/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/88/659088/1/check/openstack-tox-py35/944a4df/ : SUCCESS in 4m 13s\n- openstack-tox-validate http://logs.openstack.org/88/659088/1/check/openstack-tox-validate/d28213e/ : SUCCESS in 3m 49s\n- requirements-tox-babel http://logs.openstack.org/88/659088/1/check/requirements-tox-babel/02da973/ : SUCCESS in 4m 24s\n- requirements-tox-py27-check-uc http://logs.openstack.org/88/659088/1/check/requirements-tox-py27-check-uc/9cb5de7/ : SUCCESS in 12m 29s\n- cross-cinder-py27 http://logs.openstack.org/88/659088/1/check/cross-cinder-py27/0ca79d6/ : SUCCESS in 10m 49s\n- cross-cinder-py35 http://logs.openstack.org/88/659088/1/check/cross-cinder-py35/7e51c9c/ : SUCCESS in 10m 53s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/88/659088/1/check/cross-glance-py27/fe009c9/ : SUCCESS in 11m 21s\n- cross-glance-py35 http://logs.openstack.org/88/659088/1/check/cross-glance-py35/84db76a/ : SUCCESS in 9m 50s\n- cross-horizon-py27 http://logs.openstack.org/88/659088/1/check/cross-horizon-py27/f592e61/ : SUCCESS in 9m 58s\n- cross-horizon-py35 http://logs.openstack.org/88/659088/1/check/cross-horizon-py35/3ae62bb/ : SUCCESS in 11m 13s\n- cross-keystone-py27 http://logs.openstack.org/88/659088/1/check/cross-keystone-py27/f5db812/ : SUCCESS in 12m 06s\n- cross-keystone-py35 http://logs.openstack.org/88/659088/1/check/cross-keystone-py35/2f1f427/ : SUCCESS in 11m 12s\n- cross-neutron-py27 http://logs.openstack.org/88/659088/1/check/cross-neutron-py27/c7159ee/ : SUCCESS in 31m 12s\n- cross-neutron-py35 http://logs.openstack.org/88/659088/1/check/cross-neutron-py35/21e67f8/ : SUCCESS in 21m 40s\n- cross-nova-py27 http://logs.openstack.org/88/659088/1/check/cross-nova-py27/fef1453/ : SUCCESS in 12m 43s\n- cross-nova-py35 http://logs.openstack.org/88/659088/1/check/cross-nova-py35/05ca83b/ : SUCCESS in 10m 44s\n- cross-nova-functional http://logs.openstack.org/88/659088/1/check/cross-nova-functional/33878c0/ : SUCCESS in 18m 42s\n- cross-swift-py27 http://logs.openstack.org/88/659088/1/check/cross-swift-py27/6a38a70/ : SUCCESS in 13m 56s\n- cross-swift-py35 http://logs.openstack.org/88/659088/1/check/cross-swift-py35/6cd164e/ : SUCCESS in 2m 57s\n- requirements-integration http://logs.openstack.org/88/659088/1/check/requirements-integration/d1ebe84/ : SUCCESS in 56m 56s\n- neutron-grenade-multinode http://logs.openstack.org/88/659088/1/check/neutron-grenade-multinode/ea184bc/ : SUCCESS in 1h 14m 17s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/88/659088/1/check/legacy-tempest-dsvm-neutron-full/c0f74a9/ : SUCCESS in 1h 26m 06s\n- neutron-grenade http://logs.openstack.org/88/659088/1/check/neutron-grenade/cb0ad39/ : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7d6f07a5037c03f2969ef08e31ded3622a0c1701","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 17:31:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81da0d0ed644dbbd38e8aa544f408c27cd5e9b7f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-14 19:50:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16502807ca3524c904cb6536a21d8c82f3f28656","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 19:50:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf8a8787b7a8bc97e3bd26e9d22093db1f2742fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 22:43:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/88/659088/1/gate/openstack-tox-pep8/f802cfd/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/88/659088/1/gate/openstack-tox-py27/d51134e/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/88/659088/1/gate/openstack-tox-py35/fd36a48/ : SUCCESS in 3m 46s\n- openstack-tox-validate http://logs.openstack.org/88/659088/1/gate/openstack-tox-validate/c32dc0e/ : SUCCESS in 3m 20s\n- requirements-tox-babel http://logs.openstack.org/88/659088/1/gate/requirements-tox-babel/32a9b65/ : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc http://logs.openstack.org/88/659088/1/gate/requirements-tox-py27-check-uc/98c0238/ : SUCCESS in 6m 34s\n- cross-cinder-py27 http://logs.openstack.org/88/659088/1/gate/cross-cinder-py27/3b2a7f4/ : SUCCESS in 8m 02s\n- cross-glance-py27 http://logs.openstack.org/88/659088/1/gate/cross-glance-py27/e171e65/ : SUCCESS in 8m 33s\n- cross-glance-py35 http://logs.openstack.org/88/659088/1/gate/cross-glance-py35/eff0bd8/ : SUCCESS in 9m 52s\n- cross-horizon-py27 http://logs.openstack.org/88/659088/1/gate/cross-horizon-py27/74782d3/ : SUCCESS in 10m 22s\n- cross-horizon-py35 http://logs.openstack.org/88/659088/1/gate/cross-horizon-py35/2e5c5fc/ : SUCCESS in 9m 43s\n- cross-keystone-py27 http://logs.openstack.org/88/659088/1/gate/cross-keystone-py27/3ef5e8d/ : SUCCESS in 12m 00s\n- cross-keystone-py35 http://logs.openstack.org/88/659088/1/gate/cross-keystone-py35/50b7427/ : SUCCESS in 11m 39s\n- cross-neutron-py27 http://logs.openstack.org/88/659088/1/gate/cross-neutron-py27/2728380/ : SUCCESS in 28m 28s\n- cross-neutron-py35 http://logs.openstack.org/88/659088/1/gate/cross-neutron-py35/726ae90/ : SUCCESS in 23m 48s\n- cross-nova-py27 http://logs.openstack.org/88/659088/1/gate/cross-nova-py27/e23b56d/ : SUCCESS in 11m 29s\n- cross-nova-py35 http://logs.openstack.org/88/659088/1/gate/cross-nova-py35/5a462b0/ : SUCCESS in 12m 47s\n- cross-nova-functional http://logs.openstack.org/88/659088/1/gate/cross-nova-functional/421f605/ : SUCCESS in 22m 25s\n- cross-swift-py27 http://logs.openstack.org/88/659088/1/gate/cross-swift-py27/2e26336/ : FAILURE in 7m 35s\n- cross-swift-py35 http://logs.openstack.org/88/659088/1/gate/cross-swift-py35/30e8fe3/ : SUCCESS in 3m 29s\n- requirements-integration http://logs.openstack.org/88/659088/1/gate/requirements-integration/2e4f47b/ : SUCCESS in 46m 13s\n- neutron-grenade-multinode http://logs.openstack.org/88/659088/1/gate/neutron-grenade-multinode/ab74eb4/ : SUCCESS in 58m 55s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/88/659088/1/gate/legacy-tempest-dsvm-neutron-full/a4ef4cf/ : SUCCESS in 1h 33m 07s\n- neutron-grenade http://logs.openstack.org/88/659088/1/gate/neutron-grenade/b7b8e78/ : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2cfb3e76f85d5c06bc63fbeec731a3df0e5ac129","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 23:27:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc1e785b91d249bd57b881728f99fa0bc9d6ff21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 00:45:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/88/659088/1/check/openstack-tox-pypy/95875ad/ : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/659088/1/check/openstack-tox-pep8/f409a7f/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/88/659088/1/check/openstack-tox-py27/eebb1f5/ : SUCCESS in 5m 04s\n- openstack-tox-py35 http://logs.openstack.org/88/659088/1/check/openstack-tox-py35/25922e3/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/88/659088/1/check/openstack-tox-validate/ccfe82f/ : SUCCESS in 3m 02s\n- requirements-tox-babel http://logs.openstack.org/88/659088/1/check/requirements-tox-babel/5d2633e/ : SUCCESS in 4m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/88/659088/1/check/requirements-tox-py27-check-uc/b5cf424/ : SUCCESS in 5m 51s\n- cross-cinder-py27 http://logs.openstack.org/88/659088/1/check/cross-cinder-py27/f0f6013/ : SUCCESS in 8m 43s\n- cross-cinder-py35 http://logs.openstack.org/88/659088/1/check/cross-cinder-py35/fe738a0/ : SUCCESS in 8m 29s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/88/659088/1/check/cross-glance-py27/f76b532/ : SUCCESS in 8m 38s\n- cross-glance-py35 http://logs.openstack.org/88/659088/1/check/cross-glance-py35/f7db165/ : SUCCESS in 9m 27s\n- cross-horizon-py27 http://logs.openstack.org/88/659088/1/check/cross-horizon-py27/db639bc/ : SUCCESS in 11m 12s\n- cross-horizon-py35 http://logs.openstack.org/88/659088/1/check/cross-horizon-py35/c35ae2f/ : SUCCESS in 9m 59s\n- cross-keystone-py27 http://logs.openstack.org/88/659088/1/check/cross-keystone-py27/d45af4a/ : SUCCESS in 13m 28s\n- cross-keystone-py35 http://logs.openstack.org/88/659088/1/check/cross-keystone-py35/4919d21/ : SUCCESS in 17m 58s\n- cross-neutron-py27 http://logs.openstack.org/88/659088/1/check/cross-neutron-py27/0e4d6b0/ : SUCCESS in 23m 49s\n- cross-neutron-py35 http://logs.openstack.org/88/659088/1/check/cross-neutron-py35/2b23811/ : SUCCESS in 24m 31s\n- cross-nova-py27 http://logs.openstack.org/88/659088/1/check/cross-nova-py27/65fecab/ : SUCCESS in 13m 50s\n- cross-nova-py35 http://logs.openstack.org/88/659088/1/check/cross-nova-py35/87fb5e7/ : SUCCESS in 13m 39s\n- cross-nova-functional http://logs.openstack.org/88/659088/1/check/cross-nova-functional/367a2b8/ : SUCCESS in 18m 53s\n- cross-swift-py27 http://logs.openstack.org/88/659088/1/check/cross-swift-py27/b43314a/ : SUCCESS in 9m 41s\n- cross-swift-py35 http://logs.openstack.org/88/659088/1/check/cross-swift-py35/6618abc/ : SUCCESS in 2m 59s\n- requirements-integration http://logs.openstack.org/88/659088/1/check/requirements-integration/d6d833a/ : SUCCESS in 44m 24s\n- neutron-grenade-multinode http://logs.openstack.org/88/659088/1/check/neutron-grenade-multinode/680b8eb/ : SUCCESS in 54m 40s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/88/659088/1/check/legacy-tempest-dsvm-neutron-full/038dc9c/ : FAILURE in 1h 13m 28s\n- neutron-grenade http://logs.openstack.org/88/659088/1/check/neutron-grenade/e9d1923/ : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d37ef83a1c28af361c0a0eb35608b177caa008d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-15 00:52:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b84e8fff9071a47a8f7b7375073411384cb63fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 02:28:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/88/659088/1/check/openstack-tox-pypy/0d39279/ : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/659088/1/check/openstack-tox-pep8/ba22f72/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/88/659088/1/check/openstack-tox-py27/5ddffba/ : SUCCESS in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/88/659088/1/check/openstack-tox-py35/ea9fb20/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/88/659088/1/check/openstack-tox-validate/b04bb3a/ : SUCCESS in 4m 36s\n- requirements-tox-babel http://logs.openstack.org/88/659088/1/check/requirements-tox-babel/3c550d5/ : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/88/659088/1/check/requirements-tox-py27-check-uc/3404577/ : SUCCESS in 5m 50s\n- cross-cinder-py27 http://logs.openstack.org/88/659088/1/check/cross-cinder-py27/682f77a/ : SUCCESS in 7m 43s\n- cross-cinder-py35 http://logs.openstack.org/88/659088/1/check/cross-cinder-py35/9fe3653/ : SUCCESS in 6m 27s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/88/659088/1/check/cross-glance-py27/1961c62/ : SUCCESS in 11m 18s\n- cross-glance-py35 http://logs.openstack.org/88/659088/1/check/cross-glance-py35/c6bb1b8/ : SUCCESS in 9m 40s\n- cross-horizon-py27 http://logs.openstack.org/88/659088/1/check/cross-horizon-py27/007004f/ : SUCCESS in 8m 43s\n- cross-horizon-py35 http://logs.openstack.org/88/659088/1/check/cross-horizon-py35/77ab36f/ : SUCCESS in 9m 37s\n- cross-keystone-py27 http://logs.openstack.org/88/659088/1/check/cross-keystone-py27/39e8968/ : SUCCESS in 11m 05s\n- cross-keystone-py35 http://logs.openstack.org/88/659088/1/check/cross-keystone-py35/3e30e56/ : SUCCESS in 37m 29s\n- cross-neutron-py27 http://logs.openstack.org/88/659088/1/check/cross-neutron-py27/b74dc4c/ : SUCCESS in 22m 27s\n- cross-neutron-py35 http://logs.openstack.org/88/659088/1/check/cross-neutron-py35/aaab5c9/ : SUCCESS in 22m 34s\n- cross-nova-py27 http://logs.openstack.org/88/659088/1/check/cross-nova-py27/a721be2/ : SUCCESS in 11m 48s\n- cross-nova-py35 http://logs.openstack.org/88/659088/1/check/cross-nova-py35/76d29da/ : SUCCESS in 16m 50s\n- cross-nova-functional http://logs.openstack.org/88/659088/1/check/cross-nova-functional/bf883ea/ : SUCCESS in 19m 23s\n- cross-swift-py27 http://logs.openstack.org/88/659088/1/check/cross-swift-py27/556eb2a/ : SUCCESS in 8m 48s\n- cross-swift-py35 http://logs.openstack.org/88/659088/1/check/cross-swift-py35/0f0f29e/ : SUCCESS in 5m 16s\n- requirements-integration http://logs.openstack.org/88/659088/1/check/requirements-integration/7e08cd1/ : SUCCESS in 49m 41s\n- neutron-grenade-multinode http://logs.openstack.org/88/659088/1/check/neutron-grenade-multinode/99c7a72/ : SUCCESS in 1h 07m 34s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/88/659088/1/check/legacy-tempest-dsvm-neutron-full/3b6e575/ : SUCCESS in 1h 30m 14s\n- neutron-grenade http://logs.openstack.org/88/659088/1/check/neutron-grenade/5ec79ce/ : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"052462a0f612bdcc2003e4fe75be61fdbf69286f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 02:28:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6949983ffb8852af5ca1d1e600511a64d98169bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 03:54:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/659088/1/gate/openstack-tox-pep8/fb12220/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/88/659088/1/gate/openstack-tox-py27/c26a220/ : SUCCESS in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/88/659088/1/gate/openstack-tox-py35/2dc8f94/ : SUCCESS in 5m 48s\n- openstack-tox-validate http://logs.openstack.org/88/659088/1/gate/openstack-tox-validate/f640577/ : SUCCESS in 5m 49s\n- requirements-tox-babel http://logs.openstack.org/88/659088/1/gate/requirements-tox-babel/ebbd7c7/ : SUCCESS in 4m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/88/659088/1/gate/requirements-tox-py27-check-uc/61a1c37/ : SUCCESS in 6m 58s\n- cross-cinder-py27 http://logs.openstack.org/88/659088/1/gate/cross-cinder-py27/4e135b0/ : SUCCESS in 9m 07s\n- cross-glance-py27 http://logs.openstack.org/88/659088/1/gate/cross-glance-py27/7cd6d83/ : SUCCESS in 10m 33s\n- cross-glance-py35 http://logs.openstack.org/88/659088/1/gate/cross-glance-py35/da6a6b6/ : SUCCESS in 9m 50s\n- cross-horizon-py27 http://logs.openstack.org/88/659088/1/gate/cross-horizon-py27/94d91cd/ : SUCCESS in 10m 28s\n- cross-horizon-py35 http://logs.openstack.org/88/659088/1/gate/cross-horizon-py35/62c2f96/ : SUCCESS in 9m 43s\n- cross-keystone-py27 http://logs.openstack.org/88/659088/1/gate/cross-keystone-py27/86c3efc/ : SUCCESS in 10m 34s\n- cross-keystone-py35 http://logs.openstack.org/88/659088/1/gate/cross-keystone-py35/dc086e6/ : SUCCESS in 13m 01s\n- cross-neutron-py27 http://logs.openstack.org/88/659088/1/gate/cross-neutron-py27/8f0a107/ : SUCCESS in 20m 58s\n- cross-neutron-py35 http://logs.openstack.org/88/659088/1/gate/cross-neutron-py35/ab849ba/ : SUCCESS in 21m 40s\n- cross-nova-py27 http://logs.openstack.org/88/659088/1/gate/cross-nova-py27/146db72/ : SUCCESS in 13m 05s\n- cross-nova-py35 http://logs.openstack.org/88/659088/1/gate/cross-nova-py35/6aca59a/ : SUCCESS in 12m 30s\n- cross-nova-functional http://logs.openstack.org/88/659088/1/gate/cross-nova-functional/6af4a77/ : SUCCESS in 19m 16s\n- cross-swift-py27 http://logs.openstack.org/88/659088/1/gate/cross-swift-py27/2f3aead/ : SUCCESS in 7m 11s\n- cross-swift-py35 http://logs.openstack.org/88/659088/1/gate/cross-swift-py35/733f36f/ : SUCCESS in 3m 20s\n- requirements-integration http://logs.openstack.org/88/659088/1/gate/requirements-integration/0c307c8/ : SUCCESS in 38m 55s\n- neutron-grenade-multinode http://logs.openstack.org/88/659088/1/gate/neutron-grenade-multinode/3f55fdd/ : SUCCESS in 59m 30s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/88/659088/1/gate/legacy-tempest-dsvm-neutron-full/c47c88c/ : SUCCESS in 1h 20m 45s\n- neutron-grenade http://logs.openstack.org/88/659088/1/gate/neutron-grenade/2b9f010/ : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a4b5a5b38f573764fa81355730ec48bcf2a89bec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 03:54:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0607ad8f2097b337e6ebaa0df7c6702f85baf2d1","revisions":{"0607ad8f2097b337e6ebaa0df7c6702f85baf2d1":{"kind":"REWORK","_number":1,"created":"2019-05-14 14:51:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/659088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/659088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/659088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/659088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/659088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/659088/1"}}},"commit":{"parents":[{"commit":"36780a9026f95f268bddec14f17a33c7004e4c21","subject":"Merge \"update constraint for python-troveclient to new release 2.12.1\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36780a9026f95f268bddec14f17a33c7004e4c21"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-14 14:51:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-14 14:51:08.000000000","tz":0},"subject":"update constraint for os-net-config to new release 7.3.9","message":"update constraint for os-net-config to new release 7.3.9\n\nChange-Id: I9bf4483ae59a207057ed3425e59505451743e36c\nmeta:version: 7.3.9\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I9216306aef6974b9275efeebf32c31631ffa63a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0607ad8f2097b337e6ebaa0df7c6702f85baf2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0607ad8f2097b337e6ebaa0df7c6702f85baf2d1"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
