)]}'
{"id":"openstack%2Frequirements~659368","triplet_id":"openstack%2Frequirements~master~I85e3dc9525893de6be3fac4c952272bfb3474255","project":"openstack/requirements","branch":"master","topic":"test","hashtags":[],"change_id":"I85e3dc9525893de6be3fac4c952272bfb3474255","subject":"Make requirements-check output more obvious","status":"MERGED","created":"2019-05-15 19:59:19.000000000","updated":"2019-05-28 15:20:39.000000000","submitted":"2019-05-28 15:20:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659368-1559056839604-a2cc76c5","meta_rev_id":"655f08bc687f0ffa079bd2ed0e2028d8c477a699","_number":659368,"virtual_id_number":659368,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-28 15:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-16 01:12:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-23 06:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-28 13:37:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 01:12:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-23 06:21:06.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-28 15:20:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2af800d102578289003c72013d28bf3173698df","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-15 19:59:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"495e37b913a1c02f919cae866649ebdc45482d3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 21:35:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/659368/1/check/openstack-tox-docs/b43a408/html/ : SUCCESS in 2m 56s\n- openstack-tox-pypy http://logs.openstack.org/68/659368/1/check/openstack-tox-pypy/6430b5e/ : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/68/659368/1/check/openstack-tox-pep8/e77a969/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/68/659368/1/check/openstack-tox-py27/82cc960/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/68/659368/1/check/openstack-tox-py36/29bcbd6/ : SUCCESS in 3m 07s\n- openstack-tox-validate http://logs.openstack.org/68/659368/1/check/openstack-tox-validate/d6a9145/ : SUCCESS in 3m 11s\n- requirements-integration http://logs.openstack.org/68/659368/1/check/requirements-integration/d159b33/ : SUCCESS in 53m 07s\n- neutron-grenade-multinode http://logs.openstack.org/68/659368/1/check/neutron-grenade-multinode/4e34ce6/ : SUCCESS in 1h 31m 39s\n- tempest-full http://logs.openstack.org/68/659368/1/check/tempest-full/6248998/ : SUCCESS in 1h 32m 42s\n- neutron-grenade http://logs.openstack.org/68/659368/1/check/neutron-grenade/4baa227/ : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2961ad408dfb4cc06354c195a12e9e27c05fa196","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 01:12:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c448e84af7bc938d8ad142ff72964aa2025f5f2a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:21:06.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me.\n\nnot +W as the topic is \u0027test\u0027 so I\u0027m not 100% certain this is meant for primetime.  If it is Sean I think it\u0027s okay for you to just approve it as it has 2 cores approval.","accounts_in_message":[],"_revision_number":1},{"id":"648b45e47cbcdc13512e8056c85d21231f0c96a9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-28 13:37:24.000000000","message":"Patch Set 1: Workflow+1\n\nNot even sure why I had a topic of \"test\" on this. :)","accounts_in_message":[],"_revision_number":1},{"id":"6255cff373db1e3dbc3eb831512d5e1f4dfb6c08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 13:37:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ba0c8eaa3222974bde44c6dfdcf8bc236cf6aef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:20:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/659368/1/gate/openstack-tox-docs/b54d295/html/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/68/659368/1/gate/openstack-tox-pep8/2fc1c66/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/68/659368/1/gate/openstack-tox-py27/e950167/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/68/659368/1/gate/openstack-tox-py36/5d0ba87/ : SUCCESS in 2m 50s\n- openstack-tox-validate http://logs.openstack.org/68/659368/1/gate/openstack-tox-validate/a66c605/ : SUCCESS in 4m 52s\n- requirements-integration http://logs.openstack.org/68/659368/1/gate/requirements-integration/7382759/ : SUCCESS in 1h 16m 31s\n- neutron-grenade-multinode http://logs.openstack.org/68/659368/1/gate/neutron-grenade-multinode/e1c8ccb/ : SUCCESS in 1h 05m 36s\n- tempest-full http://logs.openstack.org/68/659368/1/gate/tempest-full/bd19092/ : SUCCESS in 1h 37m 13s\n- neutron-grenade http://logs.openstack.org/68/659368/1/gate/neutron-grenade/ec0c2f1/ : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":1},{"id":"58cae8128b3223b99457b04e5987820551eb783f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:20:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7904ae1b528ef9d5a406443524b2d7055b8b61e9","revisions":{"7904ae1b528ef9d5a406443524b2d7055b8b61e9":{"kind":"REWORK","_number":1,"created":"2019-05-15 19:59:19.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/68/659368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/659368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/659368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/659368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/659368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/659368/1"}}},"commit":{"parents":[{"commit":"abda9d216df814cf5ddb6282357419227a470467","subject":"Merge \"Add cinder extras storops library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abda9d216df814cf5ddb6282357419227a470467"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-15 19:58:22.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-15 19:58:22.000000000","tz":-300},"subject":"Make requirements-check output more obvious","message":"Make requirements-check output more obvious\n\nThis makes the output from checking requirements a little more verbose\nwhen issues are found to help make it more obvious what the issue is\nwhen there is a failure.\n\nChange-Id: I85e3dc9525893de6be3fac4c952272bfb3474255\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7904ae1b528ef9d5a406443524b2d7055b8b61e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7904ae1b528ef9d5a406443524b2d7055b8b61e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
