)]}'
{"id":"openstack%2Frequirements~659449","triplet_id":"openstack%2Frequirements~stable%2Fstein~I9a1c35621a9a820dbc853a5017f013423b2d6f2f","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I9a1c35621a9a820dbc853a5017f013423b2d6f2f","subject":"update constraint for oslo.cache to new release 1.33.3","status":"MERGED","created":"2019-05-16 06:31:24.000000000","updated":"2019-05-17 18:30:22.000000000","submitted":"2019-05-17 18:30:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659449-1558117822513-978e37b0","meta_rev_id":"d76d35027bb25f06a46df894b08bbc2b249be482","_number":659449,"virtual_id_number":659449,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-17 18:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-16 14:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-17 15:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-17 15:58:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 14:12:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-17 15:58:06.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-17 18:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48666e7dfc50a371c5243c59cadd65bfe09fbe54","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-16 06:31:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02c89950693e1ca73e568902fe1dbe9b9deb512a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 08:22:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/49/659449/1/check/openstack-tox-pypy/d7130c3/ : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/659449/1/check/openstack-tox-pep8/4cc9a58/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/49/659449/1/check/openstack-tox-py27/6b7c32f/ : SUCCESS in 2m 52s\n- openstack-tox-py35 http://logs.openstack.org/49/659449/1/check/openstack-tox-py35/addf207/ : SUCCESS in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/49/659449/1/check/openstack-tox-py36/41df6c6/ : SUCCESS in 2m 53s\n- openstack-tox-validate http://logs.openstack.org/49/659449/1/check/openstack-tox-validate/c2689be/ : SUCCESS in 3m 12s\n- requirements-tox-babel http://logs.openstack.org/49/659449/1/check/requirements-tox-babel/b0b461f/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/49/659449/1/check/requirements-tox-py27-check-uc/a24bc58/ : SUCCESS in 6m 45s\n- requirements-tox-py35-check-uc http://logs.openstack.org/49/659449/1/check/requirements-tox-py35-check-uc/fdaa7bc/ : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc http://logs.openstack.org/49/659449/1/check/requirements-tox-py36-check-uc/0ac8a83/ : SUCCESS in 6m 11s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/49/659449/1/check/cross-cinder-py27/59051ce/ : SUCCESS in 10m 15s\n- cross-cinder-py35 http://logs.openstack.org/49/659449/1/check/cross-cinder-py35/936f86a/ : SUCCESS in 8m 55s\n- cross-glance-py27 http://logs.openstack.org/49/659449/1/check/cross-glance-py27/d19c6e0/ : SUCCESS in 4m 31s\n- cross-glance-py35 http://logs.openstack.org/49/659449/1/check/cross-glance-py35/11aee35/ : SUCCESS in 3m 52s\n- cross-horizon-py27 http://logs.openstack.org/49/659449/1/check/cross-horizon-py27/22e2d2f/ : SUCCESS in 10m 44s\n- cross-horizon-py35 http://logs.openstack.org/49/659449/1/check/cross-horizon-py35/df76313/ : SUCCESS in 11m 18s\n- cross-horizon-npm http://logs.openstack.org/49/659449/1/check/cross-horizon-npm/792dd0b/ : SUCCESS in 5m 39s\n- cross-keystone-py27 http://logs.openstack.org/49/659449/1/check/cross-keystone-py27/17c342c/ : SUCCESS in 24m 23s\n- cross-keystone-py35 http://logs.openstack.org/49/659449/1/check/cross-keystone-py35/62889a2/ : SUCCESS in 25m 01s\n- cross-neutron-py27 http://logs.openstack.org/49/659449/1/check/cross-neutron-py27/5abe874/ : SUCCESS in 22m 44s\n- cross-neutron-py35 http://logs.openstack.org/49/659449/1/check/cross-neutron-py35/840779d/ : SUCCESS in 25m 32s\n- cross-nova-py27 http://logs.openstack.org/49/659449/1/check/cross-nova-py27/cce8c2c/ : SUCCESS in 12m 23s\n- cross-nova-py35 http://logs.openstack.org/49/659449/1/check/cross-nova-py35/a4a60fc/ : SUCCESS in 14m 22s\n- cross-nova-functional http://logs.openstack.org/49/659449/1/check/cross-nova-functional/35f65e9/ : SUCCESS in 20m 55s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/49/659449/1/check/cross-osc-build-sphinx-docs/7179513/html/ : SUCCESS in 5m 12s\n- cross-swift-py27 http://logs.openstack.org/49/659449/1/check/cross-swift-py27/964aee4/ : SUCCESS in 7m 39s\n- cross-swift-py35 http://logs.openstack.org/49/659449/1/check/cross-swift-py35/4aa6921/ : SUCCESS in 7m 59s\n- requirements-integration http://logs.openstack.org/49/659449/1/check/requirements-integration/464f0df/ : SUCCESS in 45m 08s\n- neutron-grenade-multinode http://logs.openstack.org/49/659449/1/check/neutron-grenade-multinode/3dbcefe/ : SUCCESS in 1h 04m 48s\n- tempest-full http://logs.openstack.org/49/659449/1/check/tempest-full/aa0dcb5/ : SUCCESS in 1h 49m 17s\n- neutron-grenade http://logs.openstack.org/49/659449/1/check/neutron-grenade/d46fe86/ : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":1},{"id":"51a8487f305a963ad030e907140b8049bdf00146","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 14:12:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"367c250743fc253f462a4d199fd9b0433864d4c9","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-17 15:58:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8848ba623722121ad6707b55e9860254b7c59c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 15:58:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c233484290c419dbc1de11c94574780970e1f074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:30:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9e0c70ad8734c81112014a7ed81b624bdadc5076","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:30:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/49/659449/1/gate/openstack-tox-pep8/1d7e3c8/ : SUCCESS in 2m 39s\n- openstack-tox-py27 http://logs.openstack.org/49/659449/1/gate/openstack-tox-py27/8732502/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/49/659449/1/gate/openstack-tox-py35/a5a5ffb/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/49/659449/1/gate/openstack-tox-py36/b8ccbcb/ : SUCCESS in 4m 46s\n- openstack-tox-validate http://logs.openstack.org/49/659449/1/gate/openstack-tox-validate/d88246e/ : SUCCESS in 2m 43s\n- requirements-tox-babel http://logs.openstack.org/49/659449/1/gate/requirements-tox-babel/e18d5f0/ : SUCCESS in 3m 37s\n- requirements-tox-py27-check-uc http://logs.openstack.org/49/659449/1/gate/requirements-tox-py27-check-uc/61d89b0/ : SUCCESS in 9m 12s\n- requirements-tox-py35-check-uc http://logs.openstack.org/49/659449/1/gate/requirements-tox-py35-check-uc/38c5cd7/ : SUCCESS in 6m 08s\n- cross-cinder-py27 http://logs.openstack.org/49/659449/1/gate/cross-cinder-py27/90a0cb8/ : SUCCESS in 8m 07s\n- cross-cinder-py35 http://logs.openstack.org/49/659449/1/gate/cross-cinder-py35/a2d93fe/ : SUCCESS in 9m 48s\n- cross-glance-py27 http://logs.openstack.org/49/659449/1/gate/cross-glance-py27/99036b9/ : SUCCESS in 4m 15s\n- cross-glance-py35 http://logs.openstack.org/49/659449/1/gate/cross-glance-py35/68bd77e/ : SUCCESS in 4m 16s\n- cross-horizon-py27 http://logs.openstack.org/49/659449/1/gate/cross-horizon-py27/fe52a88/ : SUCCESS in 10m 40s\n- cross-horizon-py35 http://logs.openstack.org/49/659449/1/gate/cross-horizon-py35/c0a4f9c/ : SUCCESS in 9m 52s\n- cross-horizon-npm http://logs.openstack.org/49/659449/1/gate/cross-horizon-npm/90812e1/ : SUCCESS in 5m 38s\n- cross-keystone-py27 http://logs.openstack.org/49/659449/1/gate/cross-keystone-py27/30a3b00/ : SUCCESS in 21m 55s\n- cross-keystone-py35 http://logs.openstack.org/49/659449/1/gate/cross-keystone-py35/96e9b35/ : SUCCESS in 26m 24s\n- cross-neutron-py27 http://logs.openstack.org/49/659449/1/gate/cross-neutron-py27/d1072e9/ : SUCCESS in 23m 24s\n- cross-neutron-py35 http://logs.openstack.org/49/659449/1/gate/cross-neutron-py35/fb1cac7/ : SUCCESS in 22m 42s\n- cross-nova-py27 http://logs.openstack.org/49/659449/1/gate/cross-nova-py27/d04566c/ : SUCCESS in 11m 38s\n- cross-nova-py35 http://logs.openstack.org/49/659449/1/gate/cross-nova-py35/60dfd2c/ : SUCCESS in 12m 12s\n- cross-nova-functional http://logs.openstack.org/49/659449/1/gate/cross-nova-functional/4dd3c7a/ : SUCCESS in 17m 54s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/49/659449/1/gate/cross-osc-build-sphinx-docs/05733d9/html/ : SUCCESS in 5m 06s\n- cross-swift-py27 http://logs.openstack.org/49/659449/1/gate/cross-swift-py27/93f0f66/ : SUCCESS in 7m 31s\n- cross-swift-py35 http://logs.openstack.org/49/659449/1/gate/cross-swift-py35/e023a3e/ : SUCCESS in 12m 16s\n- requirements-integration http://logs.openstack.org/49/659449/1/gate/requirements-integration/ab96a57/ : SUCCESS in 51m 46s\n- neutron-grenade-multinode http://logs.openstack.org/49/659449/1/gate/neutron-grenade-multinode/10a2e22/ : SUCCESS in 1h 09m 06s\n- tempest-full http://logs.openstack.org/49/659449/1/gate/tempest-full/c4c8e4b/ : SUCCESS in 1h 47m 34s\n- neutron-grenade http://logs.openstack.org/49/659449/1/gate/neutron-grenade/80f834a/ : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae4357d5abbbc11b32018061cfb4820bad87c77a","revisions":{"ae4357d5abbbc11b32018061cfb4820bad87c77a":{"kind":"REWORK","_number":1,"created":"2019-05-16 06:31:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/659449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/659449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/659449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/659449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/659449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/659449/1"}}},"commit":{"parents":[{"commit":"913f788afc13a9b162b2dbcbcfa3a376fca872cc","subject":"Merge \"Add safety check output to the linters output\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/913f788afc13a9b162b2dbcbcfa3a376fca872cc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:31:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:31:23.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 1.33.3","message":"update constraint for oslo.cache to new release 1.33.3\n\nChange-Id: I9a1c35621a9a820dbc853a5017f013423b2d6f2f\nmeta:version: 1.33.3\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I499f962a00debb374a5c0ae967b34313026a4fe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae4357d5abbbc11b32018061cfb4820bad87c77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae4357d5abbbc11b32018061cfb4820bad87c77a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
