)]}'
{"id":"openstack%2Frequirements~659451","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ic02d8f1156bc5c98240071d356f386ff9a626581","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ic02d8f1156bc5c98240071d356f386ff9a626581","subject":"update constraint for python-cinderclient to new release 4.2.0","status":"MERGED","created":"2019-05-16 06:42:10.000000000","updated":"2019-05-17 17:13:51.000000000","submitted":"2019-05-17 17:13:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659451-1558113231784-354c625b","meta_rev_id":"6346b54627850257dbe2056225cb2195935bc15c","_number":659451,"virtual_id_number":659451,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-17 17:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-16 14:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-17 15:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-17 15:19:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 14:12:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-17 15:19:26.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-17 17:13:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebb6efdb3c7f492ecd17a7df2e0ade0b6dfe703d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-16 06:42:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"683054da09a1694b717328f4c1bf9dd9d8249492","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 08:35:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/51/659451/1/check/openstack-tox-pypy/167e470/ : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/51/659451/1/check/openstack-tox-pep8/dbef2ee/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/51/659451/1/check/openstack-tox-py27/818047a/ : SUCCESS in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/51/659451/1/check/openstack-tox-py35/99ae5cc/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/51/659451/1/check/openstack-tox-py36/11c0492/ : SUCCESS in 4m 35s\n- openstack-tox-validate http://logs.openstack.org/51/659451/1/check/openstack-tox-validate/aa23cbd/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/51/659451/1/check/requirements-tox-babel/0059366/ : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc http://logs.openstack.org/51/659451/1/check/requirements-tox-py27-check-uc/b9f8482/ : SUCCESS in 8m 24s\n- requirements-tox-py35-check-uc http://logs.openstack.org/51/659451/1/check/requirements-tox-py35-check-uc/12ab77b/ : SUCCESS in 10m 40s\n- requirements-tox-py36-check-uc http://logs.openstack.org/51/659451/1/check/requirements-tox-py36-check-uc/fa8d4b9/ : SUCCESS in 7m 08s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/51/659451/1/check/cross-cinder-py27/fbe8829/ : SUCCESS in 9m 46s\n- cross-cinder-py35 http://logs.openstack.org/51/659451/1/check/cross-cinder-py35/1066a5d/ : SUCCESS in 10m 16s\n- cross-glance-py27 http://logs.openstack.org/51/659451/1/check/cross-glance-py27/5673e06/ : SUCCESS in 5m 23s\n- cross-glance-py35 http://logs.openstack.org/51/659451/1/check/cross-glance-py35/329113d/ : SUCCESS in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/51/659451/1/check/cross-horizon-py27/8144369/ : SUCCESS in 10m 19s\n- cross-horizon-py35 http://logs.openstack.org/51/659451/1/check/cross-horizon-py35/7a8dce1/ : SUCCESS in 9m 21s\n- cross-horizon-npm http://logs.openstack.org/51/659451/1/check/cross-horizon-npm/917b895/ : SUCCESS in 7m 40s\n- cross-keystone-py27 http://logs.openstack.org/51/659451/1/check/cross-keystone-py27/2208008/ : SUCCESS in 19m 55s\n- cross-keystone-py35 http://logs.openstack.org/51/659451/1/check/cross-keystone-py35/d833a89/ : SUCCESS in 24m 07s\n- cross-neutron-py27 http://logs.openstack.org/51/659451/1/check/cross-neutron-py27/61c07e0/ : SUCCESS in 20m 42s\n- cross-neutron-py35 http://logs.openstack.org/51/659451/1/check/cross-neutron-py35/1b9a3f0/ : SUCCESS in 27m 34s\n- cross-nova-py27 http://logs.openstack.org/51/659451/1/check/cross-nova-py27/c4d0810/ : SUCCESS in 13m 29s\n- cross-nova-py35 http://logs.openstack.org/51/659451/1/check/cross-nova-py35/5dc17f3/ : SUCCESS in 13m 20s\n- cross-nova-functional http://logs.openstack.org/51/659451/1/check/cross-nova-functional/0fb04cf/ : SUCCESS in 24m 02s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/51/659451/1/check/cross-osc-build-sphinx-docs/c2c6e7c/html/ : SUCCESS in 7m 03s\n- cross-swift-py27 http://logs.openstack.org/51/659451/1/check/cross-swift-py27/2831558/ : SUCCESS in 8m 16s\n- cross-swift-py35 http://logs.openstack.org/51/659451/1/check/cross-swift-py35/e8e365f/ : SUCCESS in 12m 20s\n- requirements-integration http://logs.openstack.org/51/659451/1/check/requirements-integration/ff45ca0/ : SUCCESS in 47m 30s\n- neutron-grenade-multinode http://logs.openstack.org/51/659451/1/check/neutron-grenade-multinode/4de4bb9/ : SUCCESS in 1h 19m 08s\n- tempest-full http://logs.openstack.org/51/659451/1/check/tempest-full/f03f8b0/ : SUCCESS in 1h 36m 38s\n- neutron-grenade http://logs.openstack.org/51/659451/1/check/neutron-grenade/8666a99/ : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9f4b56adb1f264cc30a77687db0e878e22161dd7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 14:12:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2da7707a4d99064cf459a5cecacd8cf612a32f4a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-17 15:19:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"121ebb9da9f74cb44c21a8a0d30399bd682ba9a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 15:19:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"beb2274f6c43afba5e412867bc0f797a866b916d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 17:13:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5399961e8da7edf87a22261f593b196224d401ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 17:13:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/51/659451/1/gate/openstack-tox-pep8/22c0cba/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/51/659451/1/gate/openstack-tox-py27/99e724f/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/51/659451/1/gate/openstack-tox-py35/7bc5ba8/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/51/659451/1/gate/openstack-tox-py36/992f411/ : SUCCESS in 4m 36s\n- openstack-tox-validate http://logs.openstack.org/51/659451/1/gate/openstack-tox-validate/a7b2153/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/51/659451/1/gate/requirements-tox-babel/d21bf2f/ : SUCCESS in 2m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/51/659451/1/gate/requirements-tox-py27-check-uc/a67c582/ : SUCCESS in 6m 29s\n- requirements-tox-py35-check-uc http://logs.openstack.org/51/659451/1/gate/requirements-tox-py35-check-uc/ef7b0a7/ : SUCCESS in 7m 06s\n- cross-cinder-py27 http://logs.openstack.org/51/659451/1/gate/cross-cinder-py27/bec4997/ : SUCCESS in 11m 47s\n- cross-cinder-py35 http://logs.openstack.org/51/659451/1/gate/cross-cinder-py35/1333d0d/ : SUCCESS in 10m 37s\n- cross-glance-py27 http://logs.openstack.org/51/659451/1/gate/cross-glance-py27/75693b4/ : SUCCESS in 6m 28s\n- cross-glance-py35 http://logs.openstack.org/51/659451/1/gate/cross-glance-py35/86a3e10/ : SUCCESS in 7m 41s\n- cross-horizon-py27 http://logs.openstack.org/51/659451/1/gate/cross-horizon-py27/46ba20f/ : SUCCESS in 8m 52s\n- cross-horizon-py35 http://logs.openstack.org/51/659451/1/gate/cross-horizon-py35/19a89ce/ : SUCCESS in 10m 45s\n- cross-horizon-npm http://logs.openstack.org/51/659451/1/gate/cross-horizon-npm/6b3c5fc/ : SUCCESS in 5m 50s\n- cross-keystone-py27 http://logs.openstack.org/51/659451/1/gate/cross-keystone-py27/00d1142/ : SUCCESS in 21m 35s\n- cross-keystone-py35 http://logs.openstack.org/51/659451/1/gate/cross-keystone-py35/70eaee5/ : SUCCESS in 29m 08s\n- cross-neutron-py27 http://logs.openstack.org/51/659451/1/gate/cross-neutron-py27/b3c9139/ : SUCCESS in 26m 35s\n- cross-neutron-py35 http://logs.openstack.org/51/659451/1/gate/cross-neutron-py35/733e93f/ : SUCCESS in 22m 53s\n- cross-nova-py27 http://logs.openstack.org/51/659451/1/gate/cross-nova-py27/2a3ce38/ : SUCCESS in 11m 07s\n- cross-nova-py35 http://logs.openstack.org/51/659451/1/gate/cross-nova-py35/9940d90/ : SUCCESS in 14m 57s\n- cross-nova-functional http://logs.openstack.org/51/659451/1/gate/cross-nova-functional/13b6bd8/ : SUCCESS in 19m 18s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/51/659451/1/gate/cross-osc-build-sphinx-docs/1fbf7e2/html/ : SUCCESS in 5m 39s\n- cross-swift-py27 http://logs.openstack.org/51/659451/1/gate/cross-swift-py27/0444a41/ : SUCCESS in 8m 08s\n- cross-swift-py35 http://logs.openstack.org/51/659451/1/gate/cross-swift-py35/c8bc3c1/ : SUCCESS in 16m 12s\n- requirements-integration http://logs.openstack.org/51/659451/1/gate/requirements-integration/bae94ae/ : SUCCESS in 46m 14s\n- neutron-grenade-multinode http://logs.openstack.org/51/659451/1/gate/neutron-grenade-multinode/cc9a42d/ : SUCCESS in 1h 09m 56s\n- tempest-full http://logs.openstack.org/51/659451/1/gate/tempest-full/858ff0c/ : SUCCESS in 1h 32m 05s\n- neutron-grenade http://logs.openstack.org/51/659451/1/gate/neutron-grenade/904841c/ : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c4b0fe70cd94a0b423455aac2c40b54e9f5e81f","revisions":{"5c4b0fe70cd94a0b423455aac2c40b54e9f5e81f":{"kind":"REWORK","_number":1,"created":"2019-05-16 06:42:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/659451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/659451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/659451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/659451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/659451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/659451/1"}}},"commit":{"parents":[{"commit":"913f788afc13a9b162b2dbcbcfa3a376fca872cc","subject":"Merge \"Add safety check output to the linters output\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/913f788afc13a9b162b2dbcbcfa3a376fca872cc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:42:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:42:08.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 4.2.0","message":"update constraint for python-cinderclient to new release 4.2.0\n\nChange-Id: Ic02d8f1156bc5c98240071d356f386ff9a626581\nmeta:version: 4.2.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I123bf9ac5228f8d4964f4ddcd72f4dc71cffbfb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4b0fe70cd94a0b423455aac2c40b54e9f5e81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4b0fe70cd94a0b423455aac2c40b54e9f5e81f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
