)]}'
{"id":"openstack%2Frequirements~659457","triplet_id":"openstack%2Frequirements~master~I3fd1287c70bba3868ab239d01ed95f2138abdb4a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3fd1287c70bba3868ab239d01ed95f2138abdb4a","subject":"update constraint for openstacksdk to new release 0.28.0","status":"MERGED","created":"2019-05-16 06:54:16.000000000","updated":"2019-05-16 15:58:58.000000000","submitted":"2019-05-16 15:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659457-1558022338223-b2ea1c39","meta_rev_id":"f0f075493d910f82a9277a75ce91e5eacfbbacb3","_number":659457,"virtual_id_number":659457,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-16 15:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-16 14:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-16 09:53:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-16 14:13:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 09:53:50.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-05-16 14:13:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-16 15:58:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54ba9e043ae1c069925c991140cbbe3154846f0b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-16 06:54:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"872a2b49a3941d52c3f60e98dc6d8cf5670da8cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 08:39:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/57/659457/1/check/openstack-tox-docs/e534769/html/ : SUCCESS in 2m 51s\n- openstack-tox-pypy http://logs.openstack.org/57/659457/1/check/openstack-tox-pypy/f9ab132/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/57/659457/1/check/openstack-tox-pep8/290d36b/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/57/659457/1/check/openstack-tox-py27/d2aad3d/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/57/659457/1/check/openstack-tox-py36/6b74618/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/57/659457/1/check/openstack-tox-validate/97304f5/ : SUCCESS in 2m 41s\n- requirements-tox-babel http://logs.openstack.org/57/659457/1/check/requirements-tox-babel/d998cc9/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/57/659457/1/check/requirements-tox-py27-check-uc/19b876e/ : FAILURE in 3m 44s\n- requirements-tox-py36-check-uc http://logs.openstack.org/57/659457/1/check/requirements-tox-py36-check-uc/56f8fdc/ : SUCCESS in 6m 45s\n- cross-cinder-py27 http://logs.openstack.org/57/659457/1/check/cross-cinder-py27/f1c3ed0/ : FAILURE in 2m 55s\n- cross-cinder-py36 http://logs.openstack.org/57/659457/1/check/cross-cinder-py36/c4d6b42/ : FAILURE in 2m 44s\n- cross-glance-py27 http://logs.openstack.org/57/659457/1/check/cross-glance-py27/99e718a/ : FAILURE in 2m 55s\n- cross-glance-py36 http://logs.openstack.org/57/659457/1/check/cross-glance-py36/5422891/ : SUCCESS in 4m 11s\n- cross-horizon-py27 http://logs.openstack.org/57/659457/1/check/cross-horizon-py27/99f3380/ : SUCCESS in 12m 25s\n- cross-horizon-py36 http://logs.openstack.org/57/659457/1/check/cross-horizon-py36/1771eba/ : SUCCESS in 11m 05s\n- cross-horizon-npm http://logs.openstack.org/57/659457/1/check/cross-horizon-npm/6a57584/ : SUCCESS in 5m 45s\n- cross-keystone-py27 http://logs.openstack.org/57/659457/1/check/cross-keystone-py27/037d6a7/ : SUCCESS in 19m 10s\n- cross-keystone-py36 http://logs.openstack.org/57/659457/1/check/cross-keystone-py36/586c9f6/ : SUCCESS in 25m 27s\n- cross-neutron-py27 http://logs.openstack.org/57/659457/1/check/cross-neutron-py27/95843f4/ : SUCCESS in 22m 25s\n- cross-neutron-py36 http://logs.openstack.org/57/659457/1/check/cross-neutron-py36/98439ea/ : SUCCESS in 23m 29s\n- cross-nova-py27 http://logs.openstack.org/57/659457/1/check/cross-nova-py27/e1835a1/ : SUCCESS in 13m 21s\n- cross-nova-py36 http://logs.openstack.org/57/659457/1/check/cross-nova-py36/4d9d8f3/ : SUCCESS in 12m 52s\n- cross-nova-functional http://logs.openstack.org/57/659457/1/check/cross-nova-functional/0bb4326/ : SUCCESS in 17m 48s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/57/659457/1/check/cross-osc-build-sphinx-docs/63fd6be/html/ : SUCCESS in 7m 13s\n- cross-swift-py27 http://logs.openstack.org/57/659457/1/check/cross-swift-py27/707e4ab/ : SUCCESS in 7m 39s\n- cross-swift-py36 http://logs.openstack.org/57/659457/1/check/cross-swift-py36/35176f2/ : SUCCESS in 10m 50s\n- requirements-integration http://logs.openstack.org/57/659457/1/check/requirements-integration/8ddd60f/ : SUCCESS in 54m 39s\n- neutron-grenade-multinode http://logs.openstack.org/57/659457/1/check/neutron-grenade-multinode/a10c2cc/ : SUCCESS in 1h 08m 13s\n- tempest-full http://logs.openstack.org/57/659457/1/check/tempest-full/45bc763/ : SUCCESS in 1h 41m 45s\n- neutron-grenade http://logs.openstack.org/57/659457/1/check/neutron-grenade/5e45f26/ : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":1},{"id":"eae894e026dbcf6a31eb088a990daf37c06c1e29","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-16 09:53:50.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1fce95555d783639cfc0fc6858cac5501459b9e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 11:50:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/57/659457/1/check/openstack-tox-docs/4df1fc1/html/ : SUCCESS in 3m 18s\n- openstack-tox-pypy http://logs.openstack.org/57/659457/1/check/openstack-tox-pypy/0f3fe76/ : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/57/659457/1/check/openstack-tox-pep8/8475d84/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/57/659457/1/check/openstack-tox-py27/d31e51d/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/57/659457/1/check/openstack-tox-py36/c9746f8/ : SUCCESS in 3m 29s\n- openstack-tox-validate http://logs.openstack.org/57/659457/1/check/openstack-tox-validate/d7ae2d3/ : SUCCESS in 3m 26s\n- requirements-tox-babel http://logs.openstack.org/57/659457/1/check/requirements-tox-babel/d305608/ : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc http://logs.openstack.org/57/659457/1/check/requirements-tox-py27-check-uc/c04be08/ : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc http://logs.openstack.org/57/659457/1/check/requirements-tox-py36-check-uc/df7a9c7/ : SUCCESS in 7m 58s\n- cross-cinder-py27 http://logs.openstack.org/57/659457/1/check/cross-cinder-py27/eb5ba4b/ : SUCCESS in 9m 45s\n- cross-cinder-py36 http://logs.openstack.org/57/659457/1/check/cross-cinder-py36/ccd07e5/ : SUCCESS in 8m 48s\n- cross-glance-py27 http://logs.openstack.org/57/659457/1/check/cross-glance-py27/29f616f/ : SUCCESS in 4m 01s\n- cross-glance-py36 http://logs.openstack.org/57/659457/1/check/cross-glance-py36/c9fc2a5/ : SUCCESS in 4m 19s\n- cross-horizon-py27 http://logs.openstack.org/57/659457/1/check/cross-horizon-py27/d11149e/ : SUCCESS in 10m 40s\n- cross-horizon-py36 http://logs.openstack.org/57/659457/1/check/cross-horizon-py36/f7ccb0c/ : SUCCESS in 9m 37s\n- cross-horizon-npm http://logs.openstack.org/57/659457/1/check/cross-horizon-npm/c8d4234/ : SUCCESS in 6m 05s\n- cross-keystone-py27 http://logs.openstack.org/57/659457/1/check/cross-keystone-py27/ad64496/ : SUCCESS in 21m 35s\n- cross-keystone-py36 http://logs.openstack.org/57/659457/1/check/cross-keystone-py36/9979230/ : SUCCESS in 23m 52s\n- cross-neutron-py27 http://logs.openstack.org/57/659457/1/check/cross-neutron-py27/c6167ac/ : SUCCESS in 21m 57s\n- cross-neutron-py36 http://logs.openstack.org/57/659457/1/check/cross-neutron-py36/296db38/ : SUCCESS in 21m 06s\n- cross-nova-py27 http://logs.openstack.org/57/659457/1/check/cross-nova-py27/d4a7700/ : SUCCESS in 12m 19s\n- cross-nova-py36 http://logs.openstack.org/57/659457/1/check/cross-nova-py36/25d2f9b/ : SUCCESS in 12m 05s\n- cross-nova-functional http://logs.openstack.org/57/659457/1/check/cross-nova-functional/86735a4/ : SUCCESS in 18m 54s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/57/659457/1/check/cross-osc-build-sphinx-docs/578d138/html/ : SUCCESS in 6m 01s\n- cross-swift-py27 http://logs.openstack.org/57/659457/1/check/cross-swift-py27/98750df/ : SUCCESS in 8m 23s\n- cross-swift-py36 http://logs.openstack.org/57/659457/1/check/cross-swift-py36/628896d/ : SUCCESS in 7m 24s\n- requirements-integration http://logs.openstack.org/57/659457/1/check/requirements-integration/cf1e8c8/ : SUCCESS in 1h 06m 05s\n- neutron-grenade-multinode http://logs.openstack.org/57/659457/1/check/neutron-grenade-multinode/022b0dd/ : SUCCESS in 1h 09m 59s\n- tempest-full http://logs.openstack.org/57/659457/1/check/tempest-full/3e28c61/ : SUCCESS in 1h 50m 15s\n- neutron-grenade http://logs.openstack.org/57/659457/1/check/neutron-grenade/0e2d93b/ : SUCCESS in 53m 32s","accounts_in_message":[],"_revision_number":1},{"id":"49aae87b9092bdd28238d95ebeb3ac65e95183da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 14:13:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c594fd4b9f039032b843236ab5e84cfbe49fd832","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:13:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad1e21a0ff161bcb8e23099a9d119f44ca4daa1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 15:58:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/57/659457/1/gate/openstack-tox-docs/766bd0c/html/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/57/659457/1/gate/openstack-tox-pep8/f69ab37/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/57/659457/1/gate/openstack-tox-py27/9a7079f/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/57/659457/1/gate/openstack-tox-py36/a0b47d1/ : SUCCESS in 3m 27s\n- openstack-tox-validate http://logs.openstack.org/57/659457/1/gate/openstack-tox-validate/0ed0075/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/57/659457/1/gate/requirements-tox-babel/d0dff0f/ : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc http://logs.openstack.org/57/659457/1/gate/requirements-tox-py27-check-uc/6a9ec27/ : SUCCESS in 7m 06s\n- requirements-tox-py36-check-uc http://logs.openstack.org/57/659457/1/gate/requirements-tox-py36-check-uc/4fa4db5/ : SUCCESS in 6m 37s\n- cross-cinder-py27 http://logs.openstack.org/57/659457/1/gate/cross-cinder-py27/ecc7ec7/ : SUCCESS in 9m 12s\n- cross-cinder-py36 http://logs.openstack.org/57/659457/1/gate/cross-cinder-py36/4763f8d/ : SUCCESS in 8m 47s\n- cross-glance-py27 http://logs.openstack.org/57/659457/1/gate/cross-glance-py27/905e85e/ : SUCCESS in 4m 37s\n- cross-glance-py36 http://logs.openstack.org/57/659457/1/gate/cross-glance-py36/2d3a5a3/ : SUCCESS in 4m 08s\n- cross-horizon-py27 http://logs.openstack.org/57/659457/1/gate/cross-horizon-py27/88ee09d/ : SUCCESS in 9m 38s\n- cross-horizon-py36 http://logs.openstack.org/57/659457/1/gate/cross-horizon-py36/632f450/ : SUCCESS in 10m 08s\n- cross-horizon-npm http://logs.openstack.org/57/659457/1/gate/cross-horizon-npm/347d029/ : SUCCESS in 5m 55s\n- cross-keystone-py27 http://logs.openstack.org/57/659457/1/gate/cross-keystone-py27/d71ee92/ : SUCCESS in 22m 40s\n- cross-keystone-py36 http://logs.openstack.org/57/659457/1/gate/cross-keystone-py36/528ca35/ : SUCCESS in 23m 08s\n- cross-neutron-py27 http://logs.openstack.org/57/659457/1/gate/cross-neutron-py27/302b7cd/ : SUCCESS in 28m 06s\n- cross-neutron-py36 http://logs.openstack.org/57/659457/1/gate/cross-neutron-py36/63c2474/ : SUCCESS in 20m 21s\n- cross-nova-py27 http://logs.openstack.org/57/659457/1/gate/cross-nova-py27/45470d1/ : SUCCESS in 10m 44s\n- cross-nova-py36 http://logs.openstack.org/57/659457/1/gate/cross-nova-py36/602d0e9/ : SUCCESS in 10m 50s\n- cross-nova-functional http://logs.openstack.org/57/659457/1/gate/cross-nova-functional/ef04a7a/ : SUCCESS in 19m 11s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/57/659457/1/gate/cross-osc-build-sphinx-docs/9038b45/html/ : SUCCESS in 5m 48s\n- cross-swift-py27 http://logs.openstack.org/57/659457/1/gate/cross-swift-py27/47e8883/ : SUCCESS in 7m 15s\n- cross-swift-py36 http://logs.openstack.org/57/659457/1/gate/cross-swift-py36/de2d703/ : SUCCESS in 7m 45s\n- requirements-integration http://logs.openstack.org/57/659457/1/gate/requirements-integration/3e07ed4/ : SUCCESS in 55m 30s\n- neutron-grenade-multinode http://logs.openstack.org/57/659457/1/gate/neutron-grenade-multinode/9dc5734/ : SUCCESS in 1h 02m 33s\n- tempest-full http://logs.openstack.org/57/659457/1/gate/tempest-full/e88974c/ : SUCCESS in 1h 40m 04s\n- neutron-grenade http://logs.openstack.org/57/659457/1/gate/neutron-grenade/2ce5c29/ : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0fd10dd2d1ef0598bab294aa625a76c7ba5fd12d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 15:58:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a36e15de817515d75dc4d6d434bd047b9630975","revisions":{"7a36e15de817515d75dc4d6d434bd047b9630975":{"kind":"REWORK","_number":1,"created":"2019-05-16 06:54:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/659457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/659457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/659457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/659457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/659457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/659457/1"}}},"commit":{"parents":[{"commit":"2b6e70b464c8b2c646ab45c1bc98ac9005d61a90","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b6e70b464c8b2c646ab45c1bc98ac9005d61a90"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:54:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:54:14.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.28.0","message":"update constraint for openstacksdk to new release 0.28.0\n\nChange-Id: I3fd1287c70bba3868ab239d01ed95f2138abdb4a\nmeta:version: 0.28.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: I0cfdf575f60e29dd7d3eb432b6606020b7c0c5fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a36e15de817515d75dc4d6d434bd047b9630975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a36e15de817515d75dc4d6d434bd047b9630975"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
