)]}'
{"id":"openstack%2Frequirements~659458","triplet_id":"openstack%2Frequirements~master~I7e8131cbe21fa970d5e99b893e48a8c588bc3500","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7e8131cbe21fa970d5e99b893e48a8c588bc3500","subject":"update constraint for python-mistralclient to new release 3.9.0","status":"MERGED","created":"2019-05-16 06:54:55.000000000","updated":"2019-05-16 10:30:26.000000000","submitted":"2019-05-16 10:30:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659458-1558002626465-96febb6b","meta_rev_id":"47c5a095cf882a2800249e0a7a2bb0aa386e1f37","_number":659458,"virtual_id_number":659458,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-16 10:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-05-16 08:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-05-16 08:11:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 08:11:15.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-16 10:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94481a5814d4cc658b76af6c8be322038640431a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-16 06:54:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb0d91abd075e4cca0e37acdb4f1001ef719406d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-16 08:11:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e422481bdc905a2ba4c86179cb8e8ad2f5943b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 08:51:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/659458/1/check/openstack-tox-docs/3cb1403/html/ : SUCCESS in 2m 48s\n- openstack-tox-pypy http://logs.openstack.org/58/659458/1/check/openstack-tox-pypy/bfffc0a/ : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/58/659458/1/check/openstack-tox-pep8/1892d53/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/58/659458/1/check/openstack-tox-py27/4a0b19f/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/58/659458/1/check/openstack-tox-py36/c32767b/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/58/659458/1/check/openstack-tox-validate/57698bb/ : SUCCESS in 3m 05s\n- requirements-tox-babel http://logs.openstack.org/58/659458/1/check/requirements-tox-babel/fefb096/ : SUCCESS in 6m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/58/659458/1/check/requirements-tox-py27-check-uc/bca2568/ : SUCCESS in 8m 36s\n- requirements-tox-py36-check-uc http://logs.openstack.org/58/659458/1/check/requirements-tox-py36-check-uc/7ca0706/ : SUCCESS in 7m 13s\n- cross-cinder-py27 http://logs.openstack.org/58/659458/1/check/cross-cinder-py27/914f035/ : SUCCESS in 9m 18s\n- cross-cinder-py36 http://logs.openstack.org/58/659458/1/check/cross-cinder-py36/a862efc/ : SUCCESS in 7m 44s\n- cross-glance-py27 http://logs.openstack.org/58/659458/1/check/cross-glance-py27/c6c7aec/ : SUCCESS in 5m 24s\n- cross-glance-py36 http://logs.openstack.org/58/659458/1/check/cross-glance-py36/742ccb2/ : SUCCESS in 6m 00s\n- cross-horizon-py27 http://logs.openstack.org/58/659458/1/check/cross-horizon-py27/1cf1e0f/ : SUCCESS in 9m 14s\n- cross-horizon-py36 http://logs.openstack.org/58/659458/1/check/cross-horizon-py36/b0403dd/ : SUCCESS in 8m 37s\n- cross-horizon-npm http://logs.openstack.org/58/659458/1/check/cross-horizon-npm/40ebf68/ : SUCCESS in 5m 35s\n- cross-keystone-py27 http://logs.openstack.org/58/659458/1/check/cross-keystone-py27/9859750/ : SUCCESS in 21m 53s\n- cross-keystone-py36 http://logs.openstack.org/58/659458/1/check/cross-keystone-py36/89f2b5b/ : SUCCESS in 22m 41s\n- cross-neutron-py27 http://logs.openstack.org/58/659458/1/check/cross-neutron-py27/c6cb168/ : SUCCESS in 22m 04s\n- cross-neutron-py36 http://logs.openstack.org/58/659458/1/check/cross-neutron-py36/af75775/ : SUCCESS in 19m 37s\n- cross-nova-py27 http://logs.openstack.org/58/659458/1/check/cross-nova-py27/a7fbfa2/ : SUCCESS in 13m 46s\n- cross-nova-py36 http://logs.openstack.org/58/659458/1/check/cross-nova-py36/14dd373/ : SUCCESS in 11m 35s\n- cross-nova-functional http://logs.openstack.org/58/659458/1/check/cross-nova-functional/59b3365/ : SUCCESS in 17m 30s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/58/659458/1/check/cross-osc-build-sphinx-docs/23aec74/html/ : SUCCESS in 6m 01s\n- cross-swift-py27 http://logs.openstack.org/58/659458/1/check/cross-swift-py27/71336e7/ : SUCCESS in 8m 35s\n- cross-swift-py36 http://logs.openstack.org/58/659458/1/check/cross-swift-py36/e31d6dd/ : SUCCESS in 8m 41s\n- requirements-integration http://logs.openstack.org/58/659458/1/check/requirements-integration/27f92e6/ : SUCCESS in 56m 51s\n- neutron-grenade-multinode http://logs.openstack.org/58/659458/1/check/neutron-grenade-multinode/268eda7/ : SUCCESS in 1h 07m 58s\n- tempest-full http://logs.openstack.org/58/659458/1/check/tempest-full/7dfff85/ : SUCCESS in 1h 51m 17s\n- neutron-grenade http://logs.openstack.org/58/659458/1/check/neutron-grenade/bf99f67/ : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"75dfeb024b5a43b55e7538056737d254ca0318a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 08:52:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a5d96eb91f4f7518182a0fbfb1614407bce7807","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 10:30:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fe330d8d34b7337f0dc255b3c06344b0265cb1f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 10:30:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/659458/1/gate/openstack-tox-docs/d13282b/html/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/58/659458/1/gate/openstack-tox-pep8/137a962/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/58/659458/1/gate/openstack-tox-py27/85f7112/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/58/659458/1/gate/openstack-tox-py36/22b2524/ : SUCCESS in 3m 51s\n- openstack-tox-validate http://logs.openstack.org/58/659458/1/gate/openstack-tox-validate/2d86732/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/58/659458/1/gate/requirements-tox-babel/43da922/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/58/659458/1/gate/requirements-tox-py27-check-uc/f0bdc85/ : SUCCESS in 6m 53s\n- requirements-tox-py36-check-uc http://logs.openstack.org/58/659458/1/gate/requirements-tox-py36-check-uc/ee72f54/ : SUCCESS in 5m 59s\n- cross-cinder-py27 http://logs.openstack.org/58/659458/1/gate/cross-cinder-py27/5e3e076/ : SUCCESS in 9m 39s\n- cross-cinder-py36 http://logs.openstack.org/58/659458/1/gate/cross-cinder-py36/ac696f4/ : SUCCESS in 7m 58s\n- cross-glance-py27 http://logs.openstack.org/58/659458/1/gate/cross-glance-py27/6f57ab9/ : SUCCESS in 4m 08s\n- cross-glance-py36 http://logs.openstack.org/58/659458/1/gate/cross-glance-py36/b7ee57c/ : SUCCESS in 4m 49s\n- cross-horizon-py27 http://logs.openstack.org/58/659458/1/gate/cross-horizon-py27/7c90ab5/ : SUCCESS in 10m 03s\n- cross-horizon-py36 http://logs.openstack.org/58/659458/1/gate/cross-horizon-py36/a9502e1/ : SUCCESS in 7m 24s\n- cross-horizon-npm http://logs.openstack.org/58/659458/1/gate/cross-horizon-npm/44027e3/ : SUCCESS in 5m 18s\n- cross-keystone-py27 http://logs.openstack.org/58/659458/1/gate/cross-keystone-py27/d300d23/ : SUCCESS in 24m 17s\n- cross-keystone-py36 http://logs.openstack.org/58/659458/1/gate/cross-keystone-py36/e22f738/ : SUCCESS in 22m 03s\n- cross-neutron-py27 http://logs.openstack.org/58/659458/1/gate/cross-neutron-py27/445aa97/ : SUCCESS in 25m 33s\n- cross-neutron-py36 http://logs.openstack.org/58/659458/1/gate/cross-neutron-py36/5969e2d/ : SUCCESS in 20m 03s\n- cross-nova-py27 http://logs.openstack.org/58/659458/1/gate/cross-nova-py27/a0fde79/ : SUCCESS in 15m 04s\n- cross-nova-py36 http://logs.openstack.org/58/659458/1/gate/cross-nova-py36/5d03910/ : SUCCESS in 11m 18s\n- cross-nova-functional http://logs.openstack.org/58/659458/1/gate/cross-nova-functional/ca14908/ : SUCCESS in 18m 54s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/58/659458/1/gate/cross-osc-build-sphinx-docs/7bda223/html/ : SUCCESS in 7m 12s\n- cross-swift-py27 http://logs.openstack.org/58/659458/1/gate/cross-swift-py27/72cf244/ : SUCCESS in 8m 19s\n- cross-swift-py36 http://logs.openstack.org/58/659458/1/gate/cross-swift-py36/4201c7d/ : SUCCESS in 7m 34s\n- requirements-integration http://logs.openstack.org/58/659458/1/gate/requirements-integration/4edae6b/ : SUCCESS in 52m 52s\n- neutron-grenade-multinode http://logs.openstack.org/58/659458/1/gate/neutron-grenade-multinode/0c404bf/ : SUCCESS in 1h 10m 10s\n- tempest-full http://logs.openstack.org/58/659458/1/gate/tempest-full/ad0e22c/ : SUCCESS in 1h 33m 33s\n- neutron-grenade http://logs.openstack.org/58/659458/1/gate/neutron-grenade/ceb9bb7/ : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"138ecbdfd25afaf3392b2b3861da08f80b88d787","revisions":{"138ecbdfd25afaf3392b2b3861da08f80b88d787":{"kind":"REWORK","_number":1,"created":"2019-05-16 06:54:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/659458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/659458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/659458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/659458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/659458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/659458/1"}}},"commit":{"parents":[{"commit":"2b6e70b464c8b2c646ab45c1bc98ac9005d61a90","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b6e70b464c8b2c646ab45c1bc98ac9005d61a90"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:54:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:54:52.000000000","tz":0},"subject":"update constraint for python-mistralclient to new release 3.9.0","message":"update constraint for python-mistralclient to new release 3.9.0\n\nChange-Id: I7e8131cbe21fa970d5e99b893e48a8c588bc3500\nmeta:version: 3.9.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Dougal Matthews \u003cdougal@redhat.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iee9c2ad8ea90d2f494e5d5f086e6fedb81c823f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/138ecbdfd25afaf3392b2b3861da08f80b88d787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/138ecbdfd25afaf3392b2b3861da08f80b88d787"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
