)]}'
{"id":"openstack%2Frequirements~659460","triplet_id":"openstack%2Frequirements~master~Ie1b64a65d90313e02325029d3eb3e6b680b26493","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie1b64a65d90313e02325029d3eb3e6b680b26493","subject":"update constraint for oslo.serialization to new release 2.29.1","status":"MERGED","created":"2019-05-16 06:58:00.000000000","updated":"2019-05-16 16:13:49.000000000","submitted":"2019-05-16 16:13:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659460-1558023229563-01b05fd8","meta_rev_id":"ebd371500f41f0791c98c323c22a841083cf0a0d","_number":659460,"virtual_id_number":659460,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-16 16:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-16 14:13:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-16 14:13:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 14:13:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-16 16:13:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f1c1cace960f0983230223ea3e4c1e587adf9a9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-16 06:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b97232f561cfdd843e89a5dc62b235cc9d159c59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:14:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/60/659460/1/check/openstack-tox-docs/a8a11ba/html/ : SUCCESS in 3m 34s\n- openstack-tox-pypy http://logs.openstack.org/60/659460/1/check/openstack-tox-pypy/f3859cc/ : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/659460/1/check/openstack-tox-pep8/e5cf4a5/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/60/659460/1/check/openstack-tox-py27/deae023/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/60/659460/1/check/openstack-tox-py36/dd4c1d9/ : SUCCESS in 3m 08s\n- openstack-tox-validate http://logs.openstack.org/60/659460/1/check/openstack-tox-validate/d067ba4/ : SUCCESS in 2m 50s\n- requirements-tox-babel http://logs.openstack.org/60/659460/1/check/requirements-tox-babel/7b8de2b/ : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc http://logs.openstack.org/60/659460/1/check/requirements-tox-py27-check-uc/0245f1f/ : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc http://logs.openstack.org/60/659460/1/check/requirements-tox-py36-check-uc/cbe3060/ : SUCCESS in 7m 45s\n- cross-cinder-py27 http://logs.openstack.org/60/659460/1/check/cross-cinder-py27/1f63198/ : SUCCESS in 7m 57s\n- cross-cinder-py36 http://logs.openstack.org/60/659460/1/check/cross-cinder-py36/e4cd7a7/ : SUCCESS in 8m 20s\n- cross-glance-py27 http://logs.openstack.org/60/659460/1/check/cross-glance-py27/921e28c/ : SUCCESS in 4m 22s\n- cross-glance-py36 http://logs.openstack.org/60/659460/1/check/cross-glance-py36/c35360e/ : SUCCESS in 3m 51s\n- cross-horizon-py27 http://logs.openstack.org/60/659460/1/check/cross-horizon-py27/3fbf756/ : SUCCESS in 9m 56s\n- cross-horizon-py36 http://logs.openstack.org/60/659460/1/check/cross-horizon-py36/b69b5dc/ : SUCCESS in 8m 47s\n- cross-horizon-npm http://logs.openstack.org/60/659460/1/check/cross-horizon-npm/2bcf87a/ : SUCCESS in 5m 19s\n- cross-keystone-py27 http://logs.openstack.org/60/659460/1/check/cross-keystone-py27/688c05f/ : SUCCESS in 21m 38s\n- cross-keystone-py36 http://logs.openstack.org/60/659460/1/check/cross-keystone-py36/88cb6d8/ : SUCCESS in 22m 19s\n- cross-neutron-py27 http://logs.openstack.org/60/659460/1/check/cross-neutron-py27/1ce4255/ : SUCCESS in 23m 18s\n- cross-neutron-py36 http://logs.openstack.org/60/659460/1/check/cross-neutron-py36/3d33124/ : SUCCESS in 27m 12s\n- cross-nova-py27 http://logs.openstack.org/60/659460/1/check/cross-nova-py27/2cd4c0c/ : SUCCESS in 14m 00s\n- cross-nova-py36 http://logs.openstack.org/60/659460/1/check/cross-nova-py36/d32bb61/ : SUCCESS in 14m 12s\n- cross-nova-functional http://logs.openstack.org/60/659460/1/check/cross-nova-functional/59a5982/ : SUCCESS in 19m 06s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/60/659460/1/check/cross-osc-build-sphinx-docs/3425306/html/ : SUCCESS in 6m 33s\n- cross-swift-py27 http://logs.openstack.org/60/659460/1/check/cross-swift-py27/f4c088c/ : SUCCESS in 7m 27s\n- cross-swift-py36 http://logs.openstack.org/60/659460/1/check/cross-swift-py36/ecf13ad/ : SUCCESS in 8m 25s\n- requirements-integration http://logs.openstack.org/60/659460/1/check/requirements-integration/9b9c7a0/ : SUCCESS in 58m 10s\n- neutron-grenade-multinode http://logs.openstack.org/60/659460/1/check/neutron-grenade-multinode/de415b7/ : SUCCESS in 1h 09m 28s\n- tempest-full http://logs.openstack.org/60/659460/1/check/tempest-full/86092cb/ : SUCCESS in 2h 02m 12s\n- neutron-grenade http://logs.openstack.org/60/659460/1/check/neutron-grenade/d9aaf0c/ : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7a4c8ce09916a7a958b40bc7d0410b6af1ca188a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 14:13:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0b1c0da71cdf7fe5569a9fb933f97077ff5c4dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:13:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2db53019aaca1a23701e01cdd91eeefa84e841a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:13:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c2c6ef80ce730c4b81df8d8cddaf4ea74ce7ffb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:13:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/60/659460/1/gate/openstack-tox-docs/0b7b65c/html/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/60/659460/1/gate/openstack-tox-pep8/22dcc96/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/60/659460/1/gate/openstack-tox-py27/69c7f5b/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/60/659460/1/gate/openstack-tox-py36/a970cfd/ : SUCCESS in 3m 27s\n- openstack-tox-validate http://logs.openstack.org/60/659460/1/gate/openstack-tox-validate/de3dfec/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/60/659460/1/gate/requirements-tox-babel/c274c81/ : SUCCESS in 3m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/60/659460/1/gate/requirements-tox-py27-check-uc/d8f10d8/ : SUCCESS in 6m 42s\n- requirements-tox-py36-check-uc http://logs.openstack.org/60/659460/1/gate/requirements-tox-py36-check-uc/f87ac1c/ : SUCCESS in 6m 40s\n- cross-cinder-py27 http://logs.openstack.org/60/659460/1/gate/cross-cinder-py27/665bcc1/ : SUCCESS in 8m 30s\n- cross-cinder-py36 http://logs.openstack.org/60/659460/1/gate/cross-cinder-py36/e1742ef/ : SUCCESS in 8m 49s\n- cross-glance-py27 http://logs.openstack.org/60/659460/1/gate/cross-glance-py27/5eee825/ : SUCCESS in 4m 11s\n- cross-glance-py36 http://logs.openstack.org/60/659460/1/gate/cross-glance-py36/fd83d0f/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/60/659460/1/gate/cross-horizon-py27/7483d4c/ : SUCCESS in 10m 32s\n- cross-horizon-py36 http://logs.openstack.org/60/659460/1/gate/cross-horizon-py36/a988203/ : SUCCESS in 8m 58s\n- cross-horizon-npm http://logs.openstack.org/60/659460/1/gate/cross-horizon-npm/28d6f03/ : SUCCESS in 5m 14s\n- cross-keystone-py27 http://logs.openstack.org/60/659460/1/gate/cross-keystone-py27/58322fc/ : SUCCESS in 29m 39s\n- cross-keystone-py36 http://logs.openstack.org/60/659460/1/gate/cross-keystone-py36/b6fc46d/ : SUCCESS in 22m 56s\n- cross-neutron-py27 http://logs.openstack.org/60/659460/1/gate/cross-neutron-py27/c7ce77e/ : SUCCESS in 22m 43s\n- cross-neutron-py36 http://logs.openstack.org/60/659460/1/gate/cross-neutron-py36/c6aa3b7/ : SUCCESS in 25m 43s\n- cross-nova-py27 http://logs.openstack.org/60/659460/1/gate/cross-nova-py27/0856631/ : SUCCESS in 19m 36s\n- cross-nova-py36 http://logs.openstack.org/60/659460/1/gate/cross-nova-py36/ae0023e/ : SUCCESS in 12m 48s\n- cross-nova-functional http://logs.openstack.org/60/659460/1/gate/cross-nova-functional/6bd115b/ : SUCCESS in 18m 11s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/60/659460/1/gate/cross-osc-build-sphinx-docs/281eb2e/html/ : SUCCESS in 8m 32s\n- cross-swift-py27 http://logs.openstack.org/60/659460/1/gate/cross-swift-py27/8e53eac/ : SUCCESS in 8m 27s\n- cross-swift-py36 http://logs.openstack.org/60/659460/1/gate/cross-swift-py36/4727b63/ : SUCCESS in 7m 55s\n- requirements-integration http://logs.openstack.org/60/659460/1/gate/requirements-integration/590c17e/ : SUCCESS in 1h 00m 08s\n- neutron-grenade-multinode http://logs.openstack.org/60/659460/1/gate/neutron-grenade-multinode/890cd3c/ : SUCCESS in 1h 07m 25s\n- tempest-full http://logs.openstack.org/60/659460/1/gate/tempest-full/ffd10f1/ : SUCCESS in 1h 58m 21s\n- neutron-grenade http://logs.openstack.org/60/659460/1/gate/neutron-grenade/4914534/ : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25e3a7c7a2ddb83a86ba5d30d359abf4fcaf2499","revisions":{"25e3a7c7a2ddb83a86ba5d30d359abf4fcaf2499":{"kind":"REWORK","_number":1,"created":"2019-05-16 06:58:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/659460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/659460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/659460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/659460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/659460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/659460/1"}}},"commit":{"parents":[{"commit":"2b6e70b464c8b2c646ab45c1bc98ac9005d61a90","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b6e70b464c8b2c646ab45c1bc98ac9005d61a90"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:57:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:57:59.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 2.29.1","message":"update constraint for oslo.serialization to new release 2.29.1\n\nChange-Id: Ie1b64a65d90313e02325029d3eb3e6b680b26493\nmeta:version: 2.29.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ib7738f88cec17a11ff4677ff1a9af582dc423175\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25e3a7c7a2ddb83a86ba5d30d359abf4fcaf2499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25e3a7c7a2ddb83a86ba5d30d359abf4fcaf2499"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
