)]}'
{"id":"openstack%2Frequirements~659461","triplet_id":"openstack%2Frequirements~master~I01f1ec6441a41280a03f18ac42f0a409d7b6616f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I01f1ec6441a41280a03f18ac42f0a409d7b6616f","subject":"update constraint for python-novaclient to new release 14.0.0","status":"MERGED","created":"2019-05-16 06:59:25.000000000","updated":"2019-05-16 16:21:39.000000000","submitted":"2019-05-16 16:21:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659461-1558023699424-cf8f5779","meta_rev_id":"72077da4eedff8f2640db35730a160904aaba087","_number":659461,"virtual_id_number":659461,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-05-16 16:21:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-16 14:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-16 13:51:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-16 14:13:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 13:51:36.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-05-16 14:13:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-16 16:21:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb7a1a05cbe2ac3802f4897515ead8136ecdfd5d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-16 06:59:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"042e843495812f6380573d0d9ade2cfe4c5c9d6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 08:51:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/61/659461/1/check/openstack-tox-docs/8a60479/html/ : SUCCESS in 3m 09s\n- openstack-tox-pypy http://logs.openstack.org/61/659461/1/check/openstack-tox-pypy/c240fc5/ : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/61/659461/1/check/openstack-tox-pep8/b593103/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/61/659461/1/check/openstack-tox-py27/94ac0f2/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/61/659461/1/check/openstack-tox-py36/e1e5887/ : SUCCESS in 3m 27s\n- openstack-tox-validate http://logs.openstack.org/61/659461/1/check/openstack-tox-validate/5182158/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/61/659461/1/check/requirements-tox-babel/0b39d5b/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/61/659461/1/check/requirements-tox-py27-check-uc/a857efb/ : SUCCESS in 9m 23s\n- requirements-tox-py36-check-uc http://logs.openstack.org/61/659461/1/check/requirements-tox-py36-check-uc/f5b3137/ : SUCCESS in 8m 01s\n- cross-cinder-py27 http://logs.openstack.org/61/659461/1/check/cross-cinder-py27/c40bb4a/ : SUCCESS in 9m 01s\n- cross-cinder-py36 http://logs.openstack.org/61/659461/1/check/cross-cinder-py36/de46243/ : SUCCESS in 8m 15s\n- cross-glance-py27 http://logs.openstack.org/61/659461/1/check/cross-glance-py27/8569907/ : SUCCESS in 3m 55s\n- cross-glance-py36 http://logs.openstack.org/61/659461/1/check/cross-glance-py36/93a818c/ : SUCCESS in 3m 55s\n- cross-horizon-py27 http://logs.openstack.org/61/659461/1/check/cross-horizon-py27/0bf7ba5/ : SUCCESS in 9m 01s\n- cross-horizon-py36 http://logs.openstack.org/61/659461/1/check/cross-horizon-py36/e58f77c/ : SUCCESS in 7m 31s\n- cross-horizon-npm http://logs.openstack.org/61/659461/1/check/cross-horizon-npm/fa41935/ : SUCCESS in 6m 39s\n- cross-keystone-py27 http://logs.openstack.org/61/659461/1/check/cross-keystone-py27/cc49455/ : SUCCESS in 22m 40s\n- cross-keystone-py36 http://logs.openstack.org/61/659461/1/check/cross-keystone-py36/47b76fd/ : SUCCESS in 23m 17s\n- cross-neutron-py27 http://logs.openstack.org/61/659461/1/check/cross-neutron-py27/649f44b/ : SUCCESS in 20m 52s\n- cross-neutron-py36 http://logs.openstack.org/61/659461/1/check/cross-neutron-py36/a34a137/ : SUCCESS in 22m 23s\n- cross-nova-py27 http://logs.openstack.org/61/659461/1/check/cross-nova-py27/3576731/ : SUCCESS in 11m 57s\n- cross-nova-py36 http://logs.openstack.org/61/659461/1/check/cross-nova-py36/970ca6b/ : SUCCESS in 13m 55s\n- cross-nova-functional http://logs.openstack.org/61/659461/1/check/cross-nova-functional/0b4dbe1/ : SUCCESS in 17m 29s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/61/659461/1/check/cross-osc-build-sphinx-docs/34cf2d4/html/ : SUCCESS in 5m 02s\n- cross-swift-py27 http://logs.openstack.org/61/659461/1/check/cross-swift-py27/ac901aa/ : SUCCESS in 8m 09s\n- cross-swift-py36 http://logs.openstack.org/61/659461/1/check/cross-swift-py36/cf64160/ : SUCCESS in 8m 30s\n- requirements-integration http://logs.openstack.org/61/659461/1/check/requirements-integration/386efea/ : SUCCESS in 46m 44s\n- neutron-grenade-multinode http://logs.openstack.org/61/659461/1/check/neutron-grenade-multinode/344b0f3/ : SUCCESS in 58m 33s\n- tempest-full http://logs.openstack.org/61/659461/1/check/tempest-full/fc6d943/ : SUCCESS in 1h 38m 36s\n- neutron-grenade http://logs.openstack.org/61/659461/1/check/neutron-grenade/39b0339/ : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":1},{"id":"cc610c1e07f1297d3f6c45c8404413e09e7a23f9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-16 13:51:36.000000000","message":"Patch Set 1: Code-Review+1\n\nYup we need this for https://review.opendev.org/#/c/659124/.","accounts_in_message":[],"_revision_number":1},{"id":"e232bed452a41e9377cd00d40373bb6d637b300f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 14:13:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f729cc1d053c7b57b12974eed2512d18f34a88e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:13:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"893db4900cd695b9dd29c053ca8e0a4545880510","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:21:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d21b18bd152025d38965f2ba20e4e6b8d1e05d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:21:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/61/659461/1/gate/openstack-tox-docs/496759f/html/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/61/659461/1/gate/openstack-tox-pep8/670b101/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/61/659461/1/gate/openstack-tox-py27/fbd920e/ : SUCCESS in 2m 41s\n- openstack-tox-py36 http://logs.openstack.org/61/659461/1/gate/openstack-tox-py36/b40931a/ : SUCCESS in 2m 42s\n- openstack-tox-validate http://logs.openstack.org/61/659461/1/gate/openstack-tox-validate/68363b6/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/61/659461/1/gate/requirements-tox-babel/ab33fba/ : SUCCESS in 3m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/61/659461/1/gate/requirements-tox-py27-check-uc/184a7d2/ : SUCCESS in 7m 05s\n- requirements-tox-py36-check-uc http://logs.openstack.org/61/659461/1/gate/requirements-tox-py36-check-uc/55a23f6/ : SUCCESS in 6m 30s\n- cross-cinder-py27 http://logs.openstack.org/61/659461/1/gate/cross-cinder-py27/3088189/ : SUCCESS in 9m 01s\n- cross-cinder-py36 http://logs.openstack.org/61/659461/1/gate/cross-cinder-py36/5107307/ : SUCCESS in 7m 53s\n- cross-glance-py27 http://logs.openstack.org/61/659461/1/gate/cross-glance-py27/ebb042c/ : SUCCESS in 4m 30s\n- cross-glance-py36 http://logs.openstack.org/61/659461/1/gate/cross-glance-py36/5979026/ : SUCCESS in 4m 34s\n- cross-horizon-py27 http://logs.openstack.org/61/659461/1/gate/cross-horizon-py27/4218ab0/ : SUCCESS in 11m 01s\n- cross-horizon-py36 http://logs.openstack.org/61/659461/1/gate/cross-horizon-py36/e67dda1/ : SUCCESS in 8m 54s\n- cross-horizon-npm http://logs.openstack.org/61/659461/1/gate/cross-horizon-npm/341d6b3/ : SUCCESS in 8m 12s\n- cross-keystone-py27 http://logs.openstack.org/61/659461/1/gate/cross-keystone-py27/f68249b/ : SUCCESS in 23m 46s\n- cross-keystone-py36 http://logs.openstack.org/61/659461/1/gate/cross-keystone-py36/11ec41e/ : SUCCESS in 25m 11s\n- cross-neutron-py27 http://logs.openstack.org/61/659461/1/gate/cross-neutron-py27/c06f036/ : SUCCESS in 24m 05s\n- cross-neutron-py36 http://logs.openstack.org/61/659461/1/gate/cross-neutron-py36/f86493a/ : SUCCESS in 23m 08s\n- cross-nova-py27 http://logs.openstack.org/61/659461/1/gate/cross-nova-py27/c7ec9b7/ : SUCCESS in 11m 59s\n- cross-nova-py36 http://logs.openstack.org/61/659461/1/gate/cross-nova-py36/e54bd28/ : SUCCESS in 13m 28s\n- cross-nova-functional http://logs.openstack.org/61/659461/1/gate/cross-nova-functional/c50fe1c/ : SUCCESS in 17m 01s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/61/659461/1/gate/cross-osc-build-sphinx-docs/c1b2e69/html/ : SUCCESS in 5m 06s\n- cross-swift-py27 http://logs.openstack.org/61/659461/1/gate/cross-swift-py27/a283a55/ : SUCCESS in 9m 57s\n- cross-swift-py36 http://logs.openstack.org/61/659461/1/gate/cross-swift-py36/65570ad/ : SUCCESS in 8m 13s\n- requirements-integration http://logs.openstack.org/61/659461/1/gate/requirements-integration/25406bf/ : SUCCESS in 48m 52s\n- neutron-grenade-multinode http://logs.openstack.org/61/659461/1/gate/neutron-grenade-multinode/3c771ec/ : SUCCESS in 1h 03m 54s\n- tempest-full http://logs.openstack.org/61/659461/1/gate/tempest-full/3f45ca5/ : SUCCESS in 2h 01m 08s\n- neutron-grenade http://logs.openstack.org/61/659461/1/gate/neutron-grenade/295034a/ : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"090676246109e97ce784ec56ff512e9560d98ac9","revisions":{"090676246109e97ce784ec56ff512e9560d98ac9":{"kind":"REWORK","_number":1,"created":"2019-05-16 06:59:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/659461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/659461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/659461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/659461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/659461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/659461/1"}}},"commit":{"parents":[{"commit":"2b6e70b464c8b2c646ab45c1bc98ac9005d61a90","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b6e70b464c8b2c646ab45c1bc98ac9005d61a90"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:59:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:59:22.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 14.0.0","message":"update constraint for python-novaclient to new release 14.0.0\n\nChange-Id: I01f1ec6441a41280a03f18ac42f0a409d7b6616f\nmeta:version: 14.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: Ib8923b23be71d872458952e37567c39abbc6b267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/090676246109e97ce784ec56ff512e9560d98ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/090676246109e97ce784ec56ff512e9560d98ac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
