)]}'
{"id":"openstack%2Frequirements~659462","triplet_id":"openstack%2Frequirements~stable%2Frocky~I6d76614a21d40e30ae10dc47ff894a3dbb3d6c71","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"I6d76614a21d40e30ae10dc47ff894a3dbb3d6c71","subject":"update constraint for oslo.cache to new release 1.30.4","status":"MERGED","created":"2019-05-16 06:59:25.000000000","updated":"2019-05-17 18:05:01.000000000","submitted":"2019-05-17 18:05:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659462-1558116301026-61dee02a","meta_rev_id":"a3a21b456e3da5ff5d8d9558e40e16192e65b967","_number":659462,"virtual_id_number":659462,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-17 18:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-16 14:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-17 15:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-17 15:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 14:13:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-17 15:57:52.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-17 18:05:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c18fb8194a983cce8da1bae282fb7a12f06330c4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-16 06:59:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5c3a3cc5890f33c26c97428e938ffcc228ee59e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:05:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/62/659462/1/check/openstack-tox-pypy/f8a5a9a/ : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/62/659462/1/check/openstack-tox-pep8/0ec9a25/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/62/659462/1/check/openstack-tox-py27/e193327/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/62/659462/1/check/openstack-tox-py35/d5028fa/ : SUCCESS in 3m 42s\n- openstack-tox-validate http://logs.openstack.org/62/659462/1/check/openstack-tox-validate/3a01d9c/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/62/659462/1/check/requirements-tox-babel/1a4396f/ : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/62/659462/1/check/requirements-tox-py27-check-uc/a4b54de/ : SUCCESS in 6m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/62/659462/1/check/requirements-tox-py35-check-uc/cad67b1/ : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc http://logs.openstack.org/62/659462/1/check/requirements-tox-py36-check-uc/2d729fd/ : SUCCESS in 6m 40s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/62/659462/1/check/cross-cinder-py27/4d484bf/ : SUCCESS in 7m 55s\n- cross-cinder-py35 http://logs.openstack.org/62/659462/1/check/cross-cinder-py35/ae1f686/ : SUCCESS in 10m 40s\n- cross-glance-py27 http://logs.openstack.org/62/659462/1/check/cross-glance-py27/a755d29/ : SUCCESS in 4m 20s\n- cross-glance-py35 http://logs.openstack.org/62/659462/1/check/cross-glance-py35/0328c3a/ : SUCCESS in 4m 39s\n- cross-horizon-py27 http://logs.openstack.org/62/659462/1/check/cross-horizon-py27/cd4f32b/ : SUCCESS in 10m 41s\n- cross-horizon-py35 http://logs.openstack.org/62/659462/1/check/cross-horizon-py35/d56e46f/ : SUCCESS in 9m 00s\n- cross-keystone-py27 http://logs.openstack.org/62/659462/1/check/cross-keystone-py27/da6d7fc/ : SUCCESS in 15m 20s\n- cross-keystone-py35 http://logs.openstack.org/62/659462/1/check/cross-keystone-py35/4335105/ : SUCCESS in 12m 28s\n- cross-neutron-py27 http://logs.openstack.org/62/659462/1/check/cross-neutron-py27/9258151/ : SUCCESS in 26m 14s\n- cross-neutron-py35 http://logs.openstack.org/62/659462/1/check/cross-neutron-py35/1fb1977/ : SUCCESS in 31m 38s\n- cross-nova-py27 http://logs.openstack.org/62/659462/1/check/cross-nova-py27/1e73363/ : SUCCESS in 13m 11s\n- cross-nova-py35 http://logs.openstack.org/62/659462/1/check/cross-nova-py35/1c99377/ : SUCCESS in 12m 19s\n- cross-nova-functional http://logs.openstack.org/62/659462/1/check/cross-nova-functional/363aa8a/ : SUCCESS in 18m 00s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/62/659462/1/check/cross-osc-build-sphinx-docs/d2735ae/html/ : SUCCESS in 4m 45s\n- cross-swift-py27 http://logs.openstack.org/62/659462/1/check/cross-swift-py27/adfb44e/ : SUCCESS in 6m 55s\n- cross-swift-py35 http://logs.openstack.org/62/659462/1/check/cross-swift-py35/edee818/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/62/659462/1/check/openstack-tox-py36/864bac0/ : SUCCESS in 3m 06s\n- requirements-integration http://logs.openstack.org/62/659462/1/check/requirements-integration/bfe8a71/ : SUCCESS in 54m 46s\n- neutron-grenade-multinode http://logs.openstack.org/62/659462/1/check/neutron-grenade-multinode/ab49cca/ : SUCCESS in 1h 03m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/62/659462/1/check/legacy-tempest-dsvm-neutron-full/5166119/ : SUCCESS in 1h 46m 12s\n- neutron-grenade http://logs.openstack.org/62/659462/1/check/neutron-grenade/0a2237b/ : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3507a7d6019a09e628bae76f50a7f1f33aabd286","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 14:12:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"da83ecfe0e4eab63f312d7983e82e53746fdbf7d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-16 14:13:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1b8acfbe511c3b79b5107b8c71d0f00b0bc593f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 15:45:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/62/659462/1/check/openstack-tox-pypy/c5cfc51/ : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/62/659462/1/check/openstack-tox-pep8/b9b064a/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/62/659462/1/check/openstack-tox-py27/b84f29d/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/62/659462/1/check/openstack-tox-py35/0a718dc/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/62/659462/1/check/openstack-tox-validate/8a6a362/ : SUCCESS in 3m 42s\n- requirements-tox-babel http://logs.openstack.org/62/659462/1/check/requirements-tox-babel/30bae39/ : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/62/659462/1/check/requirements-tox-py27-check-uc/0449588/ : SUCCESS in 7m 16s\n- requirements-tox-py35-check-uc http://logs.openstack.org/62/659462/1/check/requirements-tox-py35-check-uc/b4d8622/ : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc http://logs.openstack.org/62/659462/1/check/requirements-tox-py36-check-uc/5e0be6f/ : SUCCESS in 6m 15s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/62/659462/1/check/cross-cinder-py27/16329cc/ : SUCCESS in 9m 25s\n- cross-cinder-py35 http://logs.openstack.org/62/659462/1/check/cross-cinder-py35/1c6bc12/ : SUCCESS in 9m 24s\n- cross-glance-py27 http://logs.openstack.org/62/659462/1/check/cross-glance-py27/f1449bd/ : SUCCESS in 4m 23s\n- cross-glance-py35 http://logs.openstack.org/62/659462/1/check/cross-glance-py35/c901d18/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/62/659462/1/check/cross-horizon-py27/429cae3/ : SUCCESS in 10m 42s\n- cross-horizon-py35 http://logs.openstack.org/62/659462/1/check/cross-horizon-py35/f6c3086/ : SUCCESS in 9m 39s\n- cross-keystone-py27 http://logs.openstack.org/62/659462/1/check/cross-keystone-py27/9ee4314/ : SUCCESS in 13m 35s\n- cross-keystone-py35 http://logs.openstack.org/62/659462/1/check/cross-keystone-py35/f63ecff/ : SUCCESS in 11m 46s\n- cross-neutron-py27 http://logs.openstack.org/62/659462/1/check/cross-neutron-py27/16e30d1/ : SUCCESS in 27m 43s\n- cross-neutron-py35 http://logs.openstack.org/62/659462/1/check/cross-neutron-py35/d0e430a/ : SUCCESS in 31m 21s\n- cross-nova-py27 http://logs.openstack.org/62/659462/1/check/cross-nova-py27/a13bc61/ : SUCCESS in 12m 14s\n- cross-nova-py35 http://logs.openstack.org/62/659462/1/check/cross-nova-py35/6164cd1/ : SUCCESS in 11m 56s\n- cross-nova-functional http://logs.openstack.org/62/659462/1/check/cross-nova-functional/9755e01/ : SUCCESS in 16m 31s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/62/659462/1/check/cross-osc-build-sphinx-docs/e701e9b/html/ : SUCCESS in 5m 06s\n- cross-swift-py27 http://logs.openstack.org/62/659462/1/check/cross-swift-py27/6557c73/ : SUCCESS in 7m 43s\n- cross-swift-py35 http://logs.openstack.org/62/659462/1/check/cross-swift-py35/9d5c004/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/62/659462/1/check/openstack-tox-py36/5977b32/ : SUCCESS in 3m 22s\n- requirements-integration http://logs.openstack.org/62/659462/1/check/requirements-integration/45f499c/ : SUCCESS in 55m 04s\n- neutron-grenade-multinode http://logs.openstack.org/62/659462/1/check/neutron-grenade-multinode/4fce810/ : SUCCESS in 1h 21m 50s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/62/659462/1/check/legacy-tempest-dsvm-neutron-full/ad1a691/ : SUCCESS in 1h 25m 44s\n- neutron-grenade http://logs.openstack.org/62/659462/1/check/neutron-grenade/ec502f7/ : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":1},{"id":"912f47493da76038c94074872d19d7b2c62d63ae","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-17 15:57:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"436c87fcf45a0178becfc398d3e92963e534ddd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 15:58:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fddada871f576290ef4b8719b6a052ce810d571d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:05:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/659462/1/gate/openstack-tox-pep8/214b53c/ : SUCCESS in 2m 41s\n- openstack-tox-py27 http://logs.openstack.org/62/659462/1/gate/openstack-tox-py27/910d2be/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/62/659462/1/gate/openstack-tox-py35/ec5a18a/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/62/659462/1/gate/openstack-tox-validate/3e37a66/ : SUCCESS in 2m 48s\n- requirements-tox-babel http://logs.openstack.org/62/659462/1/gate/requirements-tox-babel/c420963/ : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/62/659462/1/gate/requirements-tox-py27-check-uc/6fe3b4f/ : SUCCESS in 6m 15s\n- requirements-tox-py35-check-uc http://logs.openstack.org/62/659462/1/gate/requirements-tox-py35-check-uc/38a3fe0/ : SUCCESS in 9m 02s\n- cross-cinder-py27 http://logs.openstack.org/62/659462/1/gate/cross-cinder-py27/7edcf85/ : SUCCESS in 9m 01s\n- cross-cinder-py35 http://logs.openstack.org/62/659462/1/gate/cross-cinder-py35/e6cf897/ : SUCCESS in 9m 04s\n- cross-glance-py27 http://logs.openstack.org/62/659462/1/gate/cross-glance-py27/446eba9/ : SUCCESS in 4m 40s\n- cross-glance-py35 http://logs.openstack.org/62/659462/1/gate/cross-glance-py35/6a33c8b/ : SUCCESS in 4m 41s\n- cross-horizon-py27 http://logs.openstack.org/62/659462/1/gate/cross-horizon-py27/138b4be/ : SUCCESS in 10m 20s\n- cross-horizon-py35 http://logs.openstack.org/62/659462/1/gate/cross-horizon-py35/a4c297c/ : SUCCESS in 10m 02s\n- cross-keystone-py27 http://logs.openstack.org/62/659462/1/gate/cross-keystone-py27/e8d1c77/ : SUCCESS in 17m 17s\n- cross-keystone-py35 http://logs.openstack.org/62/659462/1/gate/cross-keystone-py35/5e0795f/ : SUCCESS in 12m 21s\n- cross-neutron-py27 http://logs.openstack.org/62/659462/1/gate/cross-neutron-py27/6e9cfde/ : SUCCESS in 26m 43s\n- cross-neutron-py35 http://logs.openstack.org/62/659462/1/gate/cross-neutron-py35/63d6a56/ : SUCCESS in 26m 58s\n- cross-nova-py27 http://logs.openstack.org/62/659462/1/gate/cross-nova-py27/24dc002/ : SUCCESS in 15m 53s\n- cross-nova-py35 http://logs.openstack.org/62/659462/1/gate/cross-nova-py35/ed288f7/ : SUCCESS in 12m 39s\n- cross-nova-functional http://logs.openstack.org/62/659462/1/gate/cross-nova-functional/4a5cb96/ : SUCCESS in 21m 51s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/62/659462/1/gate/cross-osc-build-sphinx-docs/c0ddbad/html/ : SUCCESS in 7m 12s\n- cross-swift-py27 http://logs.openstack.org/62/659462/1/gate/cross-swift-py27/dc76ac2/ : SUCCESS in 10m 15s\n- cross-swift-py35 http://logs.openstack.org/62/659462/1/gate/cross-swift-py35/3eca720/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/62/659462/1/gate/openstack-tox-py36/c6d5bc7/ : SUCCESS in 4m 49s\n- requirements-integration http://logs.openstack.org/62/659462/1/gate/requirements-integration/24ade17/ : SUCCESS in 1h 04m 15s\n- neutron-grenade-multinode http://logs.openstack.org/62/659462/1/gate/neutron-grenade-multinode/a327805/ : SUCCESS in 1h 14m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/62/659462/1/gate/legacy-tempest-dsvm-neutron-full/f306972/ : SUCCESS in 1h 22m 55s\n- neutron-grenade http://logs.openstack.org/62/659462/1/gate/neutron-grenade/87b2d07/ : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"684ec47cc71e99898888c9197110649d0bc46a92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:05:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7865cbcc26727b9858ec5a94781011bf42b5a2b","revisions":{"c7865cbcc26727b9858ec5a94781011bf42b5a2b":{"kind":"REWORK","_number":1,"created":"2019-05-16 06:59:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/659462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/659462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/659462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/659462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/659462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/659462/1"}}},"commit":{"parents":[{"commit":"65b77b8fb8a767b9e94ab54bc4752265ff694f8c","subject":"Merge \"Add safety check output to the linters output\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65b77b8fb8a767b9e94ab54bc4752265ff694f8c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:59:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-16 06:59:23.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 1.30.4","message":"update constraint for oslo.cache to new release 1.30.4\n\nChange-Id: I6d76614a21d40e30ae10dc47ff894a3dbb3d6c71\nmeta:version: 1.30.4\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I88dacf851a68c7b0cb6ffde1855e1fda4b96aa65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7865cbcc26727b9858ec5a94781011bf42b5a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7865cbcc26727b9858ec5a94781011bf42b5a2b"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
