)]}'
{"id":"openstack%2Frequirements~659480","triplet_id":"openstack%2Frequirements~master~Ieaf517f5b244a70821bce2a09e72e9964ff662c4","project":"openstack/requirements","branch":"master","topic":"bug/1811515","hashtags":[],"change_id":"Ieaf517f5b244a70821bce2a09e72e9964ff662c4","subject":"Bump Pyroute2 to 0.5.5","status":"ABANDONED","created":"2019-05-16 07:47:19.000000000","updated":"2019-07-01 15:26:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bb017fb4ed3cb97b7ce37756939595d8b34ccbaf","_number":659480,"virtual_id_number":659480,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-1,"date":"2019-05-16 09:36:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2019-05-16 08:22:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2019-05-16 09:00:00.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"date":"2019-05-16 07:47:44.000000000","_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"date":"2019-05-16 07:47:33.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2019-05-16 07:47:56.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"date":"2019-05-16 07:47:50.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 07:47:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-05-16 07:47:44.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-05-16 07:47:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-05-16 07:47:56.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-05-16 08:22:10.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-05-16 09:36:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1b568c17bced06a8832ca5008078ccf0dbf416e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-16 07:47:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"813b63ea8436548ea63545bb0ccdda01bdde8704","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-05-16 08:22:10.000000000","message":"Patch Set 1: Code-Review-1\n\n- why to change only upper-constraints.txt but not global-requirements.txt?\n- 0.5.4 and 0.5.5 have been excluded for reasons. if they are not relevant anymore, can you explain it in the commit message?","accounts_in_message":[],"_revision_number":1},{"id":"7b32802fe7b5aef45da63cd8d948546cbd63419b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-16 08:59:55.000000000","message":"Patch Set 1:\n\nUntil https://github.com/svinota/pyroute2/issues/599 is fixed, we need to stay in 0.5.3.\n\nI\u0027ll revisit this patch once pyroute2 solves this error.","accounts_in_message":[],"_revision_number":1},{"id":"e56090540e8911983566501b2def156948b1d3bc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-16 09:00:00.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4b636915b4dc039670822c79da6ac6619b7e9745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:36:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/80/659480/1/check/openstack-tox-docs/6168ab2/html/ : SUCCESS in 3m 05s\n- openstack-tox-pypy http://logs.openstack.org/80/659480/1/check/openstack-tox-pypy/88dc8a4/ : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/659480/1/check/openstack-tox-pep8/2b1aca6/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/80/659480/1/check/openstack-tox-py27/139181f/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/80/659480/1/check/openstack-tox-py36/b38391f/ : SUCCESS in 3m 29s\n- openstack-tox-validate http://logs.openstack.org/80/659480/1/check/openstack-tox-validate/00abfe0/ : FAILURE in 4m 37s\n- requirements-tox-babel http://logs.openstack.org/80/659480/1/check/requirements-tox-babel/f1a8433/ : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/80/659480/1/check/requirements-tox-py27-check-uc/455cd7c/ : SUCCESS in 7m 40s\n- requirements-tox-py36-check-uc http://logs.openstack.org/80/659480/1/check/requirements-tox-py36-check-uc/89424fa/ : SUCCESS in 8m 19s\n- cross-cinder-py27 http://logs.openstack.org/80/659480/1/check/cross-cinder-py27/3570ada/ : SUCCESS in 8m 55s\n- cross-cinder-py36 http://logs.openstack.org/80/659480/1/check/cross-cinder-py36/18a6615/ : SUCCESS in 7m 54s\n- cross-glance-py27 http://logs.openstack.org/80/659480/1/check/cross-glance-py27/ce25a60/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/80/659480/1/check/cross-glance-py36/01481f2/ : SUCCESS in 4m 16s\n- cross-horizon-py27 http://logs.openstack.org/80/659480/1/check/cross-horizon-py27/bcd2c54/ : SUCCESS in 10m 09s\n- cross-horizon-py36 http://logs.openstack.org/80/659480/1/check/cross-horizon-py36/680180b/ : SUCCESS in 8m 52s\n- cross-horizon-npm http://logs.openstack.org/80/659480/1/check/cross-horizon-npm/287bd6a/ : SUCCESS in 5m 52s\n- cross-keystone-py27 http://logs.openstack.org/80/659480/1/check/cross-keystone-py27/8aa48a5/ : SUCCESS in 20m 58s\n- cross-keystone-py36 http://logs.openstack.org/80/659480/1/check/cross-keystone-py36/0fd1501/ : SUCCESS in 25m 45s\n- cross-neutron-py27 http://logs.openstack.org/80/659480/1/check/cross-neutron-py27/ffc36ea/ : SUCCESS in 25m 27s\n- cross-neutron-py36 http://logs.openstack.org/80/659480/1/check/cross-neutron-py36/e839085/ : SUCCESS in 19m 54s\n- cross-nova-py27 http://logs.openstack.org/80/659480/1/check/cross-nova-py27/4388115/ : SUCCESS in 13m 44s\n- cross-nova-py36 http://logs.openstack.org/80/659480/1/check/cross-nova-py36/17a28b4/ : SUCCESS in 10m 26s\n- cross-nova-functional http://logs.openstack.org/80/659480/1/check/cross-nova-functional/1437e4f/ : SUCCESS in 21m 47s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/80/659480/1/check/cross-osc-build-sphinx-docs/fb9655d/html/ : SUCCESS in 5m 26s\n- cross-swift-py27 http://logs.openstack.org/80/659480/1/check/cross-swift-py27/1432960/ : SUCCESS in 7m 32s\n- cross-swift-py36 http://logs.openstack.org/80/659480/1/check/cross-swift-py36/e95dd1a/ : SUCCESS in 7m 34s\n- requirements-integration http://logs.openstack.org/80/659480/1/check/requirements-integration/712f4c2/ : SUCCESS in 46m 15s\n- neutron-grenade-multinode http://logs.openstack.org/80/659480/1/check/neutron-grenade-multinode/1df8730/ : SUCCESS in 1h 09m 48s\n- tempest-full http://logs.openstack.org/80/659480/1/check/tempest-full/dd8af20/ : SUCCESS in 1h 43m 59s\n- neutron-grenade http://logs.openstack.org/80/659480/1/check/neutron-grenade/9842e9b/ : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cbf92e38b45de54fab7fa95640df05cd99457c80","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-07-01 15:26:34.000000000","message":"Abandoned\n\nI\u0027ll retake this patch once we have a stable Pyroute2 version solving the problems we have in ip_lib.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0c3329d09ad53bae22106183594e987203af16f","revisions":{"d0c3329d09ad53bae22106183594e987203af16f":{"kind":"REWORK","_number":1,"created":"2019-05-16 07:47:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/659480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/659480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/659480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/659480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/659480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/659480/1"}}},"commit":{"parents":[{"commit":"2b6e70b464c8b2c646ab45c1bc98ac9005d61a90","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b6e70b464c8b2c646ab45c1bc98ac9005d61a90"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-05-15 21:04:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-05-15 21:06:39.000000000","tz":0},"subject":"Bump Pyroute2 to 0.5.5","message":"Bump Pyroute2 to 0.5.5\n\nChange-Id: Ieaf517f5b244a70821bce2a09e72e9964ff662c4\nRelated-Bug: #1811515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c3329d09ad53bae22106183594e987203af16f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c3329d09ad53bae22106183594e987203af16f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
