)]}'
{"id":"openstack%2Frequirements~659723","triplet_id":"openstack%2Frequirements~master~Ic750b4f61b770d91b19fdd26ce9ffdbed100dca5","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ic750b4f61b770d91b19fdd26ce9ffdbed100dca5","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-05-17 06:13:26.000000000","updated":"2019-05-19 04:30:05.000000000","submitted":"2019-05-19 04:30:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659723-1558240205069-deb1f29f","meta_rev_id":"911ac3188721c27c640903a5e87e7aeca4bab421","_number":659723,"virtual_id_number":659723,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-19 04:30:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-18 18:40:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-18 19:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-18 19:52:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-18 18:40:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-18 19:52:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-19 04:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07009f1bbc98685849e4c75280fea03e6655802b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-17 06:13:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f512b26fe32132071e76dfd2e7dee84bf3fac772","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 08:00:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/23/659723/1/check/openstack-tox-docs/acd9286/html/ : SUCCESS in 3m 01s\n- openstack-tox-pypy http://logs.openstack.org/23/659723/1/check/openstack-tox-pypy/acd5b58/ : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/23/659723/1/check/openstack-tox-pep8/49dd877/ : SUCCESS in 2m 29s\n- openstack-tox-py27 http://logs.openstack.org/23/659723/1/check/openstack-tox-py27/5ac166a/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/23/659723/1/check/openstack-tox-py36/8004741/ : SUCCESS in 2m 51s\n- openstack-tox-validate http://logs.openstack.org/23/659723/1/check/openstack-tox-validate/828ea1e/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/23/659723/1/check/requirements-tox-babel/ccc9c0b/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/23/659723/1/check/requirements-tox-py27-check-uc/543554c/ : FAILURE in 6m 54s\n- requirements-tox-py36-check-uc http://logs.openstack.org/23/659723/1/check/requirements-tox-py36-check-uc/08ea56c/ : FAILURE in 7m 10s\n- cross-cinder-py27 http://logs.openstack.org/23/659723/1/check/cross-cinder-py27/3e3047b/ : SUCCESS in 8m 31s\n- cross-cinder-py36 http://logs.openstack.org/23/659723/1/check/cross-cinder-py36/38bf9ab/ : SUCCESS in 8m 34s\n- cross-glance-py27 http://logs.openstack.org/23/659723/1/check/cross-glance-py27/70ee125/ : SUCCESS in 5m 33s\n- cross-glance-py36 http://logs.openstack.org/23/659723/1/check/cross-glance-py36/5489088/ : SUCCESS in 4m 28s\n- cross-horizon-py27 http://logs.openstack.org/23/659723/1/check/cross-horizon-py27/5f081f4/ : SUCCESS in 10m 26s\n- cross-horizon-py36 http://logs.openstack.org/23/659723/1/check/cross-horizon-py36/002d0b2/ : SUCCESS in 9m 25s\n- cross-horizon-npm http://logs.openstack.org/23/659723/1/check/cross-horizon-npm/dcc2ae1/ : SUCCESS in 5m 21s\n- cross-keystone-py27 http://logs.openstack.org/23/659723/1/check/cross-keystone-py27/1e7fbbf/ : SUCCESS in 21m 48s\n- cross-keystone-py36 http://logs.openstack.org/23/659723/1/check/cross-keystone-py36/a7d4341/ : SUCCESS in 26m 07s\n- cross-neutron-py27 http://logs.openstack.org/23/659723/1/check/cross-neutron-py27/be66463/ : SUCCESS in 23m 25s\n- cross-neutron-py36 http://logs.openstack.org/23/659723/1/check/cross-neutron-py36/36f6cfa/ : SUCCESS in 23m 27s\n- cross-nova-py27 http://logs.openstack.org/23/659723/1/check/cross-nova-py27/c7423a7/ : SUCCESS in 12m 08s\n- cross-nova-py36 http://logs.openstack.org/23/659723/1/check/cross-nova-py36/0092fe0/ : SUCCESS in 14m 12s\n- cross-nova-functional http://logs.openstack.org/23/659723/1/check/cross-nova-functional/f7a06dc/ : SUCCESS in 20m 41s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/23/659723/1/check/cross-osc-build-sphinx-docs/59c4d1c/html/ : SUCCESS in 5m 57s\n- cross-swift-py27 http://logs.openstack.org/23/659723/1/check/cross-swift-py27/4f41db6/ : SUCCESS in 7m 47s\n- cross-swift-py36 http://logs.openstack.org/23/659723/1/check/cross-swift-py36/cc372a7/ : SUCCESS in 8m 00s\n- requirements-integration http://logs.openstack.org/23/659723/1/check/requirements-integration/969ea11/ : SUCCESS in 57m 16s\n- neutron-grenade-multinode http://logs.openstack.org/23/659723/1/check/neutron-grenade-multinode/787381a/ : SUCCESS in 1h 04m 42s\n- tempest-full http://logs.openstack.org/23/659723/1/check/tempest-full/9f6b435/ : FAILURE in 1h 36m 03s\n- neutron-grenade http://logs.openstack.org/23/659723/1/check/neutron-grenade/d552252/ : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":1},{"id":"71969a49a47304ee4fe0b5b80d4a6a7388ae1ad4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-18 06:11:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b491cd5c94703eea4ce647682c7b7d5303bcadf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 07:48:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/23/659723/2/check/openstack-tox-docs/25a3d33/html/ : SUCCESS in 3m 24s\n- openstack-tox-pypy http://logs.openstack.org/23/659723/2/check/openstack-tox-pypy/992e7f5/ : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/23/659723/2/check/openstack-tox-pep8/b3ffdd1/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/23/659723/2/check/openstack-tox-py27/eedc8a9/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/23/659723/2/check/openstack-tox-py36/7c524ec/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/23/659723/2/check/openstack-tox-validate/4931da7/ : SUCCESS in 5m 13s\n- requirements-tox-babel http://logs.openstack.org/23/659723/2/check/requirements-tox-babel/7184485/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/23/659723/2/check/requirements-tox-py27-check-uc/e84ade2/ : FAILURE in 6m 39s\n- requirements-tox-py36-check-uc http://logs.openstack.org/23/659723/2/check/requirements-tox-py36-check-uc/da049cc/ : FAILURE in 9m 01s\n- cross-cinder-py27 http://logs.openstack.org/23/659723/2/check/cross-cinder-py27/c50158a/ : SUCCESS in 8m 25s\n- cross-cinder-py36 http://logs.openstack.org/23/659723/2/check/cross-cinder-py36/409e6ab/ : SUCCESS in 7m 43s\n- cross-glance-py27 http://logs.openstack.org/23/659723/2/check/cross-glance-py27/5302150/ : SUCCESS in 4m 20s\n- cross-glance-py36 http://logs.openstack.org/23/659723/2/check/cross-glance-py36/5021107/ : SUCCESS in 4m 23s\n- cross-horizon-py27 http://logs.openstack.org/23/659723/2/check/cross-horizon-py27/1d3a12b/ : SUCCESS in 11m 23s\n- cross-horizon-py36 http://logs.openstack.org/23/659723/2/check/cross-horizon-py36/d3e8fae/ : SUCCESS in 8m 53s\n- cross-horizon-npm http://logs.openstack.org/23/659723/2/check/cross-horizon-npm/4ea10ba/ : SUCCESS in 5m 48s\n- cross-keystone-py27 http://logs.openstack.org/23/659723/2/check/cross-keystone-py27/78c93b7/ : SUCCESS in 20m 00s\n- cross-keystone-py36 http://logs.openstack.org/23/659723/2/check/cross-keystone-py36/8cb0409/ : SUCCESS in 23m 24s\n- cross-neutron-py27 http://logs.openstack.org/23/659723/2/check/cross-neutron-py27/abcd7ba/ : SUCCESS in 21m 08s\n- cross-neutron-py36 http://logs.openstack.org/23/659723/2/check/cross-neutron-py36/acfb1d2/ : SUCCESS in 23m 01s\n- cross-nova-py27 http://logs.openstack.org/23/659723/2/check/cross-nova-py27/3590891/ : SUCCESS in 11m 49s\n- cross-nova-py36 http://logs.openstack.org/23/659723/2/check/cross-nova-py36/076ffae/ : SUCCESS in 20m 49s\n- cross-nova-functional http://logs.openstack.org/23/659723/2/check/cross-nova-functional/02a8696/ : SUCCESS in 17m 49s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/23/659723/2/check/cross-osc-build-sphinx-docs/fedd720/html/ : SUCCESS in 5m 15s\n- cross-swift-py27 http://logs.openstack.org/23/659723/2/check/cross-swift-py27/2b09c79/ : SUCCESS in 7m 39s\n- cross-swift-py36 http://logs.openstack.org/23/659723/2/check/cross-swift-py36/4fae9b3/ : SUCCESS in 7m 49s\n- requirements-integration http://logs.openstack.org/23/659723/2/check/requirements-integration/e660075/ : SUCCESS in 57m 02s\n- neutron-grenade-multinode http://logs.openstack.org/23/659723/2/check/neutron-grenade-multinode/8900568/ : SUCCESS in 1h 07m 08s\n- tempest-full http://logs.openstack.org/23/659723/2/check/tempest-full/f814fdf/ : FAILURE in 1h 30m 29s\n- neutron-grenade http://logs.openstack.org/23/659723/2/check/neutron-grenade/81bfe8c/ : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0f2d25ab7fdbddae248545f9f9dcc681ba7094ce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-18 17:13:30.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"e479b157486307bcc6bf5a125a5fee8a7cc7a61e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-18 17:33:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4fbd1f6f8f0b1da89c1c3304c47589b208af91e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-18 18:40:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8c1793e2f216e515134683127b46a461d604193d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 19:33:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/23/659723/3/check/openstack-tox-docs/867d8e2/html/ : SUCCESS in 4m 15s\n- openstack-tox-pypy http://logs.openstack.org/23/659723/3/check/openstack-tox-pypy/7ea184c/ : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/23/659723/3/check/openstack-tox-pep8/3acf2d9/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/23/659723/3/check/openstack-tox-py27/c1d8187/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/23/659723/3/check/openstack-tox-py36/6650433/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/23/659723/3/check/openstack-tox-validate/5c61995/ : SUCCESS in 2m 41s\n- requirements-tox-babel http://logs.openstack.org/23/659723/3/check/requirements-tox-babel/e95e430/ : SUCCESS in 2m 24s\n- requirements-tox-py27-check-uc http://logs.openstack.org/23/659723/3/check/requirements-tox-py27-check-uc/ffd0b34/ : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc http://logs.openstack.org/23/659723/3/check/requirements-tox-py36-check-uc/c9df4c2/ : SUCCESS in 6m 33s\n- cross-cinder-py27 http://logs.openstack.org/23/659723/3/check/cross-cinder-py27/3cdcacd/ : SUCCESS in 8m 27s\n- cross-cinder-py36 http://logs.openstack.org/23/659723/3/check/cross-cinder-py36/8b40034/ : SUCCESS in 7m 40s\n- cross-glance-py27 http://logs.openstack.org/23/659723/3/check/cross-glance-py27/c166c23/ : SUCCESS in 4m 06s\n- cross-glance-py36 http://logs.openstack.org/23/659723/3/check/cross-glance-py36/b777637/ : SUCCESS in 4m 13s\n- cross-horizon-py27 http://logs.openstack.org/23/659723/3/check/cross-horizon-py27/8d6295f/ : SUCCESS in 9m 37s\n- cross-horizon-py36 http://logs.openstack.org/23/659723/3/check/cross-horizon-py36/9288686/ : SUCCESS in 7m 22s\n- cross-horizon-npm http://logs.openstack.org/23/659723/3/check/cross-horizon-npm/ecb17de/ : SUCCESS in 5m 48s\n- cross-keystone-py27 http://logs.openstack.org/23/659723/3/check/cross-keystone-py27/2b428d2/ : SUCCESS in 23m 20s\n- cross-keystone-py36 http://logs.openstack.org/23/659723/3/check/cross-keystone-py36/c62515d/ : SUCCESS in 22m 23s\n- cross-neutron-py27 http://logs.openstack.org/23/659723/3/check/cross-neutron-py27/78221bd/ : SUCCESS in 24m 51s\n- cross-neutron-py36 http://logs.openstack.org/23/659723/3/check/cross-neutron-py36/c5ef246/ : SUCCESS in 23m 21s\n- cross-nova-py27 http://logs.openstack.org/23/659723/3/check/cross-nova-py27/560174a/ : SUCCESS in 16m 50s\n- cross-nova-py36 http://logs.openstack.org/23/659723/3/check/cross-nova-py36/2745fbc/ : SUCCESS in 11m 40s\n- cross-nova-functional http://logs.openstack.org/23/659723/3/check/cross-nova-functional/0223a2c/ : SUCCESS in 18m 58s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/23/659723/3/check/cross-osc-build-sphinx-docs/1867a2f/html/ : SUCCESS in 5m 17s\n- cross-swift-py27 http://logs.openstack.org/23/659723/3/check/cross-swift-py27/df507ba/ : SUCCESS in 10m 17s\n- cross-swift-py36 http://logs.openstack.org/23/659723/3/check/cross-swift-py36/e39815f/ : SUCCESS in 7m 46s\n- requirements-integration http://logs.openstack.org/23/659723/3/check/requirements-integration/60a5fc5/ : SUCCESS in 1h 02m 22s\n- neutron-grenade-multinode http://logs.openstack.org/23/659723/3/check/neutron-grenade-multinode/76d445c/ : SUCCESS in 1h 01m 38s\n- tempest-full http://logs.openstack.org/23/659723/3/check/tempest-full/54201f8/ : SUCCESS in 1h 55m 21s\n- neutron-grenade http://logs.openstack.org/23/659723/3/check/neutron-grenade/70b626f/ : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b8a0346d381b9875c286185c1b7d8e9a0be4dc26","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-18 19:52:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f5fc8cb40b82fc3c402bfb4e36f722db6b4dd121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 19:52:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2c1c3b623af9e0f27907b107b01b8536a8a458fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 21:22:05.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/23/659723/3/gate/openstack-tox-docs/3270dbf/html/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/23/659723/3/gate/openstack-tox-pep8/6cfa838/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/23/659723/3/gate/openstack-tox-py27/6b3106e/ : SUCCESS in 2m 58s\n- openstack-tox-py36 http://logs.openstack.org/23/659723/3/gate/openstack-tox-py36/cd89ab4/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/23/659723/3/gate/openstack-tox-validate/83641ea/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/23/659723/3/gate/requirements-tox-babel/4a615e1/ : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc http://logs.openstack.org/23/659723/3/gate/requirements-tox-py27-check-uc/5537579/ : SUCCESS in 6m 09s\n- requirements-tox-py36-check-uc http://logs.openstack.org/23/659723/3/gate/requirements-tox-py36-check-uc/912f0a6/ : SUCCESS in 8m 25s\n- cross-cinder-py27 http://logs.openstack.org/23/659723/3/gate/cross-cinder-py27/b873267/ : SUCCESS in 8m 18s\n- cross-cinder-py36 http://logs.openstack.org/23/659723/3/gate/cross-cinder-py36/0b83743/ : SUCCESS in 8m 13s\n- cross-glance-py27 http://logs.openstack.org/23/659723/3/gate/cross-glance-py27/4efc3d7/ : SUCCESS in 3m 52s\n- cross-glance-py36 http://logs.openstack.org/23/659723/3/gate/cross-glance-py36/1b2e3ee/ : SUCCESS in 4m 28s\n- cross-horizon-py27 http://logs.openstack.org/23/659723/3/gate/cross-horizon-py27/d2de211/ : SUCCESS in 9m 59s\n- cross-horizon-py36 http://logs.openstack.org/23/659723/3/gate/cross-horizon-py36/3be57d8/ : SUCCESS in 7m 25s\n- cross-horizon-npm http://logs.openstack.org/23/659723/3/gate/cross-horizon-npm/406256d/ : SUCCESS in 5m 59s\n- cross-keystone-py27 http://logs.openstack.org/23/659723/3/gate/cross-keystone-py27/e185f3f/ : SUCCESS in 21m 55s\n- cross-keystone-py36 http://logs.openstack.org/23/659723/3/gate/cross-keystone-py36/97d7dde/ : SUCCESS in 23m 15s\n- cross-neutron-py27 http://logs.openstack.org/23/659723/3/gate/cross-neutron-py27/75e339b/ : SUCCESS in 19m 18s\n- cross-neutron-py36 http://logs.openstack.org/23/659723/3/gate/cross-neutron-py36/245858e/ : SUCCESS in 18m 14s\n- cross-nova-py27 http://logs.openstack.org/23/659723/3/gate/cross-nova-py27/302a127/ : SUCCESS in 12m 10s\n- cross-nova-py36 http://logs.openstack.org/23/659723/3/gate/cross-nova-py36/c8d7cb4/ : FAILURE in 24m 57s\n- cross-nova-functional http://logs.openstack.org/23/659723/3/gate/cross-nova-functional/39b31d3/ : SUCCESS in 18m 04s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/23/659723/3/gate/cross-osc-build-sphinx-docs/8d363c3/html/ : SUCCESS in 4m 48s\n- cross-swift-py27 http://logs.openstack.org/23/659723/3/gate/cross-swift-py27/5f427e1/ : SUCCESS in 22m 11s\n- cross-swift-py36 http://logs.openstack.org/23/659723/3/gate/cross-swift-py36/a4b7d6b/ : SUCCESS in 7m 42s\n- requirements-integration http://logs.openstack.org/23/659723/3/gate/requirements-integration/2d66a40/ : SUCCESS in 41m 44s\n- neutron-grenade-multinode http://logs.openstack.org/23/659723/3/gate/neutron-grenade-multinode/a624380/ : SUCCESS in 54m 07s\n- tempest-full http://logs.openstack.org/23/659723/3/gate/tempest-full/be5b33d/ : SUCCESS in 1h 21m 07s\n- neutron-grenade http://logs.openstack.org/23/659723/3/gate/neutron-grenade/830a9b1/ : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":3},{"id":"4104ce9811872f70787f18570187ee047200ffef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-18 21:41:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4b30d24a4a1b0679093e8ea992fa1e93cb73cada","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 23:10:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/23/659723/3/check/openstack-tox-docs/7ce2126/html/ : SUCCESS in 2m 46s\n- openstack-tox-pypy http://logs.openstack.org/23/659723/3/check/openstack-tox-pypy/1d241be/ : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/23/659723/3/check/openstack-tox-pep8/8fded4f/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/23/659723/3/check/openstack-tox-py27/98de96a/ : SUCCESS in 2m 46s\n- openstack-tox-py36 http://logs.openstack.org/23/659723/3/check/openstack-tox-py36/67c7f92/ : SUCCESS in 3m 04s\n- openstack-tox-validate http://logs.openstack.org/23/659723/3/check/openstack-tox-validate/a3eb9cf/ : SUCCESS in 3m 18s\n- requirements-tox-babel http://logs.openstack.org/23/659723/3/check/requirements-tox-babel/ae5592e/ : SUCCESS in 2m 49s\n- requirements-tox-py27-check-uc http://logs.openstack.org/23/659723/3/check/requirements-tox-py27-check-uc/9599ea8/ : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/23/659723/3/check/requirements-tox-py36-check-uc/2ed8d60/ : SUCCESS in 6m 49s\n- cross-cinder-py27 http://logs.openstack.org/23/659723/3/check/cross-cinder-py27/3086f0a/ : FAILURE in 8m 10s\n- cross-cinder-py36 http://logs.openstack.org/23/659723/3/check/cross-cinder-py36/f6b9db1/ : SUCCESS in 7m 16s\n- cross-glance-py27 http://logs.openstack.org/23/659723/3/check/cross-glance-py27/de9112f/ : SUCCESS in 5m 25s\n- cross-glance-py36 http://logs.openstack.org/23/659723/3/check/cross-glance-py36/eaa5534/ : SUCCESS in 3m 17s\n- cross-horizon-py27 http://logs.openstack.org/23/659723/3/check/cross-horizon-py27/be4e730/ : SUCCESS in 8m 35s\n- cross-horizon-py36 http://logs.openstack.org/23/659723/3/check/cross-horizon-py36/9fab176/ : SUCCESS in 7m 13s\n- cross-horizon-npm http://logs.openstack.org/23/659723/3/check/cross-horizon-npm/fd475b9/ : SUCCESS in 5m 06s\n- cross-keystone-py27 http://logs.openstack.org/23/659723/3/check/cross-keystone-py27/59d1158/ : SUCCESS in 20m 54s\n- cross-keystone-py36 http://logs.openstack.org/23/659723/3/check/cross-keystone-py36/71f921c/ : SUCCESS in 21m 03s\n- cross-neutron-py27 http://logs.openstack.org/23/659723/3/check/cross-neutron-py27/33214cd/ : SUCCESS in 21m 04s\n- cross-neutron-py36 http://logs.openstack.org/23/659723/3/check/cross-neutron-py36/b56ead5/ : SUCCESS in 23m 08s\n- cross-nova-py27 http://logs.openstack.org/23/659723/3/check/cross-nova-py27/5a38d13/ : SUCCESS in 12m 20s\n- cross-nova-py36 http://logs.openstack.org/23/659723/3/check/cross-nova-py36/19430ef/ : SUCCESS in 11m 10s\n- cross-nova-functional http://logs.openstack.org/23/659723/3/check/cross-nova-functional/471c200/ : SUCCESS in 17m 48s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/23/659723/3/check/cross-osc-build-sphinx-docs/4f9d7a6/html/ : SUCCESS in 4m 25s\n- cross-swift-py27 http://logs.openstack.org/23/659723/3/check/cross-swift-py27/7bb02b9/ : SUCCESS in 7m 01s\n- cross-swift-py36 http://logs.openstack.org/23/659723/3/check/cross-swift-py36/87db0a7/ : SUCCESS in 7m 00s\n- requirements-integration http://logs.openstack.org/23/659723/3/check/requirements-integration/c73f250/ : SUCCESS in 53m 57s\n- neutron-grenade-multinode http://logs.openstack.org/23/659723/3/check/neutron-grenade-multinode/fb4c578/ : SUCCESS in 1h 02m 32s\n- tempest-full http://logs.openstack.org/23/659723/3/check/tempest-full/2740bd2/ : SUCCESS in 1h 27m 48s\n- neutron-grenade http://logs.openstack.org/23/659723/3/check/neutron-grenade/28edf16/ : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4566f740e9bc77a804d0761012e90ffdfa5b1fec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-19 01:08:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a8f95466f5488134c9d58bafa5c344def07fb26e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 02:40:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/23/659723/3/check/openstack-tox-docs/0a7078f/html/ : SUCCESS in 3m 41s\n- openstack-tox-pypy http://logs.openstack.org/23/659723/3/check/openstack-tox-pypy/25a80af/ : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/23/659723/3/check/openstack-tox-pep8/812bd44/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/23/659723/3/check/openstack-tox-py27/c1230e8/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/23/659723/3/check/openstack-tox-py36/f8d77e8/ : SUCCESS in 4m 05s\n- openstack-tox-validate http://logs.openstack.org/23/659723/3/check/openstack-tox-validate/86eb43c/ : SUCCESS in 3m 40s\n- requirements-tox-babel http://logs.openstack.org/23/659723/3/check/requirements-tox-babel/4fc6d1c/ : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc http://logs.openstack.org/23/659723/3/check/requirements-tox-py27-check-uc/8645411/ : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc http://logs.openstack.org/23/659723/3/check/requirements-tox-py36-check-uc/40ae32b/ : SUCCESS in 5m 36s\n- cross-cinder-py27 http://logs.openstack.org/23/659723/3/check/cross-cinder-py27/0c2064a/ : SUCCESS in 8m 04s\n- cross-cinder-py36 http://logs.openstack.org/23/659723/3/check/cross-cinder-py36/a03b408/ : SUCCESS in 7m 10s\n- cross-glance-py27 http://logs.openstack.org/23/659723/3/check/cross-glance-py27/9b6dd26/ : SUCCESS in 4m 06s\n- cross-glance-py36 http://logs.openstack.org/23/659723/3/check/cross-glance-py36/2e68d5d/ : SUCCESS in 4m 02s\n- cross-horizon-py27 http://logs.openstack.org/23/659723/3/check/cross-horizon-py27/8095833/ : SUCCESS in 11m 29s\n- cross-horizon-py36 http://logs.openstack.org/23/659723/3/check/cross-horizon-py36/5c080df/ : SUCCESS in 7m 15s\n- cross-horizon-npm http://logs.openstack.org/23/659723/3/check/cross-horizon-npm/8b97516/ : SUCCESS in 5m 14s\n- cross-keystone-py27 http://logs.openstack.org/23/659723/3/check/cross-keystone-py27/a81a864/ : SUCCESS in 20m 59s\n- cross-keystone-py36 http://logs.openstack.org/23/659723/3/check/cross-keystone-py36/fd24da8/ : SUCCESS in 23m 59s\n- cross-neutron-py27 http://logs.openstack.org/23/659723/3/check/cross-neutron-py27/85216d6/ : SUCCESS in 19m 53s\n- cross-neutron-py36 http://logs.openstack.org/23/659723/3/check/cross-neutron-py36/74a08b0/ : SUCCESS in 21m 01s\n- cross-nova-py27 http://logs.openstack.org/23/659723/3/check/cross-nova-py27/34636bd/ : SUCCESS in 10m 24s\n- cross-nova-py36 http://logs.openstack.org/23/659723/3/check/cross-nova-py36/7802adc/ : SUCCESS in 10m 34s\n- cross-nova-functional http://logs.openstack.org/23/659723/3/check/cross-nova-functional/5f8ee41/ : SUCCESS in 17m 22s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/23/659723/3/check/cross-osc-build-sphinx-docs/3c9755d/html/ : SUCCESS in 5m 07s\n- cross-swift-py27 http://logs.openstack.org/23/659723/3/check/cross-swift-py27/b6056f1/ : SUCCESS in 7m 15s\n- cross-swift-py36 http://logs.openstack.org/23/659723/3/check/cross-swift-py36/90cc449/ : SUCCESS in 11m 43s\n- requirements-integration http://logs.openstack.org/23/659723/3/check/requirements-integration/e0d2449/ : SUCCESS in 48m 56s\n- neutron-grenade-multinode http://logs.openstack.org/23/659723/3/check/neutron-grenade-multinode/b0a2db5/ : SUCCESS in 1h 04m 20s\n- tempest-full http://logs.openstack.org/23/659723/3/check/tempest-full/9b753eb/ : SUCCESS in 1h 30m 15s\n- neutron-grenade http://logs.openstack.org/23/659723/3/check/neutron-grenade/32b8125/ : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":3},{"id":"cb3df42a5431d530e22c9722df81ab2ed2e2c3b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 02:40:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"73735ff3254dcd00057c8004599aa86c606bf67f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 04:30:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/23/659723/3/gate/openstack-tox-docs/96c8cb9/html/ : SUCCESS in 2m 52s\n- openstack-tox-pep8 http://logs.openstack.org/23/659723/3/gate/openstack-tox-pep8/d372ba5/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/23/659723/3/gate/openstack-tox-py27/3898aff/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/23/659723/3/gate/openstack-tox-py36/94f68ea/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/23/659723/3/gate/openstack-tox-validate/2a1a342/ : SUCCESS in 3m 47s\n- requirements-tox-babel http://logs.openstack.org/23/659723/3/gate/requirements-tox-babel/ca01e88/ : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc http://logs.openstack.org/23/659723/3/gate/requirements-tox-py27-check-uc/691b778/ : SUCCESS in 8m 09s\n- requirements-tox-py36-check-uc http://logs.openstack.org/23/659723/3/gate/requirements-tox-py36-check-uc/56040c9/ : SUCCESS in 7m 00s\n- cross-cinder-py27 http://logs.openstack.org/23/659723/3/gate/cross-cinder-py27/1b5dee0/ : SUCCESS in 8m 56s\n- cross-cinder-py36 http://logs.openstack.org/23/659723/3/gate/cross-cinder-py36/10c2f5b/ : SUCCESS in 7m 19s\n- cross-glance-py27 http://logs.openstack.org/23/659723/3/gate/cross-glance-py27/5c586e1/ : SUCCESS in 5m 03s\n- cross-glance-py36 http://logs.openstack.org/23/659723/3/gate/cross-glance-py36/8ae5142/ : SUCCESS in 5m 02s\n- cross-horizon-py27 http://logs.openstack.org/23/659723/3/gate/cross-horizon-py27/e844997/ : SUCCESS in 10m 12s\n- cross-horizon-py36 http://logs.openstack.org/23/659723/3/gate/cross-horizon-py36/133fb66/ : SUCCESS in 9m 17s\n- cross-horizon-npm http://logs.openstack.org/23/659723/3/gate/cross-horizon-npm/6996bc9/ : SUCCESS in 5m 32s\n- cross-keystone-py27 http://logs.openstack.org/23/659723/3/gate/cross-keystone-py27/6c64719/ : SUCCESS in 18m 52s\n- cross-keystone-py36 http://logs.openstack.org/23/659723/3/gate/cross-keystone-py36/60a3b01/ : SUCCESS in 22m 37s\n- cross-neutron-py27 http://logs.openstack.org/23/659723/3/gate/cross-neutron-py27/0ff0d8c/ : SUCCESS in 23m 11s\n- cross-neutron-py36 http://logs.openstack.org/23/659723/3/gate/cross-neutron-py36/260dac4/ : SUCCESS in 21m 13s\n- cross-nova-py27 http://logs.openstack.org/23/659723/3/gate/cross-nova-py27/363bdb0/ : SUCCESS in 14m 14s\n- cross-nova-py36 http://logs.openstack.org/23/659723/3/gate/cross-nova-py36/b63b455/ : SUCCESS in 10m 45s\n- cross-nova-functional http://logs.openstack.org/23/659723/3/gate/cross-nova-functional/c305045/ : SUCCESS in 17m 41s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/23/659723/3/gate/cross-osc-build-sphinx-docs/4d6fa54/html/ : SUCCESS in 5m 04s\n- cross-swift-py27 http://logs.openstack.org/23/659723/3/gate/cross-swift-py27/fcc09f6/ : SUCCESS in 10m 39s\n- cross-swift-py36 http://logs.openstack.org/23/659723/3/gate/cross-swift-py36/26343f5/ : SUCCESS in 7m 55s\n- requirements-integration http://logs.openstack.org/23/659723/3/gate/requirements-integration/3137465/ : SUCCESS in 48m 29s\n- neutron-grenade-multinode http://logs.openstack.org/23/659723/3/gate/neutron-grenade-multinode/7deb2f2/ : SUCCESS in 1h 04m 51s\n- tempest-full http://logs.openstack.org/23/659723/3/gate/tempest-full/98a4821/ : SUCCESS in 1h 44m 36s\n- neutron-grenade http://logs.openstack.org/23/659723/3/gate/neutron-grenade/9ab87ed/ : SUCCESS in 48m 22s","accounts_in_message":[],"_revision_number":3},{"id":"911ac3188721c27c640903a5e87e7aeca4bab421","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 04:30:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aadd6dccab3082fa5373d9ba97fb561930cc31ff","revisions":{"d0efa1acd95fca171286a2fddcf7679505d5e7c4":{"kind":"REWORK","_number":1,"created":"2019-05-17 06:13:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/659723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/659723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/659723/1"}}},"commit":{"parents":[{"commit":"3c3e972b7a1231f0ce625276e2d62f974fca3e01","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c3e972b7a1231f0ce625276e2d62f974fca3e01"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-17 06:13:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-17 06:13:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic750b4f61b770d91b19fdd26ce9ffdbed100dca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0efa1acd95fca171286a2fddcf7679505d5e7c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0efa1acd95fca171286a2fddcf7679505d5e7c4"}]},"branch":"refs/heads/master"},"ebbf2955ddc7fea0bbc291b7442050968998664e":{"kind":"REWORK","_number":2,"created":"2019-05-18 06:11:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/659723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/659723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/659723/2"}}},"commit":{"parents":[{"commit":"cfe0542a65f5557e8f22e2895772b0f08f0b58c2","subject":"Merge \"Exclude broken ironicclient versions 2.5.2 and 2.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfe0542a65f5557e8f22e2895772b0f08f0b58c2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-18 06:11:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-18 06:11:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic750b4f61b770d91b19fdd26ce9ffdbed100dca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebbf2955ddc7fea0bbc291b7442050968998664e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebbf2955ddc7fea0bbc291b7442050968998664e"}]},"branch":"refs/heads/master"},"aadd6dccab3082fa5373d9ba97fb561930cc31ff":{"kind":"REWORK","_number":3,"created":"2019-05-18 17:33:25.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/23/659723/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/659723/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/659723/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/659723/3"}}},"commit":{"parents":[{"commit":"cfe0542a65f5557e8f22e2895772b0f08f0b58c2","subject":"Merge \"Exclude broken ironicclient versions 2.5.2 and 2.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfe0542a65f5557e8f22e2895772b0f08f0b58c2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-18 06:11:24.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-18 17:32:56.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic750b4f61b770d91b19fdd26ce9ffdbed100dca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aadd6dccab3082fa5373d9ba97fb561930cc31ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aadd6dccab3082fa5373d9ba97fb561930cc31ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
