)]}'
{"id":"openstack%2Frequirements~659733","triplet_id":"openstack%2Frequirements~stable%2Fstein~I0331d061c13f5e2e109443c5210e0924b73ce0be","project":"openstack/requirements","branch":"stable/stein","topic":"ironicclient","hashtags":[],"change_id":"I0331d061c13f5e2e109443c5210e0924b73ce0be","subject":"Exclude broken ironicclient versions 2.5.2 and 2.7.1","status":"MERGED","created":"2019-05-17 07:36:29.000000000","updated":"2019-05-17 14:02:05.000000000","submitted":"2019-05-17 14:02:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659733-1558101725715-38c90c1d","meta_rev_id":"026c95a9f00e065a48cca8d8a1a8553caec98be1","_number":659733,"virtual_id_number":659733,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-05-17 14:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":2,"date":"2019-05-17 11:56:34.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2019-05-17 07:45:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":1,"date":"2019-05-17 11:56:34.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 07:45:24.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-05-17 11:56:34.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-17 14:02:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4c0eadeee76fd5ff847c36554579d9a5b5ae6f4","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-17 07:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84b8fa8d14e7273bbd73e7567e56e8a0ac405e27","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-17 07:45:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e25c3f669c9e556a9cbebfb477a363628b4868d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 09:24:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/33/659733/1/check/openstack-tox-pypy/6ffb96b/ : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/33/659733/1/check/openstack-tox-pep8/05a2dfc/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/33/659733/1/check/openstack-tox-py27/4386793/ : SUCCESS in 3m 54s\n- openstack-tox-py35 http://logs.openstack.org/33/659733/1/check/openstack-tox-py35/9b662c6/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/33/659733/1/check/openstack-tox-py36/6669963/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/33/659733/1/check/openstack-tox-validate/2fd18b3/ : SUCCESS in 2m 32s\n- requirements-tox-babel http://logs.openstack.org/33/659733/1/check/requirements-tox-babel/494cbb6/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/33/659733/1/check/requirements-tox-py27-check-uc/0a2dda0/ : SUCCESS in 6m 05s\n- requirements-tox-py35-check-uc http://logs.openstack.org/33/659733/1/check/requirements-tox-py35-check-uc/dda2404/ : SUCCESS in 8m 16s\n- requirements-tox-py36-check-uc http://logs.openstack.org/33/659733/1/check/requirements-tox-py36-check-uc/4df7342/ : SUCCESS in 6m 03s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/33/659733/1/check/cross-cinder-py27/97b7a65/ : SUCCESS in 10m 13s\n- cross-cinder-py35 http://logs.openstack.org/33/659733/1/check/cross-cinder-py35/faa5937/ : SUCCESS in 8m 27s\n- cross-glance-py27 http://logs.openstack.org/33/659733/1/check/cross-glance-py27/8686114/ : SUCCESS in 5m 04s\n- cross-glance-py35 http://logs.openstack.org/33/659733/1/check/cross-glance-py35/c95abd5/ : SUCCESS in 4m 03s\n- cross-horizon-py27 http://logs.openstack.org/33/659733/1/check/cross-horizon-py27/5357f78/ : SUCCESS in 9m 56s\n- cross-horizon-py35 http://logs.openstack.org/33/659733/1/check/cross-horizon-py35/b3645c2/ : SUCCESS in 8m 44s\n- cross-horizon-npm http://logs.openstack.org/33/659733/1/check/cross-horizon-npm/8999515/ : SUCCESS in 5m 36s\n- cross-keystone-py27 http://logs.openstack.org/33/659733/1/check/cross-keystone-py27/277bb43/ : SUCCESS in 18m 40s\n- cross-keystone-py35 http://logs.openstack.org/33/659733/1/check/cross-keystone-py35/185da17/ : SUCCESS in 22m 39s\n- cross-neutron-py27 http://logs.openstack.org/33/659733/1/check/cross-neutron-py27/c92a8bf/ : SUCCESS in 17m 55s\n- cross-neutron-py35 http://logs.openstack.org/33/659733/1/check/cross-neutron-py35/542f5d1/ : SUCCESS in 23m 39s\n- cross-nova-py27 http://logs.openstack.org/33/659733/1/check/cross-nova-py27/d12e9fa/ : SUCCESS in 12m 57s\n- cross-nova-py35 http://logs.openstack.org/33/659733/1/check/cross-nova-py35/4740389/ : SUCCESS in 14m 06s\n- cross-nova-functional http://logs.openstack.org/33/659733/1/check/cross-nova-functional/c4f4097/ : SUCCESS in 18m 06s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/33/659733/1/check/cross-osc-build-sphinx-docs/1b80c0e/html/ : SUCCESS in 6m 01s\n- cross-swift-py27 http://logs.openstack.org/33/659733/1/check/cross-swift-py27/e9189fe/ : SUCCESS in 7m 50s\n- cross-swift-py35 http://logs.openstack.org/33/659733/1/check/cross-swift-py35/52a32aa/ : SUCCESS in 7m 37s\n- requirements-integration http://logs.openstack.org/33/659733/1/check/requirements-integration/b2f0165/ : SUCCESS in 50m 51s\n- neutron-grenade-multinode http://logs.openstack.org/33/659733/1/check/neutron-grenade-multinode/403adcf/ : SUCCESS in 1h 05m 49s\n- tempest-full http://logs.openstack.org/33/659733/1/check/tempest-full/753fb79/ : SUCCESS in 1h 42m 32s\n- neutron-grenade http://logs.openstack.org/33/659733/1/check/neutron-grenade/8639380/ : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":1},{"id":"94b0f31009c8c4e77b24a5ab761544d66722af23","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-05-17 11:56:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nfast-approving to unblock ironic gates","accounts_in_message":[],"_revision_number":1},{"id":"9cf268128a667997c087527203216d34af1e2c2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 11:56:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a740cf48b832f49e290ede3075c8f9d9087f7d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 14:02:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"08c9a4ae2f5fe7826de2c4144366f398998ab47e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 14:02:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/33/659733/1/gate/openstack-tox-pep8/bad640d/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/33/659733/1/gate/openstack-tox-py27/262d545/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/33/659733/1/gate/openstack-tox-py35/06032b9/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/33/659733/1/gate/openstack-tox-py36/50f7498/ : SUCCESS in 3m 15s\n- openstack-tox-validate http://logs.openstack.org/33/659733/1/gate/openstack-tox-validate/7088754/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/33/659733/1/gate/requirements-tox-babel/bc6966a/ : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/33/659733/1/gate/requirements-tox-py27-check-uc/d60105b/ : SUCCESS in 6m 01s\n- requirements-tox-py35-check-uc http://logs.openstack.org/33/659733/1/gate/requirements-tox-py35-check-uc/8114177/ : SUCCESS in 6m 57s\n- cross-cinder-py27 http://logs.openstack.org/33/659733/1/gate/cross-cinder-py27/77e21e6/ : SUCCESS in 8m 05s\n- cross-cinder-py35 http://logs.openstack.org/33/659733/1/gate/cross-cinder-py35/b1fee79/ : SUCCESS in 10m 06s\n- cross-glance-py27 http://logs.openstack.org/33/659733/1/gate/cross-glance-py27/ff01198/ : SUCCESS in 4m 52s\n- cross-glance-py35 http://logs.openstack.org/33/659733/1/gate/cross-glance-py35/e23503a/ : SUCCESS in 4m 04s\n- cross-horizon-py27 http://logs.openstack.org/33/659733/1/gate/cross-horizon-py27/f19bea9/ : SUCCESS in 10m 10s\n- cross-horizon-py35 http://logs.openstack.org/33/659733/1/gate/cross-horizon-py35/94821be/ : SUCCESS in 9m 37s\n- cross-horizon-npm http://logs.openstack.org/33/659733/1/gate/cross-horizon-npm/049e1f1/ : SUCCESS in 6m 51s\n- cross-keystone-py27 http://logs.openstack.org/33/659733/1/gate/cross-keystone-py27/cb9d258/ : SUCCESS in 23m 52s\n- cross-keystone-py35 http://logs.openstack.org/33/659733/1/gate/cross-keystone-py35/c4023ac/ : SUCCESS in 23m 03s\n- cross-neutron-py27 http://logs.openstack.org/33/659733/1/gate/cross-neutron-py27/5ea4a57/ : SUCCESS in 20m 12s\n- cross-neutron-py35 http://logs.openstack.org/33/659733/1/gate/cross-neutron-py35/6982c49/ : SUCCESS in 26m 16s\n- cross-nova-py27 http://logs.openstack.org/33/659733/1/gate/cross-nova-py27/672a9df/ : SUCCESS in 13m 27s\n- cross-nova-py35 http://logs.openstack.org/33/659733/1/gate/cross-nova-py35/132afa8/ : SUCCESS in 12m 22s\n- cross-nova-functional http://logs.openstack.org/33/659733/1/gate/cross-nova-functional/1066a3d/ : SUCCESS in 19m 25s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/33/659733/1/gate/cross-osc-build-sphinx-docs/0980e80/html/ : SUCCESS in 5m 07s\n- cross-swift-py27 http://logs.openstack.org/33/659733/1/gate/cross-swift-py27/022ce10/ : SUCCESS in 7m 40s\n- cross-swift-py35 http://logs.openstack.org/33/659733/1/gate/cross-swift-py35/521f1b9/ : SUCCESS in 7m 37s\n- requirements-integration http://logs.openstack.org/33/659733/1/gate/requirements-integration/cf9d4a7/ : SUCCESS in 47m 17s\n- neutron-grenade-multinode http://logs.openstack.org/33/659733/1/gate/neutron-grenade-multinode/a657be8/ : SUCCESS in 1h 10m 22s\n- tempest-full http://logs.openstack.org/33/659733/1/gate/tempest-full/20c68cd/ : SUCCESS in 1h 46m 55s\n- neutron-grenade http://logs.openstack.org/33/659733/1/gate/neutron-grenade/e4591ff/ : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df8a8a16f65b72ed17e24c2dce9e62d5f8774b7d","revisions":{"df8a8a16f65b72ed17e24c2dce9e62d5f8774b7d":{"kind":"REWORK","_number":1,"created":"2019-05-17 07:36:29.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/33/659733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/659733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/659733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/659733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/659733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/659733/1"}}},"commit":{"parents":[{"commit":"913f788afc13a9b162b2dbcbcfa3a376fca872cc","subject":"Merge \"Add safety check output to the linters output\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/913f788afc13a9b162b2dbcbcfa3a376fca872cc"}]}],"author":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-05-16 16:06:35.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-05-17 07:35:53.000000000","tz":120},"subject":"Exclude broken ironicclient versions 2.5.2 and 2.7.1","message":"Exclude broken ironicclient versions 2.5.2 and 2.7.1\n\nA change was merged to these versions that somehow broke Nova accessing\nIronic. While a fix has been posted in https://review.opendev.org/#/c/659592/\nwe should block these releases to unbreak the CI for now.\n\nChange-Id: I0331d061c13f5e2e109443c5210e0924b73ce0be\nStory: #2005723\nTask: #31055\n(cherry picked from commit ba7f936fd797eba30946f3f263e6050181905988)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8a8a16f65b72ed17e24c2dce9e62d5f8774b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8a8a16f65b72ed17e24c2dce9e62d5f8774b7d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
