)]}'
{"id":"openstack%2Frequirements~659734","triplet_id":"openstack%2Frequirements~stable%2Frocky~I0331d061c13f5e2e109443c5210e0924b73ce0be","project":"openstack/requirements","branch":"stable/rocky","topic":"ironicclient","hashtags":[],"change_id":"I0331d061c13f5e2e109443c5210e0924b73ce0be","subject":"Exclude broken ironicclient versions 2.5.2 and 2.7.1","status":"MERGED","created":"2019-05-17 07:37:50.000000000","updated":"2019-05-17 17:27:26.000000000","submitted":"2019-05-17 17:27:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659734-1558114046873-d1a6a7d1","meta_rev_id":"faf45dfa315687436ae68b20f1d3b174a4ce67db","_number":659734,"virtual_id_number":659734,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-17 15:32:52.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-05-17 17:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-17 14:04:52.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-17 11:56:43.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2019-05-17 07:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-17 11:56:43.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 07:45:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-05-17 11:56:43.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-17 15:32:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-17 17:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d521e82d7f811b984205c3f09ea98a3279a8739","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-17 07:37:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01c52645249e0668b46d72fbf6527e675946bf9c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-17 07:45:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5c921ed36ffe7571af379c86c4117a7dfc8560d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 09:04:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/34/659734/1/check/openstack-tox-pypy/0e66eb5/ : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/34/659734/1/check/openstack-tox-pep8/14f3f7d/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/34/659734/1/check/openstack-tox-py27/07098a9/ : SUCCESS in 3m 17s\n- openstack-tox-py35 http://logs.openstack.org/34/659734/1/check/openstack-tox-py35/509a11c/ : SUCCESS in 3m 28s\n- openstack-tox-validate http://logs.openstack.org/34/659734/1/check/openstack-tox-validate/3ba6d67/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/34/659734/1/check/requirements-tox-babel/d7e11f8/ : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/34/659734/1/check/requirements-tox-py27-check-uc/f278768/ : SUCCESS in 10m 17s\n- requirements-tox-py35-check-uc http://logs.openstack.org/34/659734/1/check/requirements-tox-py35-check-uc/c09e49f/ : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc http://logs.openstack.org/34/659734/1/check/requirements-tox-py36-check-uc/4c4c0f4/ : SUCCESS in 5m 43s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/34/659734/1/check/cross-cinder-py27/19ff566/ : SUCCESS in 8m 45s\n- cross-cinder-py35 http://logs.openstack.org/34/659734/1/check/cross-cinder-py35/b62b92f/ : SUCCESS in 9m 09s\n- cross-glance-py27 http://logs.openstack.org/34/659734/1/check/cross-glance-py27/ed437c0/ : SUCCESS in 4m 13s\n- cross-glance-py35 http://logs.openstack.org/34/659734/1/check/cross-glance-py35/f26c847/ : SUCCESS in 4m 13s\n- cross-horizon-py27 http://logs.openstack.org/34/659734/1/check/cross-horizon-py27/1bfd4c8/ : SUCCESS in 9m 10s\n- cross-horizon-py35 http://logs.openstack.org/34/659734/1/check/cross-horizon-py35/4da4907/ : SUCCESS in 9m 03s\n- cross-keystone-py27 http://logs.openstack.org/34/659734/1/check/cross-keystone-py27/7975a05/ : SUCCESS in 12m 08s\n- cross-keystone-py35 http://logs.openstack.org/34/659734/1/check/cross-keystone-py35/cc06235/ : SUCCESS in 11m 40s\n- cross-neutron-py27 http://logs.openstack.org/34/659734/1/check/cross-neutron-py27/68f9d98/ : SUCCESS in 23m 50s\n- cross-neutron-py35 http://logs.openstack.org/34/659734/1/check/cross-neutron-py35/25e450f/ : SUCCESS in 24m 27s\n- cross-nova-py27 http://logs.openstack.org/34/659734/1/check/cross-nova-py27/981bf10/ : SUCCESS in 12m 18s\n- cross-nova-py35 http://logs.openstack.org/34/659734/1/check/cross-nova-py35/c185caf/ : SUCCESS in 11m 45s\n- cross-nova-functional http://logs.openstack.org/34/659734/1/check/cross-nova-functional/9f57a7c/ : SUCCESS in 17m 25s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/34/659734/1/check/cross-osc-build-sphinx-docs/ae3489f/html/ : SUCCESS in 4m 39s\n- cross-swift-py27 http://logs.openstack.org/34/659734/1/check/cross-swift-py27/b0d847f/ : SUCCESS in 7m 33s\n- cross-swift-py35 http://logs.openstack.org/34/659734/1/check/cross-swift-py35/3d954e9/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/34/659734/1/check/openstack-tox-py36/4d92a71/ : SUCCESS in 3m 01s\n- requirements-integration http://logs.openstack.org/34/659734/1/check/requirements-integration/8d6113f/ : SUCCESS in 54m 14s\n- neutron-grenade-multinode http://logs.openstack.org/34/659734/1/check/neutron-grenade-multinode/6c59799/ : SUCCESS in 1h 01m 04s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/34/659734/1/check/legacy-tempest-dsvm-neutron-full/40ee932/ : SUCCESS in 1h 21m 12s\n- neutron-grenade http://logs.openstack.org/34/659734/1/check/neutron-grenade/9093d17/ : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6ec6e7ed20e02f398810c5a8ba523ff4b145faea","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-05-17 11:56:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nfast-approving to unblock ironic gates","accounts_in_message":[],"_revision_number":1},{"id":"d9ff23c334670f13bcfb4ca0b8a713ff22c6013a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 11:56:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75ce412620857e9e14c33d0c5bfb1f254bd7607f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 14:02:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/659734/1/gate/openstack-tox-pep8/67e17ac/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/34/659734/1/gate/openstack-tox-py27/938b97c/ : SUCCESS in 4m 22s\n- openstack-tox-py35 http://logs.openstack.org/34/659734/1/gate/openstack-tox-py35/b9a4982/ : SUCCESS in 3m 54s\n- openstack-tox-validate http://logs.openstack.org/34/659734/1/gate/openstack-tox-validate/23da038/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/34/659734/1/gate/requirements-tox-babel/bbd7cf4/ : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc http://logs.openstack.org/34/659734/1/gate/requirements-tox-py27-check-uc/27b2ce1/ : SUCCESS in 6m 11s\n- requirements-tox-py35-check-uc http://logs.openstack.org/34/659734/1/gate/requirements-tox-py35-check-uc/46c95f1/ : SUCCESS in 6m 03s\n- cross-cinder-py27 http://logs.openstack.org/34/659734/1/gate/cross-cinder-py27/b7609b3/ : SUCCESS in 8m 26s\n- cross-cinder-py35 http://logs.openstack.org/34/659734/1/gate/cross-cinder-py35/98a109b/ : SUCCESS in 8m 04s\n- cross-glance-py27 http://logs.openstack.org/34/659734/1/gate/cross-glance-py27/ce12791/ : SUCCESS in 3m 42s\n- cross-glance-py35 http://logs.openstack.org/34/659734/1/gate/cross-glance-py35/c5873da/ : FAILURE in 4m 46s\n- cross-horizon-py27 http://logs.openstack.org/34/659734/1/gate/cross-horizon-py27/00f0086/ : SUCCESS in 8m 48s\n- cross-horizon-py35 http://logs.openstack.org/34/659734/1/gate/cross-horizon-py35/1f6690b/ : SUCCESS in 8m 31s\n- cross-keystone-py27 http://logs.openstack.org/34/659734/1/gate/cross-keystone-py27/b580232/ : SUCCESS in 19m 20s\n- cross-keystone-py35 http://logs.openstack.org/34/659734/1/gate/cross-keystone-py35/616ed88/ : SUCCESS in 11m 16s\n- cross-neutron-py27 http://logs.openstack.org/34/659734/1/gate/cross-neutron-py27/f0697fc/ : SUCCESS in 23m 27s\n- cross-neutron-py35 http://logs.openstack.org/34/659734/1/gate/cross-neutron-py35/787023d/ : SUCCESS in 26m 06s\n- cross-nova-py27 http://logs.openstack.org/34/659734/1/gate/cross-nova-py27/436e10e/ : SUCCESS in 12m 33s\n- cross-nova-py35 http://logs.openstack.org/34/659734/1/gate/cross-nova-py35/a6442bb/ : SUCCESS in 15m 21s\n- cross-nova-functional http://logs.openstack.org/34/659734/1/gate/cross-nova-functional/e75444f/ : SUCCESS in 15m 32s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/34/659734/1/gate/cross-osc-build-sphinx-docs/818f71a/html/ : SUCCESS in 6m 02s\n- cross-swift-py27 http://logs.openstack.org/34/659734/1/gate/cross-swift-py27/79aaf94/ : SUCCESS in 8m 20s\n- cross-swift-py35 http://logs.openstack.org/34/659734/1/gate/cross-swift-py35/6f5023e/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/34/659734/1/gate/openstack-tox-py36/cb6a49c/ : SUCCESS in 3m 56s\n- requirements-integration http://logs.openstack.org/34/659734/1/gate/requirements-integration/6e2066f/ : SUCCESS in 54m 38s\n- neutron-grenade-multinode http://logs.openstack.org/34/659734/1/gate/neutron-grenade-multinode/5125e27/ : SUCCESS in 1h 08m 14s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/34/659734/1/gate/legacy-tempest-dsvm-neutron-full/c943a37/ : SUCCESS in 1h 25m 31s\n- neutron-grenade http://logs.openstack.org/34/659734/1/gate/neutron-grenade/9ec8f8a/ : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8b900aa15cd2423249f1fe1d0b09176f044ff157","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-17 14:04:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d36eb31d542b50b161f7b94c4736065cac0fa8c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-17 15:32:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05dfbce221a9e42be928f0830628466916a64b9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 15:39:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/34/659734/1/check/openstack-tox-pypy/0209578/ : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/34/659734/1/check/openstack-tox-pep8/202009b/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/34/659734/1/check/openstack-tox-py27/4efa9ae/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/34/659734/1/check/openstack-tox-py35/c18fcac/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/34/659734/1/check/openstack-tox-validate/a14a72d/ : SUCCESS in 3m 33s\n- requirements-tox-babel http://logs.openstack.org/34/659734/1/check/requirements-tox-babel/5425f88/ : SUCCESS in 2m 58s\n- requirements-tox-py27-check-uc http://logs.openstack.org/34/659734/1/check/requirements-tox-py27-check-uc/164e458/ : SUCCESS in 6m 50s\n- requirements-tox-py35-check-uc http://logs.openstack.org/34/659734/1/check/requirements-tox-py35-check-uc/aeee1f5/ : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc http://logs.openstack.org/34/659734/1/check/requirements-tox-py36-check-uc/c290427/ : SUCCESS in 5m 58s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/34/659734/1/check/cross-cinder-py27/491ea2e/ : SUCCESS in 7m 57s\n- cross-cinder-py35 http://logs.openstack.org/34/659734/1/check/cross-cinder-py35/81ab933/ : SUCCESS in 9m 05s\n- cross-glance-py27 http://logs.openstack.org/34/659734/1/check/cross-glance-py27/77eda5e/ : SUCCESS in 4m 28s\n- cross-glance-py35 http://logs.openstack.org/34/659734/1/check/cross-glance-py35/93682c5/ : SUCCESS in 3m 50s\n- cross-horizon-py27 http://logs.openstack.org/34/659734/1/check/cross-horizon-py27/f50108e/ : SUCCESS in 11m 05s\n- cross-horizon-py35 http://logs.openstack.org/34/659734/1/check/cross-horizon-py35/ed140f5/ : SUCCESS in 9m 50s\n- cross-keystone-py27 http://logs.openstack.org/34/659734/1/check/cross-keystone-py27/e80d624/ : SUCCESS in 14m 25s\n- cross-keystone-py35 http://logs.openstack.org/34/659734/1/check/cross-keystone-py35/995cb01/ : SUCCESS in 21m 38s\n- cross-neutron-py27 http://logs.openstack.org/34/659734/1/check/cross-neutron-py27/02ea032/ : SUCCESS in 22m 58s\n- cross-neutron-py35 http://logs.openstack.org/34/659734/1/check/cross-neutron-py35/72d33ed/ : SUCCESS in 25m 15s\n- cross-nova-py27 http://logs.openstack.org/34/659734/1/check/cross-nova-py27/2ac385e/ : SUCCESS in 17m 28s\n- cross-nova-py35 http://logs.openstack.org/34/659734/1/check/cross-nova-py35/78448aa/ : SUCCESS in 18m 44s\n- cross-nova-functional http://logs.openstack.org/34/659734/1/check/cross-nova-functional/efde3ad/ : SUCCESS in 23m 34s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/34/659734/1/check/cross-osc-build-sphinx-docs/d172f78/html/ : SUCCESS in 5m 00s\n- cross-swift-py27 http://logs.openstack.org/34/659734/1/check/cross-swift-py27/84ead46/ : SUCCESS in 7m 07s\n- cross-swift-py35 http://logs.openstack.org/34/659734/1/check/cross-swift-py35/3155611/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/34/659734/1/check/openstack-tox-py36/02dcac5/ : SUCCESS in 4m 07s\n- requirements-integration http://logs.openstack.org/34/659734/1/check/requirements-integration/66d4cea/ : SUCCESS in 57m 20s\n- neutron-grenade-multinode http://logs.openstack.org/34/659734/1/check/neutron-grenade-multinode/1eaf03c/ : SUCCESS in 1h 03m 04s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/34/659734/1/check/legacy-tempest-dsvm-neutron-full/24d10d1/ : SUCCESS in 1h 28m 33s\n- neutron-grenade http://logs.openstack.org/34/659734/1/check/neutron-grenade/fc91213/ : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b86bf7927484bbea2cd487493a9b1fbf8339e31a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 15:39:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46cd5567d44405b50dd1c0e154b12678b3cf56ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 17:27:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"72ab3c4d7ecba302f426d1b5bbe48e47f0aa4863","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 17:27:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/34/659734/1/gate/openstack-tox-pep8/77e8bfe/ : SUCCESS in 2m 39s\n- openstack-tox-py27 http://logs.openstack.org/34/659734/1/gate/openstack-tox-py27/c82e88e/ : SUCCESS in 2m 51s\n- openstack-tox-py35 http://logs.openstack.org/34/659734/1/gate/openstack-tox-py35/ba04ccb/ : SUCCESS in 3m 08s\n- openstack-tox-validate http://logs.openstack.org/34/659734/1/gate/openstack-tox-validate/6fcafd0/ : SUCCESS in 2m 50s\n- requirements-tox-babel http://logs.openstack.org/34/659734/1/gate/requirements-tox-babel/1f97901/ : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/34/659734/1/gate/requirements-tox-py27-check-uc/9c401b8/ : SUCCESS in 9m 04s\n- requirements-tox-py35-check-uc http://logs.openstack.org/34/659734/1/gate/requirements-tox-py35-check-uc/a31fbd7/ : SUCCESS in 7m 04s\n- cross-cinder-py27 http://logs.openstack.org/34/659734/1/gate/cross-cinder-py27/aa25e5b/ : SUCCESS in 9m 17s\n- cross-cinder-py35 http://logs.openstack.org/34/659734/1/gate/cross-cinder-py35/4ce371f/ : SUCCESS in 9m 00s\n- cross-glance-py27 http://logs.openstack.org/34/659734/1/gate/cross-glance-py27/b08d581/ : SUCCESS in 4m 15s\n- cross-glance-py35 http://logs.openstack.org/34/659734/1/gate/cross-glance-py35/02c0d1c/ : SUCCESS in 4m 08s\n- cross-horizon-py27 http://logs.openstack.org/34/659734/1/gate/cross-horizon-py27/fe6b4c7/ : SUCCESS in 9m 42s\n- cross-horizon-py35 http://logs.openstack.org/34/659734/1/gate/cross-horizon-py35/961fe1b/ : SUCCESS in 8m 15s\n- cross-keystone-py27 http://logs.openstack.org/34/659734/1/gate/cross-keystone-py27/f62c388/ : SUCCESS in 12m 24s\n- cross-keystone-py35 http://logs.openstack.org/34/659734/1/gate/cross-keystone-py35/6a66c10/ : SUCCESS in 24m 55s\n- cross-neutron-py27 http://logs.openstack.org/34/659734/1/gate/cross-neutron-py27/26d2ef8/ : SUCCESS in 25m 02s\n- cross-neutron-py35 http://logs.openstack.org/34/659734/1/gate/cross-neutron-py35/fc1c639/ : SUCCESS in 28m 53s\n- cross-nova-py27 http://logs.openstack.org/34/659734/1/gate/cross-nova-py27/453cb13/ : SUCCESS in 16m 25s\n- cross-nova-py35 http://logs.openstack.org/34/659734/1/gate/cross-nova-py35/7da8ca9/ : SUCCESS in 16m 41s\n- cross-nova-functional http://logs.openstack.org/34/659734/1/gate/cross-nova-functional/55f1a0b/ : SUCCESS in 20m 36s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/34/659734/1/gate/cross-osc-build-sphinx-docs/d6d3499/html/ : SUCCESS in 6m 57s\n- cross-swift-py27 http://logs.openstack.org/34/659734/1/gate/cross-swift-py27/cfa1f85/ : SUCCESS in 11m 07s\n- cross-swift-py35 http://logs.openstack.org/34/659734/1/gate/cross-swift-py35/6042431/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/34/659734/1/gate/openstack-tox-py36/fde007d/ : SUCCESS in 3m 38s\n- requirements-integration http://logs.openstack.org/34/659734/1/gate/requirements-integration/be515f0/ : SUCCESS in 55m 59s\n- neutron-grenade-multinode http://logs.openstack.org/34/659734/1/gate/neutron-grenade-multinode/d7d93ad/ : SUCCESS in 57m 29s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/34/659734/1/gate/legacy-tempest-dsvm-neutron-full/3d3c469/ : SUCCESS in 1h 35m 51s\n- neutron-grenade http://logs.openstack.org/34/659734/1/gate/neutron-grenade/86f30db/ : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f55f58097a7cf37cb0611756c603e5b9c5761a9f","revisions":{"f55f58097a7cf37cb0611756c603e5b9c5761a9f":{"kind":"REWORK","_number":1,"created":"2019-05-17 07:37:50.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/34/659734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/659734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/659734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/659734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/659734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/659734/1"}}},"commit":{"parents":[{"commit":"65b77b8fb8a767b9e94ab54bc4752265ff694f8c","subject":"Merge \"Add safety check output to the linters output\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65b77b8fb8a767b9e94ab54bc4752265ff694f8c"}]}],"author":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-05-16 16:06:35.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-05-17 07:37:34.000000000","tz":120},"subject":"Exclude broken ironicclient versions 2.5.2 and 2.7.1","message":"Exclude broken ironicclient versions 2.5.2 and 2.7.1\n\nA change was merged to these versions that somehow broke Nova accessing\nIronic. While a fix has been posted in https://review.opendev.org/#/c/659592/\nwe should block these releases to unbreak the CI for now.\n\nChange-Id: I0331d061c13f5e2e109443c5210e0924b73ce0be\nStory: #2005723\nTask: #31055\n(cherry picked from commit ba7f936fd797eba30946f3f263e6050181905988)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f55f58097a7cf37cb0611756c603e5b9c5761a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f55f58097a7cf37cb0611756c603e5b9c5761a9f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
