)]}'
{"id":"openstack%2Frequirements~660254","triplet_id":"openstack%2Frequirements~master~I105eb01daef038c13a87d68e677b1bc5fd0b0ded","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I105eb01daef038c13a87d68e677b1bc5fd0b0ded","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-05-21 06:14:07.000000000","updated":"2019-05-22 10:05:21.000000000","submitted":"2019-05-22 10:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660254-1558519521139-19075ab0","meta_rev_id":"d08510bac10b51c7d5910aae87569e94cc67ed60","_number":660254,"virtual_id_number":660254,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-22 10:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-22 03:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-22 00:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-22 08:09:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-22 03:48:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 00:29:09.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-22 03:48:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 08:09:46.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-22 10:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"926a7dc6861f1dd6fde4f9636632e41c6101b8de","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-21 06:14:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dcc239fad77d8e25723a2fec785b2244f0186ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 08:24:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/54/660254/1/check/openstack-tox-docs/1fc7dad/html/ : SUCCESS in 2m 33s\n- openstack-tox-pypy http://logs.openstack.org/54/660254/1/check/openstack-tox-pypy/bbe1273/ : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/54/660254/1/check/openstack-tox-pep8/cd82476/ : SUCCESS in 2m 30s\n- openstack-tox-py27 http://logs.openstack.org/54/660254/1/check/openstack-tox-py27/d2a1f27/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/54/660254/1/check/openstack-tox-py36/266fc90/ : SUCCESS in 4m 13s\n- openstack-tox-validate http://logs.openstack.org/54/660254/1/check/openstack-tox-validate/805fc15/ : SUCCESS in 3m 19s\n- requirements-tox-babel http://logs.openstack.org/54/660254/1/check/requirements-tox-babel/8bb68fd/ : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/54/660254/1/check/requirements-tox-py27-check-uc/e6ea1e5/ : FAILURE in 7m 37s\n- requirements-tox-py36-check-uc http://logs.openstack.org/54/660254/1/check/requirements-tox-py36-check-uc/1e7a3f1/ : FAILURE in 6m 16s\n- cross-cinder-py27 http://logs.openstack.org/54/660254/1/check/cross-cinder-py27/591ef15/ : SUCCESS in 8m 55s\n- cross-cinder-py36 http://logs.openstack.org/54/660254/1/check/cross-cinder-py36/8bc1903/ : SUCCESS in 8m 13s\n- cross-glance-py27 http://logs.openstack.org/54/660254/1/check/cross-glance-py27/b0019d0/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/54/660254/1/check/cross-glance-py36/4b5c771/ : SUCCESS in 4m 15s\n- cross-horizon-py27 http://logs.openstack.org/54/660254/1/check/cross-horizon-py27/991bfea/ : SUCCESS in 9m 42s\n- cross-horizon-py36 http://logs.openstack.org/54/660254/1/check/cross-horizon-py36/9f7a5f2/ : SUCCESS in 8m 54s\n- cross-horizon-npm http://logs.openstack.org/54/660254/1/check/cross-horizon-npm/ca0ab2e/ : SUCCESS in 6m 06s\n- cross-keystone-py27 http://logs.openstack.org/54/660254/1/check/cross-keystone-py27/7ccc29b/ : SUCCESS in 20m 46s\n- cross-keystone-py36 http://logs.openstack.org/54/660254/1/check/cross-keystone-py36/b1264e8/ : SUCCESS in 22m 56s\n- cross-neutron-py27 http://logs.openstack.org/54/660254/1/check/cross-neutron-py27/4a60b1b/ : SUCCESS in 25m 17s\n- cross-neutron-py36 http://logs.openstack.org/54/660254/1/check/cross-neutron-py36/69f7e27/ : SUCCESS in 21m 46s\n- cross-nova-py27 http://logs.openstack.org/54/660254/1/check/cross-nova-py27/c6c820e/ : SUCCESS in 12m 02s\n- cross-nova-py36 http://logs.openstack.org/54/660254/1/check/cross-nova-py36/9a41513/ : SUCCESS in 11m 42s\n- cross-nova-functional http://logs.openstack.org/54/660254/1/check/cross-nova-functional/070c030/ : FAILURE in 19m 56s\n- cross-osc-tox-docs http://logs.openstack.org/54/660254/1/check/cross-osc-tox-docs/73be945/html/ : SUCCESS in 5m 13s\n- cross-swift-py27 http://logs.openstack.org/54/660254/1/check/cross-swift-py27/a374482/ : SUCCESS in 7m 00s\n- cross-swift-py36 http://logs.openstack.org/54/660254/1/check/cross-swift-py36/afdce8a/ : SUCCESS in 8m 39s\n- requirements-integration http://logs.openstack.org/54/660254/1/check/requirements-integration/2a9a174/ : SUCCESS in 49m 32s\n- neutron-grenade-multinode http://logs.openstack.org/54/660254/1/check/neutron-grenade-multinode/eb6e0ed/ : SUCCESS in 57m 26s\n- tempest-full http://logs.openstack.org/54/660254/1/check/tempest-full/4b4d031/ : FAILURE in 1h 41m 10s\n- neutron-grenade http://logs.openstack.org/54/660254/1/check/neutron-grenade/aa5ad26/ : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"051376119152c8fc365177d184b8cd9bd80b58fc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-21 16:11:55.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":1},{"id":"5ec3005ef5f4068374502debd953c4a7dbff5a61","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-21 16:13:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec12c5a7c67bff75617bbcbc3b393461f62f7a77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 22:44:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/54/660254/2/check/openstack-tox-docs/db7544d/html/ : SUCCESS in 3m 13s\n- openstack-tox-pypy http://logs.openstack.org/54/660254/2/check/openstack-tox-pypy/deb2e74/ : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/54/660254/2/check/openstack-tox-pep8/b3401b0/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/54/660254/2/check/openstack-tox-py27/f9c756d/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/54/660254/2/check/openstack-tox-py36/22d129c/ : SUCCESS in 4m 59s\n- openstack-tox-validate http://logs.openstack.org/54/660254/2/check/openstack-tox-validate/3c4fc73/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/54/660254/2/check/requirements-tox-babel/af3e529/ : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc http://logs.openstack.org/54/660254/2/check/requirements-tox-py27-check-uc/c497f26/ : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc http://logs.openstack.org/54/660254/2/check/requirements-tox-py36-check-uc/fdd0f11/ : SUCCESS in 6m 57s\n- cross-cinder-py27 http://logs.openstack.org/54/660254/2/check/cross-cinder-py27/ac928df/ : SUCCESS in 8m 27s\n- cross-cinder-py36 http://logs.openstack.org/54/660254/2/check/cross-cinder-py36/6565a15/ : SUCCESS in 7m 38s\n- cross-glance-py27 http://logs.openstack.org/54/660254/2/check/cross-glance-py27/e2589fd/ : SUCCESS in 3m 52s\n- cross-glance-py36 http://logs.openstack.org/54/660254/2/check/cross-glance-py36/f82bca0/ : SUCCESS in 3m 44s\n- cross-horizon-py27 http://logs.openstack.org/54/660254/2/check/cross-horizon-py27/446efc1/ : SUCCESS in 10m 36s\n- cross-horizon-py36 http://logs.openstack.org/54/660254/2/check/cross-horizon-py36/2a2e4bc/ : SUCCESS in 9m 07s\n- cross-horizon-npm http://logs.openstack.org/54/660254/2/check/cross-horizon-npm/6311193/ : SUCCESS in 6m 07s\n- cross-keystone-py27 http://logs.openstack.org/54/660254/2/check/cross-keystone-py27/98774ef/ : SUCCESS in 21m 39s\n- cross-keystone-py36 http://logs.openstack.org/54/660254/2/check/cross-keystone-py36/4df410d/ : SUCCESS in 25m 59s\n- cross-neutron-py27 http://logs.openstack.org/54/660254/2/check/cross-neutron-py27/eaec6b7/ : SUCCESS in 29m 45s\n- cross-neutron-py36 http://logs.openstack.org/54/660254/2/check/cross-neutron-py36/0ff63ad/ : SUCCESS in 24m 50s\n- cross-nova-py27 http://logs.openstack.org/54/660254/2/check/cross-nova-py27/f2d4104/ : SUCCESS in 14m 44s\n- cross-nova-py36 http://logs.openstack.org/54/660254/2/check/cross-nova-py36/9ea9e49/ : SUCCESS in 11m 25s\n- cross-nova-functional http://logs.openstack.org/54/660254/2/check/cross-nova-functional/a0617a9/ : FAILURE in 20m 32s\n- cross-osc-tox-docs http://logs.openstack.org/54/660254/2/check/cross-osc-tox-docs/07b9d60/html/ : SUCCESS in 5m 33s\n- cross-swift-py27 http://logs.openstack.org/54/660254/2/check/cross-swift-py27/ca1096f/ : SUCCESS in 7m 41s\n- cross-swift-py36 http://logs.openstack.org/54/660254/2/check/cross-swift-py36/b359dfd/ : SUCCESS in 8m 42s\n- requirements-integration http://logs.openstack.org/54/660254/2/check/requirements-integration/8a316c1/ : SUCCESS in 52m 34s\n- neutron-grenade-multinode http://logs.openstack.org/54/660254/2/check/neutron-grenade-multinode/e096308/ : SUCCESS in 1h 11m 15s\n- tempest-full http://logs.openstack.org/54/660254/2/check/tempest-full/0920c1a/ : FAILURE in 1h 32m 57s\n- neutron-grenade http://logs.openstack.org/54/660254/2/check/neutron-grenade/52ed0d4/ : SUCCESS in 49m 09s","accounts_in_message":[],"_revision_number":2},{"id":"71b64edde5469440d8cc2f072f2638a300d98a7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-21 22:52:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bdbadf505116ba072061f20f4eadcd8651320065","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-21 22:56:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b6ce88309380b8ee3034330982ca9cefe2d2a33","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-22 00:29:09.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks\n\nLooks good to me","accounts_in_message":[],"_revision_number":3},{"id":"14e916e29d3759e2f71268fd089e914c9a364b2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 01:22:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/54/660254/3/check/openstack-tox-docs/de9fa8c/html/ : SUCCESS in 3m 17s\n- openstack-tox-pypy http://logs.openstack.org/54/660254/3/check/openstack-tox-pypy/29b7d8e/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/54/660254/3/check/openstack-tox-pep8/c087019/ : SUCCESS in 2m 45s\n- openstack-tox-py27 http://logs.openstack.org/54/660254/3/check/openstack-tox-py27/5d1f1c5/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/54/660254/3/check/openstack-tox-py36/4a8edd9/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/54/660254/3/check/openstack-tox-validate/4f42aa5/ : SUCCESS in 2m 36s\n- requirements-tox-babel http://logs.openstack.org/54/660254/3/check/requirements-tox-babel/61e1e79/ : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/54/660254/3/check/requirements-tox-py27-check-uc/e473770/ : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc http://logs.openstack.org/54/660254/3/check/requirements-tox-py36-check-uc/17fe496/ : SUCCESS in 5m 47s\n- cross-cinder-py27 http://logs.openstack.org/54/660254/3/check/cross-cinder-py27/ee41899/ : SUCCESS in 8m 08s\n- cross-cinder-py36 http://logs.openstack.org/54/660254/3/check/cross-cinder-py36/f486f01/ : SUCCESS in 7m 47s\n- cross-glance-py27 http://logs.openstack.org/54/660254/3/check/cross-glance-py27/813b007/ : SUCCESS in 3m 59s\n- cross-glance-py36 http://logs.openstack.org/54/660254/3/check/cross-glance-py36/fb06559/ : SUCCESS in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/54/660254/3/check/cross-horizon-py27/d1e86f9/ : SUCCESS in 10m 33s\n- cross-horizon-py36 http://logs.openstack.org/54/660254/3/check/cross-horizon-py36/f5193a1/ : SUCCESS in 8m 18s\n- cross-horizon-npm http://logs.openstack.org/54/660254/3/check/cross-horizon-npm/fd561de/ : SUCCESS in 5m 59s\n- cross-keystone-py27 http://logs.openstack.org/54/660254/3/check/cross-keystone-py27/177a740/ : SUCCESS in 21m 56s\n- cross-keystone-py36 http://logs.openstack.org/54/660254/3/check/cross-keystone-py36/803bcd4/ : SUCCESS in 21m 31s\n- cross-neutron-py27 http://logs.openstack.org/54/660254/3/check/cross-neutron-py27/46c076a/ : SUCCESS in 30m 05s\n- cross-neutron-py36 http://logs.openstack.org/54/660254/3/check/cross-neutron-py36/f9d087f/ : SUCCESS in 23m 34s\n- cross-nova-py27 http://logs.openstack.org/54/660254/3/check/cross-nova-py27/3eef5b0/ : SUCCESS in 12m 07s\n- cross-nova-py36 http://logs.openstack.org/54/660254/3/check/cross-nova-py36/2e2d88c/ : SUCCESS in 24m 39s\n- cross-nova-functional http://logs.openstack.org/54/660254/3/check/cross-nova-functional/4843c44/ : SUCCESS in 18m 02s\n- cross-osc-tox-docs http://logs.openstack.org/54/660254/3/check/cross-osc-tox-docs/ad25376/html/ : SUCCESS in 4m 53s\n- cross-swift-py27 http://logs.openstack.org/54/660254/3/check/cross-swift-py27/03ca9d2/ : SUCCESS in 8m 28s\n- cross-swift-py36 http://logs.openstack.org/54/660254/3/check/cross-swift-py36/935430b/ : SUCCESS in 13m 42s\n- requirements-integration http://logs.openstack.org/54/660254/3/check/requirements-integration/4df5057/ : SUCCESS in 46m 31s\n- neutron-grenade-multinode http://logs.openstack.org/54/660254/3/check/neutron-grenade-multinode/3edd504/ : SUCCESS in 1h 06m 43s\n- tempest-full http://logs.openstack.org/54/660254/3/check/tempest-full/07cf893/ : SUCCESS in 1h 39m 31s\n- neutron-grenade http://logs.openstack.org/54/660254/3/check/neutron-grenade/246fcdf/ : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":3},{"id":"7190544d47cd4c3e95ec6db993699ff7d0adb698","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 03:48:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"74688e6ac1a4786016284fe54265c0249d875a62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 03:48:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"518e475bef979bebf5df1444c8bdc41b1d49e38c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:19:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/54/660254/3/gate/openstack-tox-docs/2002583/html/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/54/660254/3/gate/openstack-tox-pep8/1b42457/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/54/660254/3/gate/openstack-tox-py27/4054099/ : SUCCESS in 2m 53s\n- openstack-tox-py36 http://logs.openstack.org/54/660254/3/gate/openstack-tox-py36/1ad400a/ : SUCCESS in 2m 42s\n- openstack-tox-validate http://logs.openstack.org/54/660254/3/gate/openstack-tox-validate/d575661/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/54/660254/3/gate/requirements-tox-babel/e5d4d6f/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/54/660254/3/gate/requirements-tox-py27-check-uc/24f2d6e/ : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/54/660254/3/gate/requirements-tox-py36-check-uc/ddbd1d2/ : SUCCESS in 6m 10s\n- cross-cinder-py27 http://logs.openstack.org/54/660254/3/gate/cross-cinder-py27/9646075/ : SUCCESS in 8m 48s\n- cross-cinder-py36 http://logs.openstack.org/54/660254/3/gate/cross-cinder-py36/347cb41/ : SUCCESS in 7m 21s\n- cross-glance-py27 http://logs.openstack.org/54/660254/3/gate/cross-glance-py27/5ee1af2/ : SUCCESS in 4m 03s\n- cross-glance-py36 http://logs.openstack.org/54/660254/3/gate/cross-glance-py36/aa8a738/ : FAILURE in 4m 40s\n- cross-horizon-py27 http://logs.openstack.org/54/660254/3/gate/cross-horizon-py27/a7d4393/ : SUCCESS in 8m 51s\n- cross-horizon-py36 http://logs.openstack.org/54/660254/3/gate/cross-horizon-py36/373ff18/ : SUCCESS in 7m 33s\n- cross-horizon-npm http://logs.openstack.org/54/660254/3/gate/cross-horizon-npm/11b0cb2/ : SUCCESS in 5m 14s\n- cross-keystone-py27 http://logs.openstack.org/54/660254/3/gate/cross-keystone-py27/1fcd3c7/ : SUCCESS in 20m 50s\n- cross-keystone-py36 http://logs.openstack.org/54/660254/3/gate/cross-keystone-py36/8405b98/ : SUCCESS in 22m 36s\n- cross-neutron-py27 http://logs.openstack.org/54/660254/3/gate/cross-neutron-py27/c189aa4/ : SUCCESS in 21m 34s\n- cross-neutron-py36 http://logs.openstack.org/54/660254/3/gate/cross-neutron-py36/4c0887d/ : SUCCESS in 17m 47s\n- cross-nova-py27 http://logs.openstack.org/54/660254/3/gate/cross-nova-py27/320a146/ : SUCCESS in 13m 26s\n- cross-nova-py36 http://logs.openstack.org/54/660254/3/gate/cross-nova-py36/a60393d/ : SUCCESS in 10m 51s\n- cross-nova-functional http://logs.openstack.org/54/660254/3/gate/cross-nova-functional/26a3fb5/ : SUCCESS in 19m 12s\n- cross-osc-tox-docs http://logs.openstack.org/54/660254/3/gate/cross-osc-tox-docs/1d5836e/html/ : SUCCESS in 5m 20s\n- cross-swift-py27 http://logs.openstack.org/54/660254/3/gate/cross-swift-py27/b491b0a/ : SUCCESS in 7m 29s\n- cross-swift-py36 http://logs.openstack.org/54/660254/3/gate/cross-swift-py36/6b63301/ : SUCCESS in 11m 20s\n- requirements-integration http://logs.openstack.org/54/660254/3/gate/requirements-integration/7c3f6c6/ : SUCCESS in 49m 10s\n- neutron-grenade-multinode http://logs.openstack.org/54/660254/3/gate/neutron-grenade-multinode/07b997c/ : SUCCESS in 1h 05m 21s\n- tempest-full http://logs.openstack.org/54/660254/3/gate/tempest-full/01f57bd/ : SUCCESS in 1h 30m 19s\n- neutron-grenade http://logs.openstack.org/54/660254/3/gate/neutron-grenade/b954e3a/ : SUCCESS in 50m 40s","accounts_in_message":[],"_revision_number":3},{"id":"d7e9b7e6e68de547dff26ed3978a609489d773eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 05:21:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"92b96681df7638cd99481fe6861ffde92f11ea1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 07:10:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/54/660254/3/check/openstack-tox-docs/a928180/html/ : SUCCESS in 2m 52s\n- openstack-tox-pypy http://logs.openstack.org/54/660254/3/check/openstack-tox-pypy/552033a/ : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/54/660254/3/check/openstack-tox-pep8/27b2cee/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/54/660254/3/check/openstack-tox-py27/006802d/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/54/660254/3/check/openstack-tox-py36/8bdd8e3/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/54/660254/3/check/openstack-tox-validate/1df0572/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/54/660254/3/check/requirements-tox-babel/26ea958/ : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc http://logs.openstack.org/54/660254/3/check/requirements-tox-py27-check-uc/ea6127e/ : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc http://logs.openstack.org/54/660254/3/check/requirements-tox-py36-check-uc/d43fadf/ : SUCCESS in 6m 28s\n- cross-cinder-py27 http://logs.openstack.org/54/660254/3/check/cross-cinder-py27/7f34930/ : SUCCESS in 8m 04s\n- cross-cinder-py36 http://logs.openstack.org/54/660254/3/check/cross-cinder-py36/084bdd9/ : SUCCESS in 7m 08s\n- cross-glance-py27 http://logs.openstack.org/54/660254/3/check/cross-glance-py27/aeeba18/ : SUCCESS in 4m 20s\n- cross-glance-py36 http://logs.openstack.org/54/660254/3/check/cross-glance-py36/902864b/ : SUCCESS in 4m 04s\n- cross-horizon-py27 http://logs.openstack.org/54/660254/3/check/cross-horizon-py27/a69733b/ : SUCCESS in 10m 23s\n- cross-horizon-py36 http://logs.openstack.org/54/660254/3/check/cross-horizon-py36/9ce133b/ : SUCCESS in 7m 30s\n- cross-horizon-npm http://logs.openstack.org/54/660254/3/check/cross-horizon-npm/85f8e6a/ : SUCCESS in 5m 51s\n- cross-keystone-py27 http://logs.openstack.org/54/660254/3/check/cross-keystone-py27/85b495a/ : SUCCESS in 22m 34s\n- cross-keystone-py36 http://logs.openstack.org/54/660254/3/check/cross-keystone-py36/264c7f9/ : SUCCESS in 23m 17s\n- cross-neutron-py27 http://logs.openstack.org/54/660254/3/check/cross-neutron-py27/eb2dca9/ : SUCCESS in 23m 38s\n- cross-neutron-py36 http://logs.openstack.org/54/660254/3/check/cross-neutron-py36/307fd8a/ : SUCCESS in 18m 53s\n- cross-nova-py27 http://logs.openstack.org/54/660254/3/check/cross-nova-py27/1045eb6/ : SUCCESS in 10m 41s\n- cross-nova-py36 http://logs.openstack.org/54/660254/3/check/cross-nova-py36/ae88ed0/ : SUCCESS in 12m 16s\n- cross-nova-functional http://logs.openstack.org/54/660254/3/check/cross-nova-functional/4e2857b/ : SUCCESS in 16m 35s\n- cross-osc-tox-docs http://logs.openstack.org/54/660254/3/check/cross-osc-tox-docs/7a33c88/html/ : SUCCESS in 5m 12s\n- cross-swift-py27 http://logs.openstack.org/54/660254/3/check/cross-swift-py27/41eb9c8/ : SUCCESS in 8m 07s\n- cross-swift-py36 http://logs.openstack.org/54/660254/3/check/cross-swift-py36/47821b8/ : SUCCESS in 7m 42s\n- requirements-integration http://logs.openstack.org/54/660254/3/check/requirements-integration/6dfeb5f/ : SUCCESS in 42m 29s\n- neutron-grenade-multinode http://logs.openstack.org/54/660254/3/check/neutron-grenade-multinode/db6e066/ : SUCCESS in 1h 02m 28s\n- tempest-full http://logs.openstack.org/54/660254/3/check/tempest-full/b852c54/ : SUCCESS in 1h 46m 00s\n- neutron-grenade http://logs.openstack.org/54/660254/3/check/neutron-grenade/26164f8/ : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":3},{"id":"eeeda70e5af8af4c78838bb3b3c6cf693e25225d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 07:10:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"619ecbe1165aa09a90933d934a2277ac110d5c4a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 08:09:46.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd68dde889dc52f4a7c8107f53f9def8ac38d11b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:05:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/54/660254/3/gate/openstack-tox-docs/440283e/html/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/54/660254/3/gate/openstack-tox-pep8/9ccaa75/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/54/660254/3/gate/openstack-tox-py27/62a781d/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/54/660254/3/gate/openstack-tox-py36/0bdf45b/ : SUCCESS in 3m 07s\n- openstack-tox-validate http://logs.openstack.org/54/660254/3/gate/openstack-tox-validate/b9473e4/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/54/660254/3/gate/requirements-tox-babel/3d23806/ : SUCCESS in 2m 27s\n- requirements-tox-py27-check-uc http://logs.openstack.org/54/660254/3/gate/requirements-tox-py27-check-uc/428df2c/ : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc http://logs.openstack.org/54/660254/3/gate/requirements-tox-py36-check-uc/6f308f5/ : SUCCESS in 7m 17s\n- cross-cinder-py27 http://logs.openstack.org/54/660254/3/gate/cross-cinder-py27/889539e/ : SUCCESS in 7m 50s\n- cross-cinder-py36 http://logs.openstack.org/54/660254/3/gate/cross-cinder-py36/fb00ee2/ : SUCCESS in 7m 49s\n- cross-glance-py27 http://logs.openstack.org/54/660254/3/gate/cross-glance-py27/dc22064/ : SUCCESS in 4m 13s\n- cross-glance-py36 http://logs.openstack.org/54/660254/3/gate/cross-glance-py36/9f6c755/ : SUCCESS in 4m 07s\n- cross-horizon-py27 http://logs.openstack.org/54/660254/3/gate/cross-horizon-py27/70c9967/ : SUCCESS in 9m 52s\n- cross-horizon-py36 http://logs.openstack.org/54/660254/3/gate/cross-horizon-py36/05b5332/ : SUCCESS in 8m 11s\n- cross-horizon-npm http://logs.openstack.org/54/660254/3/gate/cross-horizon-npm/e180e2f/ : SUCCESS in 6m 21s\n- cross-keystone-py27 http://logs.openstack.org/54/660254/3/gate/cross-keystone-py27/e824871/ : SUCCESS in 23m 38s\n- cross-keystone-py36 http://logs.openstack.org/54/660254/3/gate/cross-keystone-py36/57f45cc/ : SUCCESS in 23m 19s\n- cross-neutron-py27 http://logs.openstack.org/54/660254/3/gate/cross-neutron-py27/d5f225d/ : SUCCESS in 23m 34s\n- cross-neutron-py36 http://logs.openstack.org/54/660254/3/gate/cross-neutron-py36/5d7fcdf/ : SUCCESS in 24m 11s\n- cross-nova-py27 http://logs.openstack.org/54/660254/3/gate/cross-nova-py27/c9e6926/ : SUCCESS in 12m 20s\n- cross-nova-py36 http://logs.openstack.org/54/660254/3/gate/cross-nova-py36/579a028/ : SUCCESS in 11m 48s\n- cross-nova-functional http://logs.openstack.org/54/660254/3/gate/cross-nova-functional/2f16d91/ : SUCCESS in 18m 23s\n- cross-osc-tox-docs http://logs.openstack.org/54/660254/3/gate/cross-osc-tox-docs/39fbead/html/ : SUCCESS in 6m 53s\n- cross-swift-py27 http://logs.openstack.org/54/660254/3/gate/cross-swift-py27/c4340eb/ : SUCCESS in 10m 55s\n- cross-swift-py36 http://logs.openstack.org/54/660254/3/gate/cross-swift-py36/772d008/ : SUCCESS in 8m 54s\n- requirements-integration http://logs.openstack.org/54/660254/3/gate/requirements-integration/9be1236/ : SUCCESS in 50m 22s\n- neutron-grenade-multinode http://logs.openstack.org/54/660254/3/gate/neutron-grenade-multinode/7d186f4/ : SUCCESS in 1h 10m 33s\n- tempest-full http://logs.openstack.org/54/660254/3/gate/tempest-full/e5f1551/ : SUCCESS in 1h 53m 44s\n- neutron-grenade http://logs.openstack.org/54/660254/3/gate/neutron-grenade/e97293b/ : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":3},{"id":"d08510bac10b51c7d5910aae87569e94cc67ed60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:05:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"147a3cf3e305cfd29fcaad2f403a45f1e1a43d19","revisions":{"4c64ccee602ef1cbacc42d154e53ff0c2e562aaa":{"kind":"REWORK","_number":1,"created":"2019-05-21 06:14:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/660254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/660254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/660254/1"}}},"commit":{"parents":[{"commit":"c2b2b03bcf8c1f48412ab695e0058d0d34a0ad97","subject":"Merge \"Add cinder extras purestorage library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2b2b03bcf8c1f48412ab695e0058d0d34a0ad97"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 06:14:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 06:14:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I105eb01daef038c13a87d68e677b1bc5fd0b0ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c64ccee602ef1cbacc42d154e53ff0c2e562aaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c64ccee602ef1cbacc42d154e53ff0c2e562aaa"}]},"branch":"refs/heads/master"},"f46580729f5ac8009e8ac7417246bb26dcb74e2f":{"kind":"REWORK","_number":2,"created":"2019-05-21 16:13:20.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/54/660254/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/660254/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/660254/2"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 06:14:04.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-21 16:12:55.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I105eb01daef038c13a87d68e677b1bc5fd0b0ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f46580729f5ac8009e8ac7417246bb26dcb74e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f46580729f5ac8009e8ac7417246bb26dcb74e2f"}]},"branch":"refs/heads/master"},"147a3cf3e305cfd29fcaad2f403a45f1e1a43d19":{"kind":"REWORK","_number":3,"created":"2019-05-21 22:56:36.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/54/660254/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/660254/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/660254/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/660254/3"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 06:14:04.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-21 22:56:19.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I105eb01daef038c13a87d68e677b1bc5fd0b0ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/147a3cf3e305cfd29fcaad2f403a45f1e1a43d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/147a3cf3e305cfd29fcaad2f403a45f1e1a43d19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
