)]}'
{"id":"openstack%2Frequirements~660465","triplet_id":"openstack%2Frequirements~master~I906e252e6c09aab9d6fab9de7d7a0d335a097a9a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I906e252e6c09aab9d6fab9de7d7a0d335a097a9a","subject":"update constraint for taskflow to new release 3.6.0","status":"MERGED","created":"2019-05-21 17:29:37.000000000","updated":"2019-05-22 09:37:02.000000000","submitted":"2019-05-22 09:37:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660465-1558517822566-cd6229ff","meta_rev_id":"32cd8441391187758d9c35380922106cd4676797","_number":660465,"virtual_id_number":660465,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-22 09:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-22 05:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-22 06:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-22 05:48:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2019-05-22 08:45:48.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 05:48:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 06:55:39.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-22 08:45:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-22 09:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"910ff8130d67e0e56b05c555851939c4c6490dcd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-21 17:29:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea112977d4d5a4c5ab873d97d2d94a91a98f6357","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 23:50:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/65/660465/1/check/openstack-tox-docs/1aa9d01/html/ : SUCCESS in 2m 55s\n- openstack-tox-pypy http://logs.openstack.org/65/660465/1/check/openstack-tox-pypy/62ca529/ : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/660465/1/check/openstack-tox-pep8/070ec46/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/65/660465/1/check/openstack-tox-py27/459f18f/ : SUCCESS in 2m 50s\n- openstack-tox-py36 http://logs.openstack.org/65/660465/1/check/openstack-tox-py36/54706dc/ : SUCCESS in 4m 34s\n- openstack-tox-validate http://logs.openstack.org/65/660465/1/check/openstack-tox-validate/073d953/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/65/660465/1/check/requirements-tox-babel/7f426cf/ : SUCCESS in 4m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/660465/1/check/requirements-tox-py27-check-uc/447301d/ : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/660465/1/check/requirements-tox-py36-check-uc/c5318a3/ : SUCCESS in 8m 00s\n- cross-cinder-py27 http://logs.openstack.org/65/660465/1/check/cross-cinder-py27/6514dd0/ : SUCCESS in 8m 08s\n- cross-cinder-py36 http://logs.openstack.org/65/660465/1/check/cross-cinder-py36/a0fc9d9/ : SUCCESS in 7m 44s\n- cross-glance-py27 http://logs.openstack.org/65/660465/1/check/cross-glance-py27/dcb5896/ : SUCCESS in 4m 29s\n- cross-glance-py36 http://logs.openstack.org/65/660465/1/check/cross-glance-py36/0240542/ : SUCCESS in 4m 32s\n- cross-horizon-py27 http://logs.openstack.org/65/660465/1/check/cross-horizon-py27/414db42/ : SUCCESS in 10m 27s\n- cross-horizon-py36 http://logs.openstack.org/65/660465/1/check/cross-horizon-py36/91a9f75/ : SUCCESS in 9m 10s\n- cross-horizon-npm http://logs.openstack.org/65/660465/1/check/cross-horizon-npm/622b8c2/ : SUCCESS in 6m 04s\n- cross-keystone-py27 http://logs.openstack.org/65/660465/1/check/cross-keystone-py27/210f04f/ : SUCCESS in 22m 58s\n- cross-keystone-py36 http://logs.openstack.org/65/660465/1/check/cross-keystone-py36/861cb1a/ : SUCCESS in 23m 32s\n- cross-neutron-py27 http://logs.openstack.org/65/660465/1/check/cross-neutron-py27/84794f5/ : SUCCESS in 29m 04s\n- cross-neutron-py36 http://logs.openstack.org/65/660465/1/check/cross-neutron-py36/5b00305/ : SUCCESS in 21m 55s\n- cross-nova-py27 http://logs.openstack.org/65/660465/1/check/cross-nova-py27/8afc534/ : SUCCESS in 13m 56s\n- cross-nova-py36 http://logs.openstack.org/65/660465/1/check/cross-nova-py36/6c2f2c5/ : SUCCESS in 11m 33s\n- cross-nova-functional http://logs.openstack.org/65/660465/1/check/cross-nova-functional/47585d9/ : SUCCESS in 19m 13s\n- cross-osc-tox-docs http://logs.openstack.org/65/660465/1/check/cross-osc-tox-docs/256b28d/html/ : SUCCESS in 5m 43s\n- cross-swift-py27 http://logs.openstack.org/65/660465/1/check/cross-swift-py27/57edaaf/ : SUCCESS in 7m 05s\n- cross-swift-py36 http://logs.openstack.org/65/660465/1/check/cross-swift-py36/afd7fad/ : SUCCESS in 7m 33s\n- requirements-integration http://logs.openstack.org/65/660465/1/check/requirements-integration/0d687bc/ : SUCCESS in 55m 40s\n- neutron-grenade-multinode http://logs.openstack.org/65/660465/1/check/neutron-grenade-multinode/ff9ca93/ : SUCCESS in 1h 17m 48s\n- tempest-full http://logs.openstack.org/65/660465/1/check/tempest-full/150b2e5/ : SUCCESS in 1h 38m 37s\n- neutron-grenade http://logs.openstack.org/65/660465/1/check/neutron-grenade/72191bd/ : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":1},{"id":"55b76a4024e69c8b0eb48da4b3cc0ebdf4d25655","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 05:48:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66adcf8e056eb7cabbaf953c600c3586d78da39c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:48:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f0dd3fb69b8447b62bb324e13af431d0bf15051","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-22 06:55:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8a00ca70db21fef6fe1afa958058c786e43150f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:37:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/65/660465/1/gate/openstack-tox-docs/40e2ad3/html/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/65/660465/1/gate/openstack-tox-pep8/b0cdb2a/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/65/660465/1/gate/openstack-tox-py27/9e9f5fe/ : SUCCESS in 2m 54s\n- openstack-tox-py36 http://logs.openstack.org/65/660465/1/gate/openstack-tox-py36/c45cb6e/ : SUCCESS in 2m 58s\n- openstack-tox-validate http://logs.openstack.org/65/660465/1/gate/openstack-tox-validate/8aa38b9/ : SUCCESS in 3m 06s\n- requirements-tox-babel http://logs.openstack.org/65/660465/1/gate/requirements-tox-babel/ae21030/ : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/660465/1/gate/requirements-tox-py27-check-uc/2122986/ : SUCCESS in 6m 54s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/660465/1/gate/requirements-tox-py36-check-uc/1f1740c/ : SUCCESS in 6m 58s\n- cross-cinder-py27 http://logs.openstack.org/65/660465/1/gate/cross-cinder-py27/3db859e/ : SUCCESS in 7m 48s\n- cross-cinder-py36 http://logs.openstack.org/65/660465/1/gate/cross-cinder-py36/32449a3/ : SUCCESS in 8m 51s\n- cross-glance-py27 http://logs.openstack.org/65/660465/1/gate/cross-glance-py27/baa4348/ : SUCCESS in 3m 58s\n- cross-glance-py36 http://logs.openstack.org/65/660465/1/gate/cross-glance-py36/fb47176/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/65/660465/1/gate/cross-horizon-py27/2635bb3/ : SUCCESS in 9m 11s\n- cross-horizon-py36 http://logs.openstack.org/65/660465/1/gate/cross-horizon-py36/f510101/ : SUCCESS in 8m 09s\n- cross-horizon-npm http://logs.openstack.org/65/660465/1/gate/cross-horizon-npm/4128882/ : SUCCESS in 6m 55s\n- cross-keystone-py27 http://logs.openstack.org/65/660465/1/gate/cross-keystone-py27/48060d7/ : SUCCESS in 26m 15s\n- cross-keystone-py36 http://logs.openstack.org/65/660465/1/gate/cross-keystone-py36/97a08ab/ : SUCCESS in 23m 18s\n- cross-neutron-py27 http://logs.openstack.org/65/660465/1/gate/cross-neutron-py27/d7136d7/ : SUCCESS in 23m 01s\n- cross-neutron-py36 http://logs.openstack.org/65/660465/1/gate/cross-neutron-py36/24f6fc1/ : SUCCESS in 23m 01s\n- cross-nova-py27 http://logs.openstack.org/65/660465/1/gate/cross-nova-py27/831691e/ : SUCCESS in 14m 19s\n- cross-nova-py36 http://logs.openstack.org/65/660465/1/gate/cross-nova-py36/d664501/ : SUCCESS in 13m 22s\n- cross-nova-functional http://logs.openstack.org/65/660465/1/gate/cross-nova-functional/3e650bb/ : SUCCESS in 16m 44s\n- cross-osc-tox-docs http://logs.openstack.org/65/660465/1/gate/cross-osc-tox-docs/e0dab71/html/ : SUCCESS in 5m 16s\n- cross-swift-py27 http://logs.openstack.org/65/660465/1/gate/cross-swift-py27/91748ba/ : SUCCESS in 9m 52s\n- cross-swift-py36 http://logs.openstack.org/65/660465/1/gate/cross-swift-py36/6b44a09/ : SUCCESS in 8m 27s\n- requirements-integration http://logs.openstack.org/65/660465/1/gate/requirements-integration/9f66698/ : SUCCESS in 51m 44s\n- neutron-grenade-multinode http://logs.openstack.org/65/660465/1/gate/neutron-grenade-multinode/11f146f/ : SUCCESS in 1h 09m 34s\n- tempest-full http://logs.openstack.org/65/660465/1/gate/tempest-full/d009ec1/ : SUCCESS in 1h 42m 46s\n- neutron-grenade http://logs.openstack.org/65/660465/1/gate/neutron-grenade/4826c9f/ : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1c284ac166c995de0c87b5891b22f735de7dab29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:37:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"356e448cf5e885e716f29f45d3eeda44e7d39b44","revisions":{"356e448cf5e885e716f29f45d3eeda44e7d39b44":{"kind":"REWORK","_number":1,"created":"2019-05-21 17:29:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/660465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/660465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/660465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/660465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/660465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/660465/1"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:29:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:29:36.000000000","tz":0},"subject":"update constraint for taskflow to new release 3.6.0","message":"update constraint for taskflow to new release 3.6.0\n\nChange-Id: I906e252e6c09aab9d6fab9de7d7a0d335a097a9a\nmeta:version: 3.6.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I47c78915181ece3ecea576028edce3e90203d87e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/356e448cf5e885e716f29f45d3eeda44e7d39b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/356e448cf5e885e716f29f45d3eeda44e7d39b44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
