)]}'
{"id":"openstack%2Frequirements~660467","triplet_id":"openstack%2Frequirements~master~Id79e24be60741fc7df57cb80899e4992f73cd14f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id79e24be60741fc7df57cb80899e4992f73cd14f","subject":"update constraint for oslo.cache to new release 1.35.0","status":"MERGED","created":"2019-05-21 17:29:51.000000000","updated":"2019-05-22 09:31:06.000000000","submitted":"2019-05-22 09:31:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660467-1558517466749-3248f38c","meta_rev_id":"0c80f0d394fed5cf06f74cc5311f61b33865fe95","_number":660467,"virtual_id_number":660467,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-22 09:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-22 05:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-22 06:55:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-22 05:48:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 05:48:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 06:55:48.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-22 09:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87f607d9fc2fcc3b7ed9fa333682b5025b987689","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-21 17:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56b55d0ef015393242c41f3bbeda7f2f8354e51a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 23:42:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/660467/1/check/openstack-tox-docs/9eb1b30/html/ : SUCCESS in 2m 58s\n- openstack-tox-pypy http://logs.openstack.org/67/660467/1/check/openstack-tox-pypy/ff084b3/ : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/660467/1/check/openstack-tox-pep8/2eb646a/ : SUCCESS in 2m 45s\n- openstack-tox-py27 http://logs.openstack.org/67/660467/1/check/openstack-tox-py27/fd7df6f/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/67/660467/1/check/openstack-tox-py36/da5c0bf/ : SUCCESS in 3m 50s\n- openstack-tox-validate http://logs.openstack.org/67/660467/1/check/openstack-tox-validate/965faf3/ : SUCCESS in 2m 52s\n- requirements-tox-babel http://logs.openstack.org/67/660467/1/check/requirements-tox-babel/6353e43/ : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/660467/1/check/requirements-tox-py27-check-uc/bf36265/ : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/660467/1/check/requirements-tox-py36-check-uc/afc506b/ : SUCCESS in 6m 25s\n- cross-cinder-py27 http://logs.openstack.org/67/660467/1/check/cross-cinder-py27/eae6f16/ : SUCCESS in 8m 37s\n- cross-cinder-py36 http://logs.openstack.org/67/660467/1/check/cross-cinder-py36/d095738/ : SUCCESS in 8m 38s\n- cross-glance-py27 http://logs.openstack.org/67/660467/1/check/cross-glance-py27/a130153/ : SUCCESS in 4m 20s\n- cross-glance-py36 http://logs.openstack.org/67/660467/1/check/cross-glance-py36/a7d6aa8/ : SUCCESS in 3m 56s\n- cross-horizon-py27 http://logs.openstack.org/67/660467/1/check/cross-horizon-py27/e451300/ : SUCCESS in 9m 34s\n- cross-horizon-py36 http://logs.openstack.org/67/660467/1/check/cross-horizon-py36/a744221/ : SUCCESS in 8m 40s\n- cross-horizon-npm http://logs.openstack.org/67/660467/1/check/cross-horizon-npm/f278083/ : SUCCESS in 6m 01s\n- cross-keystone-py27 http://logs.openstack.org/67/660467/1/check/cross-keystone-py27/f4bf898/ : SUCCESS in 19m 22s\n- cross-keystone-py36 http://logs.openstack.org/67/660467/1/check/cross-keystone-py36/ea1504c/ : SUCCESS in 25m 46s\n- cross-neutron-py27 http://logs.openstack.org/67/660467/1/check/cross-neutron-py27/99b04b5/ : SUCCESS in 21m 32s\n- cross-neutron-py36 http://logs.openstack.org/67/660467/1/check/cross-neutron-py36/06c50b7/ : SUCCESS in 21m 23s\n- cross-nova-py27 http://logs.openstack.org/67/660467/1/check/cross-nova-py27/3c15afd/ : SUCCESS in 11m 29s\n- cross-nova-py36 http://logs.openstack.org/67/660467/1/check/cross-nova-py36/2839796/ : SUCCESS in 13m 41s\n- cross-nova-functional http://logs.openstack.org/67/660467/1/check/cross-nova-functional/e4864e9/ : SUCCESS in 18m 55s\n- cross-osc-tox-docs http://logs.openstack.org/67/660467/1/check/cross-osc-tox-docs/703a4b3/html/ : SUCCESS in 5m 38s\n- cross-swift-py27 http://logs.openstack.org/67/660467/1/check/cross-swift-py27/bac06e0/ : SUCCESS in 7m 35s\n- cross-swift-py36 http://logs.openstack.org/67/660467/1/check/cross-swift-py36/9588e2f/ : SUCCESS in 8m 03s\n- requirements-integration http://logs.openstack.org/67/660467/1/check/requirements-integration/f17b65f/ : SUCCESS in 52m 56s\n- neutron-grenade-multinode http://logs.openstack.org/67/660467/1/check/neutron-grenade-multinode/8a09964/ : SUCCESS in 1h 07m 46s\n- tempest-full http://logs.openstack.org/67/660467/1/check/tempest-full/49caef8/ : SUCCESS in 1h 41m 52s\n- neutron-grenade http://logs.openstack.org/67/660467/1/check/neutron-grenade/541038d/ : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a4c1042a7e24b583e94c0e76dd09679ec759806a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 05:48:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82213f3f7e66b6c41f4220ca6a05d8ee3254703d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:48:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e51b85552577bc69665149646ad39dd309ee937","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-22 06:55:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a864680cea6e8adde76d8907a290d940131c71c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:31:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4eee01146902a981bc5620a24ad33985bef477a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:31:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/660467/1/gate/openstack-tox-docs/f30dd98/html/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/67/660467/1/gate/openstack-tox-pep8/d78c334/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/67/660467/1/gate/openstack-tox-py27/426e8a9/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/67/660467/1/gate/openstack-tox-py36/25d732f/ : SUCCESS in 3m 07s\n- openstack-tox-validate http://logs.openstack.org/67/660467/1/gate/openstack-tox-validate/32c623a/ : SUCCESS in 2m 55s\n- requirements-tox-babel http://logs.openstack.org/67/660467/1/gate/requirements-tox-babel/6a7585b/ : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/660467/1/gate/requirements-tox-py27-check-uc/d99543c/ : SUCCESS in 6m 05s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/660467/1/gate/requirements-tox-py36-check-uc/e6dc3e8/ : SUCCESS in 6m 07s\n- cross-cinder-py27 http://logs.openstack.org/67/660467/1/gate/cross-cinder-py27/8e3468f/ : SUCCESS in 9m 52s\n- cross-cinder-py36 http://logs.openstack.org/67/660467/1/gate/cross-cinder-py36/a0f4634/ : SUCCESS in 8m 00s\n- cross-glance-py27 http://logs.openstack.org/67/660467/1/gate/cross-glance-py27/ff1ae1a/ : SUCCESS in 5m 56s\n- cross-glance-py36 http://logs.openstack.org/67/660467/1/gate/cross-glance-py36/f475706/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/67/660467/1/gate/cross-horizon-py27/10b6a76/ : SUCCESS in 8m 44s\n- cross-horizon-py36 http://logs.openstack.org/67/660467/1/gate/cross-horizon-py36/7bc98fb/ : SUCCESS in 10m 41s\n- cross-horizon-npm http://logs.openstack.org/67/660467/1/gate/cross-horizon-npm/02e6dfa/ : SUCCESS in 5m 03s\n- cross-keystone-py27 http://logs.openstack.org/67/660467/1/gate/cross-keystone-py27/236a9d3/ : SUCCESS in 21m 05s\n- cross-keystone-py36 http://logs.openstack.org/67/660467/1/gate/cross-keystone-py36/d7b885f/ : SUCCESS in 26m 03s\n- cross-neutron-py27 http://logs.openstack.org/67/660467/1/gate/cross-neutron-py27/39e775d/ : SUCCESS in 25m 27s\n- cross-neutron-py36 http://logs.openstack.org/67/660467/1/gate/cross-neutron-py36/2d87fe1/ : SUCCESS in 20m 34s\n- cross-nova-py27 http://logs.openstack.org/67/660467/1/gate/cross-nova-py27/ae2e690/ : SUCCESS in 13m 41s\n- cross-nova-py36 http://logs.openstack.org/67/660467/1/gate/cross-nova-py36/3e11567/ : SUCCESS in 10m 35s\n- cross-nova-functional http://logs.openstack.org/67/660467/1/gate/cross-nova-functional/b72cc93/ : SUCCESS in 21m 27s\n- cross-osc-tox-docs http://logs.openstack.org/67/660467/1/gate/cross-osc-tox-docs/039d153/html/ : SUCCESS in 5m 27s\n- cross-swift-py27 http://logs.openstack.org/67/660467/1/gate/cross-swift-py27/4d8be37/ : SUCCESS in 7m 41s\n- cross-swift-py36 http://logs.openstack.org/67/660467/1/gate/cross-swift-py36/e66b59a/ : SUCCESS in 10m 45s\n- requirements-integration http://logs.openstack.org/67/660467/1/gate/requirements-integration/881463a/ : SUCCESS in 52m 37s\n- neutron-grenade-multinode http://logs.openstack.org/67/660467/1/gate/neutron-grenade-multinode/758c43d/ : SUCCESS in 1h 07m 28s\n- tempest-full http://logs.openstack.org/67/660467/1/gate/tempest-full/6bfcd33/ : SUCCESS in 1h 37m 21s\n- neutron-grenade http://logs.openstack.org/67/660467/1/gate/neutron-grenade/17e85b3/ : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef552dc152eb37ea6c7dc06b6d4cd278cc54885f","revisions":{"ef552dc152eb37ea6c7dc06b6d4cd278cc54885f":{"kind":"REWORK","_number":1,"created":"2019-05-21 17:29:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/660467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/660467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/660467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/660467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/660467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/660467/1"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:29:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:29:48.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 1.35.0","message":"update constraint for oslo.cache to new release 1.35.0\n\nChange-Id: Id79e24be60741fc7df57cb80899e4992f73cd14f\nmeta:version: 1.35.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I47c78915181ece3ecea576028edce3e90203d87e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef552dc152eb37ea6c7dc06b6d4cd278cc54885f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef552dc152eb37ea6c7dc06b6d4cd278cc54885f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
