)]}'
{"id":"openstack%2Frequirements~660469","triplet_id":"openstack%2Frequirements~master~I544f1f20194f8e0d5e89ba6f0754c0f2af639680","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I544f1f20194f8e0d5e89ba6f0754c0f2af639680","subject":"update constraint for oslo.log to new release 3.44.0","status":"MERGED","created":"2019-05-21 17:30:22.000000000","updated":"2019-05-22 09:31:04.000000000","submitted":"2019-05-22 09:31:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660469-1558517464245-f3b1de75","meta_rev_id":"19de53671cbb1f57b0fe922531685b2dfa97627b","_number":660469,"virtual_id_number":660469,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-22 09:31:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-22 05:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-22 06:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-22 05:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 05:48:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 06:54:31.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-22 09:31:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa132584810108ad338c5c56875a4eb152a7627c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-21 17:30:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f42acc487870bf7d563badb7536d9d92722dc26c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 23:52:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/660469/1/check/openstack-tox-docs/50dbe9f/html/ : SUCCESS in 2m 58s\n- openstack-tox-pypy http://logs.openstack.org/69/660469/1/check/openstack-tox-pypy/5c0b29b/ : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/69/660469/1/check/openstack-tox-pep8/de06d27/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/69/660469/1/check/openstack-tox-py27/8c18835/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/69/660469/1/check/openstack-tox-py36/4d53a7c/ : SUCCESS in 3m 15s\n- openstack-tox-validate http://logs.openstack.org/69/660469/1/check/openstack-tox-validate/250c62c/ : SUCCESS in 2m 42s\n- requirements-tox-babel http://logs.openstack.org/69/660469/1/check/requirements-tox-babel/893572b/ : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/69/660469/1/check/requirements-tox-py27-check-uc/5739f7a/ : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc http://logs.openstack.org/69/660469/1/check/requirements-tox-py36-check-uc/f21580b/ : SUCCESS in 6m 00s\n- cross-cinder-py27 http://logs.openstack.org/69/660469/1/check/cross-cinder-py27/ae2964d/ : SUCCESS in 9m 37s\n- cross-cinder-py36 http://logs.openstack.org/69/660469/1/check/cross-cinder-py36/49cd672/ : SUCCESS in 7m 39s\n- cross-glance-py27 http://logs.openstack.org/69/660469/1/check/cross-glance-py27/f3904c0/ : SUCCESS in 4m 03s\n- cross-glance-py36 http://logs.openstack.org/69/660469/1/check/cross-glance-py36/7741746/ : SUCCESS in 4m 32s\n- cross-horizon-py27 http://logs.openstack.org/69/660469/1/check/cross-horizon-py27/89b462e/ : SUCCESS in 10m 12s\n- cross-horizon-py36 http://logs.openstack.org/69/660469/1/check/cross-horizon-py36/44c8c6e/ : SUCCESS in 7m 53s\n- cross-horizon-npm http://logs.openstack.org/69/660469/1/check/cross-horizon-npm/cf83b56/ : SUCCESS in 6m 05s\n- cross-keystone-py27 http://logs.openstack.org/69/660469/1/check/cross-keystone-py27/455628c/ : SUCCESS in 23m 16s\n- cross-keystone-py36 http://logs.openstack.org/69/660469/1/check/cross-keystone-py36/e49d4e0/ : SUCCESS in 27m 56s\n- cross-neutron-py27 http://logs.openstack.org/69/660469/1/check/cross-neutron-py27/57102b6/ : SUCCESS in 27m 16s\n- cross-neutron-py36 http://logs.openstack.org/69/660469/1/check/cross-neutron-py36/5b78071/ : SUCCESS in 24m 51s\n- cross-nova-py27 http://logs.openstack.org/69/660469/1/check/cross-nova-py27/d9da438/ : SUCCESS in 13m 32s\n- cross-nova-py36 http://logs.openstack.org/69/660469/1/check/cross-nova-py36/60c0e3f/ : SUCCESS in 12m 30s\n- cross-nova-functional http://logs.openstack.org/69/660469/1/check/cross-nova-functional/43a51e6/ : SUCCESS in 22m 51s\n- cross-osc-tox-docs http://logs.openstack.org/69/660469/1/check/cross-osc-tox-docs/48a7396/html/ : SUCCESS in 5m 04s\n- cross-swift-py27 http://logs.openstack.org/69/660469/1/check/cross-swift-py27/860eb61/ : SUCCESS in 8m 03s\n- cross-swift-py36 http://logs.openstack.org/69/660469/1/check/cross-swift-py36/67e705d/ : SUCCESS in 8m 00s\n- requirements-integration http://logs.openstack.org/69/660469/1/check/requirements-integration/a15376e/ : SUCCESS in 51m 28s\n- neutron-grenade-multinode http://logs.openstack.org/69/660469/1/check/neutron-grenade-multinode/5d2003b/ : SUCCESS in 1h 12m 19s\n- tempest-full http://logs.openstack.org/69/660469/1/check/tempest-full/ad03ccb/ : SUCCESS in 1h 50m 14s\n- neutron-grenade http://logs.openstack.org/69/660469/1/check/neutron-grenade/a904fc1/ : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"abaebbef6cd6859e767faf8f6fbf7f190b6e5c9a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 05:48:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d5abb015e80c8293df22c792ceecc54e78bde33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:48:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f458bd4615427af9f411172c9ca142d5146d1ba2","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-22 06:54:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff6ce90b4fca5a4beb493352cb61b5313bdf716e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:31:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/660469/1/gate/openstack-tox-docs/8d65832/html/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/69/660469/1/gate/openstack-tox-pep8/9200627/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/69/660469/1/gate/openstack-tox-py27/bc4440d/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/69/660469/1/gate/openstack-tox-py36/d94b9ca/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/69/660469/1/gate/openstack-tox-validate/b592d5b/ : SUCCESS in 3m 00s\n- requirements-tox-babel http://logs.openstack.org/69/660469/1/gate/requirements-tox-babel/7942443/ : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/69/660469/1/gate/requirements-tox-py27-check-uc/ae84253/ : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc http://logs.openstack.org/69/660469/1/gate/requirements-tox-py36-check-uc/d1f66a7/ : SUCCESS in 7m 44s\n- cross-cinder-py27 http://logs.openstack.org/69/660469/1/gate/cross-cinder-py27/b8db5e6/ : SUCCESS in 8m 12s\n- cross-cinder-py36 http://logs.openstack.org/69/660469/1/gate/cross-cinder-py36/3376c1c/ : SUCCESS in 7m 09s\n- cross-glance-py27 http://logs.openstack.org/69/660469/1/gate/cross-glance-py27/61f16e6/ : SUCCESS in 3m 48s\n- cross-glance-py36 http://logs.openstack.org/69/660469/1/gate/cross-glance-py36/79e646c/ : SUCCESS in 5m 25s\n- cross-horizon-py27 http://logs.openstack.org/69/660469/1/gate/cross-horizon-py27/b58b0d8/ : SUCCESS in 10m 45s\n- cross-horizon-py36 http://logs.openstack.org/69/660469/1/gate/cross-horizon-py36/292b973/ : SUCCESS in 9m 33s\n- cross-horizon-npm http://logs.openstack.org/69/660469/1/gate/cross-horizon-npm/b12f464/ : SUCCESS in 5m 11s\n- cross-keystone-py27 http://logs.openstack.org/69/660469/1/gate/cross-keystone-py27/bf205b2/ : SUCCESS in 24m 39s\n- cross-keystone-py36 http://logs.openstack.org/69/660469/1/gate/cross-keystone-py36/814dcc1/ : SUCCESS in 24m 23s\n- cross-neutron-py27 http://logs.openstack.org/69/660469/1/gate/cross-neutron-py27/865d667/ : SUCCESS in 22m 48s\n- cross-neutron-py36 http://logs.openstack.org/69/660469/1/gate/cross-neutron-py36/55e60fb/ : SUCCESS in 23m 42s\n- cross-nova-py27 http://logs.openstack.org/69/660469/1/gate/cross-nova-py27/0a4fe2b/ : SUCCESS in 13m 51s\n- cross-nova-py36 http://logs.openstack.org/69/660469/1/gate/cross-nova-py36/22520c3/ : SUCCESS in 12m 34s\n- cross-nova-functional http://logs.openstack.org/69/660469/1/gate/cross-nova-functional/5c83005/ : SUCCESS in 19m 11s\n- cross-osc-tox-docs http://logs.openstack.org/69/660469/1/gate/cross-osc-tox-docs/7e0de83/html/ : SUCCESS in 5m 33s\n- cross-swift-py27 http://logs.openstack.org/69/660469/1/gate/cross-swift-py27/ff9a7dd/ : SUCCESS in 8m 15s\n- cross-swift-py36 http://logs.openstack.org/69/660469/1/gate/cross-swift-py36/251cf83/ : SUCCESS in 8m 40s\n- requirements-integration http://logs.openstack.org/69/660469/1/gate/requirements-integration/5cad1cb/ : SUCCESS in 53m 43s\n- neutron-grenade-multinode http://logs.openstack.org/69/660469/1/gate/neutron-grenade-multinode/5e52aa5/ : SUCCESS in 1h 02m 58s\n- tempest-full http://logs.openstack.org/69/660469/1/gate/tempest-full/aa4aee9/ : SUCCESS in 1h 52m 26s\n- neutron-grenade http://logs.openstack.org/69/660469/1/gate/neutron-grenade/7c5ade1/ : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"19995392198bc8d4a09f2006e2b9fdfe0f297368","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:31:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c47c5620acdde7dc39408638c34e456f665aa2a","revisions":{"0c47c5620acdde7dc39408638c34e456f665aa2a":{"kind":"REWORK","_number":1,"created":"2019-05-21 17:30:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/660469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/660469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/660469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/660469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/660469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/660469/1"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:30:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:30:20.000000000","tz":0},"subject":"update constraint for oslo.log to new release 3.44.0","message":"update constraint for oslo.log to new release 3.44.0\n\nChange-Id: I544f1f20194f8e0d5e89ba6f0754c0f2af639680\nmeta:version: 3.44.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I47c78915181ece3ecea576028edce3e90203d87e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c47c5620acdde7dc39408638c34e456f665aa2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c47c5620acdde7dc39408638c34e456f665aa2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
