)]}'
{"id":"openstack%2Frequirements~660470","triplet_id":"openstack%2Frequirements~master~I1a3e05721a42fcbdf36662c68857d26d3a54aa65","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1a3e05721a42fcbdf36662c68857d26d3a54aa65","subject":"update constraint for metalsmith to new release 0.12.0","status":"MERGED","created":"2019-05-21 17:30:27.000000000","updated":"2019-05-22 09:52:49.000000000","submitted":"2019-05-22 09:52:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660470-1558518769932-74850baa","meta_rev_id":"eee2d5ccdf284384a0bfa98833bccfee7ac33fbb","_number":660470,"virtual_id_number":660470,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-22 09:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-22 05:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-22 06:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-22 05:48:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 05:48:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 06:55:33.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-22 09:52:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e99320432bfec54bfa21c1ded2eedbfd3b6f1c3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-21 17:30:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea86ebc268956e0d0214986d2b09a78c03b3a791","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 23:54:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/660470/1/check/openstack-tox-docs/670473d/html/ : SUCCESS in 3m 06s\n- openstack-tox-pypy http://logs.openstack.org/70/660470/1/check/openstack-tox-pypy/0765bc3/ : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/660470/1/check/openstack-tox-pep8/4d14937/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/70/660470/1/check/openstack-tox-py27/7ee3e92/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/70/660470/1/check/openstack-tox-py36/d8f7b94/ : SUCCESS in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/70/660470/1/check/openstack-tox-validate/db5feaf/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/70/660470/1/check/requirements-tox-babel/fccb87b/ : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/70/660470/1/check/requirements-tox-py27-check-uc/c574842/ : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc http://logs.openstack.org/70/660470/1/check/requirements-tox-py36-check-uc/f2dc04e/ : SUCCESS in 5m 50s\n- cross-cinder-py27 http://logs.openstack.org/70/660470/1/check/cross-cinder-py27/5562acc/ : SUCCESS in 8m 06s\n- cross-cinder-py36 http://logs.openstack.org/70/660470/1/check/cross-cinder-py36/b7d0b37/ : SUCCESS in 8m 13s\n- cross-glance-py27 http://logs.openstack.org/70/660470/1/check/cross-glance-py27/1366dc5/ : SUCCESS in 4m 05s\n- cross-glance-py36 http://logs.openstack.org/70/660470/1/check/cross-glance-py36/f900ce1/ : SUCCESS in 4m 26s\n- cross-horizon-py27 http://logs.openstack.org/70/660470/1/check/cross-horizon-py27/b5f7bf7/ : SUCCESS in 13m 51s\n- cross-horizon-py36 http://logs.openstack.org/70/660470/1/check/cross-horizon-py36/203447a/ : SUCCESS in 12m 32s\n- cross-horizon-npm http://logs.openstack.org/70/660470/1/check/cross-horizon-npm/727fe5c/ : SUCCESS in 8m 35s\n- cross-keystone-py27 http://logs.openstack.org/70/660470/1/check/cross-keystone-py27/871ce16/ : SUCCESS in 30m 31s\n- cross-keystone-py36 http://logs.openstack.org/70/660470/1/check/cross-keystone-py36/e75b107/ : SUCCESS in 30m 58s\n- cross-neutron-py27 http://logs.openstack.org/70/660470/1/check/cross-neutron-py27/a7d21df/ : SUCCESS in 22m 41s\n- cross-neutron-py36 http://logs.openstack.org/70/660470/1/check/cross-neutron-py36/045f7de/ : SUCCESS in 23m 22s\n- cross-nova-py27 http://logs.openstack.org/70/660470/1/check/cross-nova-py27/1b28df2/ : SUCCESS in 11m 39s\n- cross-nova-py36 http://logs.openstack.org/70/660470/1/check/cross-nova-py36/f701730/ : SUCCESS in 12m 10s\n- cross-nova-functional http://logs.openstack.org/70/660470/1/check/cross-nova-functional/3790b1e/ : SUCCESS in 18m 00s\n- cross-osc-tox-docs http://logs.openstack.org/70/660470/1/check/cross-osc-tox-docs/44f640f/html/ : SUCCESS in 5m 19s\n- cross-swift-py27 http://logs.openstack.org/70/660470/1/check/cross-swift-py27/2b52864/ : SUCCESS in 8m 02s\n- cross-swift-py36 http://logs.openstack.org/70/660470/1/check/cross-swift-py36/3d864e1/ : SUCCESS in 8m 40s\n- requirements-integration http://logs.openstack.org/70/660470/1/check/requirements-integration/5d6516b/ : SUCCESS in 51m 36s\n- neutron-grenade-multinode http://logs.openstack.org/70/660470/1/check/neutron-grenade-multinode/9ae6c9a/ : SUCCESS in 1h 05m 51s\n- tempest-full http://logs.openstack.org/70/660470/1/check/tempest-full/84e98f6/ : SUCCESS in 1h 51m 11s\n- neutron-grenade http://logs.openstack.org/70/660470/1/check/neutron-grenade/3b23873/ : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"38b3675076e054107fa8b854ca5210fb8d0e38e7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 05:48:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85f67c597f7478d16a72dde2cd1443ca4344118e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:48:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36c5efc0ee895ce39965ccc19592f696b70ace82","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-22 06:55:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"483aad81d19b6c3b82e89ba4481c6db9c5fa1ce8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:52:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"568cd65deb86594d02acf815dd20f0c0f30e53cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:52:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/660470/1/gate/openstack-tox-docs/d719e6c/html/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/70/660470/1/gate/openstack-tox-pep8/f7280f6/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/70/660470/1/gate/openstack-tox-py27/fca72d9/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/70/660470/1/gate/openstack-tox-py36/e9854f0/ : SUCCESS in 4m 36s\n- openstack-tox-validate http://logs.openstack.org/70/660470/1/gate/openstack-tox-validate/f2edeed/ : SUCCESS in 3m 05s\n- requirements-tox-babel http://logs.openstack.org/70/660470/1/gate/requirements-tox-babel/9664430/ : SUCCESS in 4m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/70/660470/1/gate/requirements-tox-py27-check-uc/14673c6/ : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc http://logs.openstack.org/70/660470/1/gate/requirements-tox-py36-check-uc/f4dc71e/ : SUCCESS in 7m 53s\n- cross-cinder-py27 http://logs.openstack.org/70/660470/1/gate/cross-cinder-py27/3466093/ : SUCCESS in 8m 11s\n- cross-cinder-py36 http://logs.openstack.org/70/660470/1/gate/cross-cinder-py36/e99bdf2/ : SUCCESS in 10m 16s\n- cross-glance-py27 http://logs.openstack.org/70/660470/1/gate/cross-glance-py27/b5ef6d8/ : SUCCESS in 3m 59s\n- cross-glance-py36 http://logs.openstack.org/70/660470/1/gate/cross-glance-py36/dac26f7/ : SUCCESS in 4m 21s\n- cross-horizon-py27 http://logs.openstack.org/70/660470/1/gate/cross-horizon-py27/121ca09/ : SUCCESS in 10m 32s\n- cross-horizon-py36 http://logs.openstack.org/70/660470/1/gate/cross-horizon-py36/0f6d4ae/ : SUCCESS in 9m 16s\n- cross-horizon-npm http://logs.openstack.org/70/660470/1/gate/cross-horizon-npm/9a9e649/ : SUCCESS in 5m 47s\n- cross-keystone-py27 http://logs.openstack.org/70/660470/1/gate/cross-keystone-py27/5050a5a/ : SUCCESS in 25m 06s\n- cross-keystone-py36 http://logs.openstack.org/70/660470/1/gate/cross-keystone-py36/f3218a8/ : SUCCESS in 23m 39s\n- cross-neutron-py27 http://logs.openstack.org/70/660470/1/gate/cross-neutron-py27/35760eb/ : SUCCESS in 30m 24s\n- cross-neutron-py36 http://logs.openstack.org/70/660470/1/gate/cross-neutron-py36/3240b66/ : SUCCESS in 26m 19s\n- cross-nova-py27 http://logs.openstack.org/70/660470/1/gate/cross-nova-py27/568bae8/ : SUCCESS in 17m 49s\n- cross-nova-py36 http://logs.openstack.org/70/660470/1/gate/cross-nova-py36/2d03d4b/ : SUCCESS in 19m 14s\n- cross-nova-functional http://logs.openstack.org/70/660470/1/gate/cross-nova-functional/9a58a21/ : SUCCESS in 19m 21s\n- cross-osc-tox-docs http://logs.openstack.org/70/660470/1/gate/cross-osc-tox-docs/08fb26b/html/ : SUCCESS in 5m 32s\n- cross-swift-py27 http://logs.openstack.org/70/660470/1/gate/cross-swift-py27/45a0f9d/ : SUCCESS in 7m 54s\n- cross-swift-py36 http://logs.openstack.org/70/660470/1/gate/cross-swift-py36/dccd589/ : SUCCESS in 11m 57s\n- requirements-integration http://logs.openstack.org/70/660470/1/gate/requirements-integration/387d9d0/ : SUCCESS in 1h 05m 09s\n- neutron-grenade-multinode http://logs.openstack.org/70/660470/1/gate/neutron-grenade-multinode/c742336/ : SUCCESS in 1h 12m 35s\n- tempest-full http://logs.openstack.org/70/660470/1/gate/tempest-full/6990b01/ : SUCCESS in 1h 55m 04s\n- neutron-grenade http://logs.openstack.org/70/660470/1/gate/neutron-grenade/ec6b5d6/ : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c635820d25f48202d39136a306d8874bb2c7b336","revisions":{"c635820d25f48202d39136a306d8874bb2c7b336":{"kind":"REWORK","_number":1,"created":"2019-05-21 17:30:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/660470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/660470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/660470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/660470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/660470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/660470/1"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:30:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:30:24.000000000","tz":0},"subject":"update constraint for metalsmith to new release 0.12.0","message":"update constraint for metalsmith to new release 0.12.0\n\nChange-Id: I1a3e05721a42fcbdf36662c68857d26d3a54aa65\nmeta:version: 0.12.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Dmitry Tantsur \u003cdivius.inside@gmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdivius.inside@gmail.com\u003e\nmeta:release:Change-Id: Id09a3eb3d4403613559dfd407669645ae3831282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c635820d25f48202d39136a306d8874bb2c7b336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c635820d25f48202d39136a306d8874bb2c7b336"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
