)]}'
{"id":"openstack%2Frequirements~660471","triplet_id":"openstack%2Frequirements~master~I882a072d711d93dd4b92f2690d79b265b7ab58e4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I882a072d711d93dd4b92f2690d79b265b7ab58e4","subject":"update constraint for oslo.vmware to new release 2.34.0","status":"MERGED","created":"2019-05-21 17:31:24.000000000","updated":"2019-05-23 10:43:46.000000000","submitted":"2019-05-23 10:43:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660471-1558608226101-7d69f64c","meta_rev_id":"5745349af0a9b8aeafe7de81878e50b099cbc90f","_number":660471,"virtual_id_number":660471,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-23 10:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-22 05:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-22 06:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-22 05:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 05:48:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 06:54:22.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-23 10:43:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a988d1d1a25d1c536c69164a377663fbf0a11d98","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-21 17:31:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4668a730f49b9397eda3821b39a01ffe13ad84cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 23:42:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/71/660471/1/check/openstack-tox-docs/bb5cff0/html/ : SUCCESS in 2m 53s\n- openstack-tox-pypy http://logs.openstack.org/71/660471/1/check/openstack-tox-pypy/1d9aea5/ : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/71/660471/1/check/openstack-tox-pep8/cbb754d/ : SUCCESS in 2m 41s\n- openstack-tox-py27 http://logs.openstack.org/71/660471/1/check/openstack-tox-py27/2ca1a41/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/71/660471/1/check/openstack-tox-py36/8bfd3bf/ : SUCCESS in 2m 54s\n- openstack-tox-validate http://logs.openstack.org/71/660471/1/check/openstack-tox-validate/0c9705e/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/71/660471/1/check/requirements-tox-babel/4e0d77f/ : SUCCESS in 4m 12s\n- requirements-tox-py27-check-uc http://logs.openstack.org/71/660471/1/check/requirements-tox-py27-check-uc/5765ec5/ : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc http://logs.openstack.org/71/660471/1/check/requirements-tox-py36-check-uc/2c72a21/ : SUCCESS in 6m 25s\n- cross-cinder-py27 http://logs.openstack.org/71/660471/1/check/cross-cinder-py27/4b79a00/ : SUCCESS in 9m 45s\n- cross-cinder-py36 http://logs.openstack.org/71/660471/1/check/cross-cinder-py36/c8deeff/ : SUCCESS in 7m 37s\n- cross-glance-py27 http://logs.openstack.org/71/660471/1/check/cross-glance-py27/a9c7043/ : SUCCESS in 4m 22s\n- cross-glance-py36 http://logs.openstack.org/71/660471/1/check/cross-glance-py36/ec13f71/ : SUCCESS in 4m 38s\n- cross-horizon-py27 http://logs.openstack.org/71/660471/1/check/cross-horizon-py27/3369c92/ : SUCCESS in 10m 25s\n- cross-horizon-py36 http://logs.openstack.org/71/660471/1/check/cross-horizon-py36/6a1ffa4/ : SUCCESS in 9m 58s\n- cross-horizon-npm http://logs.openstack.org/71/660471/1/check/cross-horizon-npm/525f4b3/ : SUCCESS in 5m 37s\n- cross-keystone-py27 http://logs.openstack.org/71/660471/1/check/cross-keystone-py27/096e016/ : SUCCESS in 24m 04s\n- cross-keystone-py36 http://logs.openstack.org/71/660471/1/check/cross-keystone-py36/d53d868/ : SUCCESS in 21m 55s\n- cross-neutron-py27 http://logs.openstack.org/71/660471/1/check/cross-neutron-py27/fe7c33e/ : SUCCESS in 22m 50s\n- cross-neutron-py36 http://logs.openstack.org/71/660471/1/check/cross-neutron-py36/35164e1/ : SUCCESS in 21m 59s\n- cross-nova-py27 http://logs.openstack.org/71/660471/1/check/cross-nova-py27/dc42e63/ : SUCCESS in 15m 26s\n- cross-nova-py36 http://logs.openstack.org/71/660471/1/check/cross-nova-py36/a370313/ : SUCCESS in 16m 43s\n- cross-nova-functional http://logs.openstack.org/71/660471/1/check/cross-nova-functional/eed24d0/ : SUCCESS in 17m 46s\n- cross-osc-tox-docs http://logs.openstack.org/71/660471/1/check/cross-osc-tox-docs/b77a407/html/ : SUCCESS in 4m 59s\n- cross-swift-py27 http://logs.openstack.org/71/660471/1/check/cross-swift-py27/13750ca/ : SUCCESS in 7m 15s\n- cross-swift-py36 http://logs.openstack.org/71/660471/1/check/cross-swift-py36/3c97f95/ : SUCCESS in 7m 22s\n- requirements-integration http://logs.openstack.org/71/660471/1/check/requirements-integration/4d3bc8d/ : SUCCESS in 53m 50s\n- neutron-grenade-multinode http://logs.openstack.org/71/660471/1/check/neutron-grenade-multinode/5e889f2/ : SUCCESS in 1h 02m 11s\n- tempest-full http://logs.openstack.org/71/660471/1/check/tempest-full/a83e8ff/ : SUCCESS in 1h 36m 40s\n- neutron-grenade http://logs.openstack.org/71/660471/1/check/neutron-grenade/71f181c/ : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":1},{"id":"64127d2ff886bd39c78a751b71c503eb2eee9f9c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 05:48:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3454bf2f103b674398d4616653b3ba0610c2b2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:48:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f15fb62e2ee783c0d7128707a6fac617b831f41c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-22 06:54:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df5d146f46e125b284370aeaed71cc7d5def5a19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:39:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/71/660471/1/gate/openstack-tox-docs/376539a/html/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/71/660471/1/gate/openstack-tox-pep8/962516b/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/71/660471/1/gate/openstack-tox-py27/23e55cf/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/71/660471/1/gate/openstack-tox-py36/08b1476/ : SUCCESS in 3m 28s\n- openstack-tox-validate http://logs.openstack.org/71/660471/1/gate/openstack-tox-validate/865a632/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/71/660471/1/gate/requirements-tox-babel/5092b4c/ : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/71/660471/1/gate/requirements-tox-py27-check-uc/e8b1e92/ : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc http://logs.openstack.org/71/660471/1/gate/requirements-tox-py36-check-uc/7810989/ : SUCCESS in 5m 53s\n- cross-cinder-py27 http://logs.openstack.org/71/660471/1/gate/cross-cinder-py27/34dabcd/ : SUCCESS in 10m 24s\n- cross-cinder-py36 http://logs.openstack.org/71/660471/1/gate/cross-cinder-py36/89fe8e1/ : SUCCESS in 11m 36s\n- cross-glance-py27 http://logs.openstack.org/71/660471/1/gate/cross-glance-py27/0cadc55/ : SUCCESS in 4m 38s\n- cross-glance-py36 http://logs.openstack.org/71/660471/1/gate/cross-glance-py36/cc3efdf/ : SUCCESS in 3m 59s\n- cross-horizon-py27 http://logs.openstack.org/71/660471/1/gate/cross-horizon-py27/331d537/ : SUCCESS in 9m 38s\n- cross-horizon-py36 http://logs.openstack.org/71/660471/1/gate/cross-horizon-py36/ded2e7b/ : SUCCESS in 10m 41s\n- cross-horizon-npm http://logs.openstack.org/71/660471/1/gate/cross-horizon-npm/4a1f0ff/ : SUCCESS in 5m 16s\n- cross-keystone-py27 http://logs.openstack.org/71/660471/1/gate/cross-keystone-py27/568b260/ : SUCCESS in 22m 16s\n- cross-keystone-py36 http://logs.openstack.org/71/660471/1/gate/cross-keystone-py36/d4950f9/ : SUCCESS in 24m 00s\n- cross-neutron-py27 http://logs.openstack.org/71/660471/1/gate/cross-neutron-py27/100a7d8/ : SUCCESS in 21m 48s\n- cross-neutron-py36 http://logs.openstack.org/71/660471/1/gate/cross-neutron-py36/68bf3c7/ : SUCCESS in 31m 27s\n- cross-nova-py27 http://logs.openstack.org/71/660471/1/gate/cross-nova-py27/4173da9/ : SUCCESS in 12m 11s\n- cross-nova-py36 http://logs.openstack.org/71/660471/1/gate/cross-nova-py36/f27dde7/ : SUCCESS in 13m 15s\n- cross-nova-functional http://logs.openstack.org/71/660471/1/gate/cross-nova-functional/c466f5e/ : SUCCESS in 25m 38s\n- cross-osc-tox-docs http://logs.openstack.org/71/660471/1/gate/cross-osc-tox-docs/310cc20/html/ : SUCCESS in 5m 23s\n- cross-swift-py27 http://logs.openstack.org/71/660471/1/gate/cross-swift-py27/a60cea8/ : FAILURE in 10m 31s\n- cross-swift-py36 http://logs.openstack.org/71/660471/1/gate/cross-swift-py36/6873d5c/ : SUCCESS in 7m 52s\n- requirements-integration http://logs.openstack.org/71/660471/1/gate/requirements-integration/91717a3/ : SUCCESS in 53m 53s\n- neutron-grenade-multinode http://logs.openstack.org/71/660471/1/gate/neutron-grenade-multinode/ad9d159/ : SUCCESS in 1h 16m 57s\n- tempest-full http://logs.openstack.org/71/660471/1/gate/tempest-full/bf3be42/ : SUCCESS in 1h 54m 00s\n- neutron-grenade http://logs.openstack.org/71/660471/1/gate/neutron-grenade/2a1e14d/ : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":1},{"id":"93b09ae2689ad05a16777188e6034609a45575e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 10:18:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e9bb6e0d5581ea3797242b1acc3c9bd71d2e7b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 14:20:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/71/660471/1/check/openstack-tox-docs/617710c/html/ : SUCCESS in 2m 53s\n- openstack-tox-pypy http://logs.openstack.org/71/660471/1/check/openstack-tox-pypy/1b6b880/ : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/71/660471/1/check/openstack-tox-pep8/c31148b/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/71/660471/1/check/openstack-tox-py27/e4b4760/ : SUCCESS in 7m 12s\n- openstack-tox-py36 http://logs.openstack.org/71/660471/1/check/openstack-tox-py36/dba8f86/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/71/660471/1/check/openstack-tox-validate/3f5fca8/ : SUCCESS in 3m 19s\n- requirements-tox-babel http://logs.openstack.org/71/660471/1/check/requirements-tox-babel/d303dd5/ : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/71/660471/1/check/requirements-tox-py27-check-uc/1dc90cf/ : SUCCESS in 6m 46s\n- requirements-tox-py36-check-uc http://logs.openstack.org/71/660471/1/check/requirements-tox-py36-check-uc/97dd4bb/ : SUCCESS in 7m 28s\n- cross-cinder-py27 http://logs.openstack.org/71/660471/1/check/cross-cinder-py27/698003e/ : SUCCESS in 9m 25s\n- cross-cinder-py36 http://logs.openstack.org/71/660471/1/check/cross-cinder-py36/1abfc7c/ : SUCCESS in 8m 13s\n- cross-glance-py27 http://logs.openstack.org/71/660471/1/check/cross-glance-py27/a7cf9e4/ : SUCCESS in 5m 20s\n- cross-glance-py36 http://logs.openstack.org/71/660471/1/check/cross-glance-py36/9b457f2/ : SUCCESS in 4m 23s\n- cross-horizon-py27 http://logs.openstack.org/71/660471/1/check/cross-horizon-py27/f84d783/ : SUCCESS in 10m 22s\n- cross-horizon-py36 http://logs.openstack.org/71/660471/1/check/cross-horizon-py36/435bb56/ : SUCCESS in 9m 47s\n- cross-horizon-npm http://logs.openstack.org/71/660471/1/check/cross-horizon-npm/d062caa/ : SUCCESS in 5m 34s\n- cross-keystone-py27 http://logs.openstack.org/71/660471/1/check/cross-keystone-py27/9eac135/ : SUCCESS in 21m 35s\n- cross-keystone-py36 http://logs.openstack.org/71/660471/1/check/cross-keystone-py36/4763ca9/ : SUCCESS in 26m 13s\n- cross-neutron-py27 http://logs.openstack.org/71/660471/1/check/cross-neutron-py27/fdab384/ : SUCCESS in 22m 00s\n- cross-neutron-py36 http://logs.openstack.org/71/660471/1/check/cross-neutron-py36/3f2d966/ : SUCCESS in 23m 04s\n- cross-nova-py27 http://logs.openstack.org/71/660471/1/check/cross-nova-py27/fa9c66e/ : SUCCESS in 11m 41s\n- cross-nova-py36 http://logs.openstack.org/71/660471/1/check/cross-nova-py36/8c98e9b/ : SUCCESS in 10m 02s\n- cross-nova-functional http://logs.openstack.org/71/660471/1/check/cross-nova-functional/2dbaec4/ : SUCCESS in 19m 01s\n- cross-osc-tox-docs http://logs.openstack.org/71/660471/1/check/cross-osc-tox-docs/b62a526/html/ : SUCCESS in 5m 20s\n- cross-swift-py27 http://logs.openstack.org/71/660471/1/check/cross-swift-py27/54b82c1/ : SUCCESS in 7m 28s\n- cross-swift-py36 http://logs.openstack.org/71/660471/1/check/cross-swift-py36/8f9ca3e/ : FAILURE in 17m 47s\n- requirements-integration http://logs.openstack.org/71/660471/1/check/requirements-integration/79432a2/ : SUCCESS in 40m 59s\n- neutron-grenade-multinode http://logs.openstack.org/71/660471/1/check/neutron-grenade-multinode/6933232/ : SUCCESS in 1h 09m 31s\n- tempest-full http://logs.openstack.org/71/660471/1/check/tempest-full/6e21a3c/ : SUCCESS in 1h 44m 29s\n- neutron-grenade http://logs.openstack.org/71/660471/1/check/neutron-grenade/45af98e/ : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8ab1b0affc8a84f2d7c870352f8d94b667eafa83","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 19:39:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca95e2f207c172e03738608222423089d244088d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:38:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/71/660471/1/check/openstack-tox-docs/da691c1/html/ : SUCCESS in 3m 21s\n- openstack-tox-pypy http://logs.openstack.org/71/660471/1/check/openstack-tox-pypy/4078471/ : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/71/660471/1/check/openstack-tox-pep8/949378d/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/71/660471/1/check/openstack-tox-py27/934e881/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/71/660471/1/check/openstack-tox-py36/1e1e1f9/ : SUCCESS in 3m 07s\n- openstack-tox-validate http://logs.openstack.org/71/660471/1/check/openstack-tox-validate/e3b432f/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/71/660471/1/check/requirements-tox-babel/372536d/ : SUCCESS in 2m 39s\n- requirements-tox-py27-check-uc http://logs.openstack.org/71/660471/1/check/requirements-tox-py27-check-uc/40b8f04/ : SUCCESS in 6m 00s\n- requirements-tox-py36-check-uc http://logs.openstack.org/71/660471/1/check/requirements-tox-py36-check-uc/fdbc345/ : SUCCESS in 6m 44s\n- cross-cinder-py27 http://logs.openstack.org/71/660471/1/check/cross-cinder-py27/33dda69/ : SUCCESS in 8m 47s\n- cross-cinder-py36 http://logs.openstack.org/71/660471/1/check/cross-cinder-py36/9d46ad2/ : SUCCESS in 8m 13s\n- cross-glance-py27 http://logs.openstack.org/71/660471/1/check/cross-glance-py27/789bb6e/ : SUCCESS in 4m 25s\n- cross-glance-py36 http://logs.openstack.org/71/660471/1/check/cross-glance-py36/a6c8af9/ : SUCCESS in 4m 29s\n- cross-horizon-py27 http://logs.openstack.org/71/660471/1/check/cross-horizon-py27/32d1e02/ : SUCCESS in 10m 00s\n- cross-horizon-py36 http://logs.openstack.org/71/660471/1/check/cross-horizon-py36/62a6391/ : SUCCESS in 9m 34s\n- cross-horizon-npm http://logs.openstack.org/71/660471/1/check/cross-horizon-npm/c220d3c/ : SUCCESS in 6m 04s\n- cross-keystone-py27 http://logs.openstack.org/71/660471/1/check/cross-keystone-py27/065e311/ : SUCCESS in 20m 24s\n- cross-keystone-py36 http://logs.openstack.org/71/660471/1/check/cross-keystone-py36/566dc1d/ : SUCCESS in 25m 15s\n- cross-neutron-py27 http://logs.openstack.org/71/660471/1/check/cross-neutron-py27/4ae0b93/ : SUCCESS in 24m 05s\n- cross-neutron-py36 http://logs.openstack.org/71/660471/1/check/cross-neutron-py36/7fcd6a0/ : SUCCESS in 19m 48s\n- cross-nova-py27 http://logs.openstack.org/71/660471/1/check/cross-nova-py27/8e724de/ : SUCCESS in 14m 11s\n- cross-nova-py36 http://logs.openstack.org/71/660471/1/check/cross-nova-py36/fa114e9/ : SUCCESS in 12m 36s\n- cross-nova-functional http://logs.openstack.org/71/660471/1/check/cross-nova-functional/b19fbb0/ : SUCCESS in 24m 32s\n- cross-osc-tox-docs http://logs.openstack.org/71/660471/1/check/cross-osc-tox-docs/06282ad/html/ : SUCCESS in 4m 40s\n- cross-swift-py27 http://logs.openstack.org/71/660471/1/check/cross-swift-py27/1962ecd/ : SUCCESS in 7m 18s\n- cross-swift-py36 http://logs.openstack.org/71/660471/1/check/cross-swift-py36/fd460c4/ : SUCCESS in 7m 52s\n- requirements-integration http://logs.openstack.org/71/660471/1/check/requirements-integration/5b57e6a/ : SUCCESS in 1h 04m 04s\n- neutron-grenade-multinode http://logs.openstack.org/71/660471/1/check/neutron-grenade-multinode/6f92ce3/ : SUCCESS in 1h 09m 43s\n- tempest-full http://logs.openstack.org/71/660471/1/check/tempest-full/d195292/ : SUCCESS in 1h 31m 31s\n- neutron-grenade http://logs.openstack.org/71/660471/1/check/neutron-grenade/2c01f0a/ : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":1},{"id":"42009f696100cd1ca1eeda00f498312f47dd1102","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:38:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01ebcee2c23b7ad4001031111ccd96a308eb0b09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 10:43:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/71/660471/1/gate/openstack-tox-docs/bdbddca/html/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/71/660471/1/gate/openstack-tox-pep8/a4c5e2e/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/71/660471/1/gate/openstack-tox-py27/d84d5f2/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/71/660471/1/gate/openstack-tox-py36/a3cbcdc/ : SUCCESS in 3m 55s\n- openstack-tox-validate http://logs.openstack.org/71/660471/1/gate/openstack-tox-validate/2358d53/ : SUCCESS in 3m 52s\n- requirements-tox-babel http://logs.openstack.org/71/660471/1/gate/requirements-tox-babel/3ffc18d/ : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/71/660471/1/gate/requirements-tox-py27-check-uc/55e3dda/ : SUCCESS in 7m 17s\n- requirements-tox-py36-check-uc http://logs.openstack.org/71/660471/1/gate/requirements-tox-py36-check-uc/c02ca54/ : SUCCESS in 6m 07s\n- cross-cinder-py27 http://logs.openstack.org/71/660471/1/gate/cross-cinder-py27/f892945/ : SUCCESS in 8m 34s\n- cross-cinder-py36 http://logs.openstack.org/71/660471/1/gate/cross-cinder-py36/63f07af/ : SUCCESS in 8m 36s\n- cross-glance-py27 http://logs.openstack.org/71/660471/1/gate/cross-glance-py27/de647bd/ : SUCCESS in 3m 56s\n- cross-glance-py36 http://logs.openstack.org/71/660471/1/gate/cross-glance-py36/78207ec/ : SUCCESS in 4m 06s\n- cross-horizon-py27 http://logs.openstack.org/71/660471/1/gate/cross-horizon-py27/7e48cee/ : SUCCESS in 9m 16s\n- cross-horizon-py36 http://logs.openstack.org/71/660471/1/gate/cross-horizon-py36/f3a6127/ : SUCCESS in 9m 55s\n- cross-horizon-npm http://logs.openstack.org/71/660471/1/gate/cross-horizon-npm/2b66b09/ : SUCCESS in 5m 59s\n- cross-keystone-py27 http://logs.openstack.org/71/660471/1/gate/cross-keystone-py27/d00efd3/ : SUCCESS in 23m 42s\n- cross-keystone-py36 http://logs.openstack.org/71/660471/1/gate/cross-keystone-py36/a167b8b/ : SUCCESS in 22m 45s\n- cross-neutron-py27 http://logs.openstack.org/71/660471/1/gate/cross-neutron-py27/41b54db/ : SUCCESS in 20m 40s\n- cross-neutron-py36 http://logs.openstack.org/71/660471/1/gate/cross-neutron-py36/a62ccfb/ : SUCCESS in 21m 22s\n- cross-nova-py27 http://logs.openstack.org/71/660471/1/gate/cross-nova-py27/cb162c2/ : SUCCESS in 11m 50s\n- cross-nova-py36 http://logs.openstack.org/71/660471/1/gate/cross-nova-py36/6bf220c/ : SUCCESS in 11m 50s\n- cross-nova-functional http://logs.openstack.org/71/660471/1/gate/cross-nova-functional/528d086/ : SUCCESS in 20m 07s\n- cross-osc-tox-docs http://logs.openstack.org/71/660471/1/gate/cross-osc-tox-docs/838eba1/html/ : SUCCESS in 5m 28s\n- cross-swift-py27 http://logs.openstack.org/71/660471/1/gate/cross-swift-py27/d9dc491/ : SUCCESS in 7m 28s\n- cross-swift-py36 http://logs.openstack.org/71/660471/1/gate/cross-swift-py36/2d16d1f/ : SUCCESS in 8m 03s\n- requirements-integration http://logs.openstack.org/71/660471/1/gate/requirements-integration/5077475/ : SUCCESS in 56m 40s\n- neutron-grenade-multinode http://logs.openstack.org/71/660471/1/gate/neutron-grenade-multinode/ed3c351/ : SUCCESS in 1h 15m 52s\n- tempest-full http://logs.openstack.org/71/660471/1/gate/tempest-full/2274a98/ : SUCCESS in 1h 32m 03s\n- neutron-grenade http://logs.openstack.org/71/660471/1/gate/neutron-grenade/0fdc326/ : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"23a08d7f258f32d0688e342ef1a7710df89b6f68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 10:43:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e90706b468c103bdc1aba43d55a613e09a68f66e","revisions":{"e90706b468c103bdc1aba43d55a613e09a68f66e":{"kind":"REWORK","_number":1,"created":"2019-05-21 17:31:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/660471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/660471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/660471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/660471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/660471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/660471/1"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:31:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:31:22.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 2.34.0","message":"update constraint for oslo.vmware to new release 2.34.0\n\nChange-Id: I882a072d711d93dd4b92f2690d79b265b7ab58e4\nmeta:version: 2.34.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I47c78915181ece3ecea576028edce3e90203d87e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e90706b468c103bdc1aba43d55a613e09a68f66e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e90706b468c103bdc1aba43d55a613e09a68f66e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
