)]}'
{"id":"openstack%2Frequirements~660476","triplet_id":"openstack%2Frequirements~master~I1311aecfcf1bc69cab6558b71eac83ef73deb2c3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1311aecfcf1bc69cab6558b71eac83ef73deb2c3","subject":"update constraint for ironic-lib to new release 2.17.0","status":"MERGED","created":"2019-05-21 17:37:35.000000000","updated":"2019-05-22 09:52:51.000000000","submitted":"2019-05-22 09:52:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660476-1558518771481-54bd5aa7","meta_rev_id":"b12186f829a165e0f2b13b3c18bb9d6428dc5173","_number":660476,"virtual_id_number":660476,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-22 09:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-22 05:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-22 05:48:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-22 06:31:26.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 05:48:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 06:31:26.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-05-22 09:52:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d26cc71b3bacdc50f8b78a06d3a0df18d12fd922","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-21 17:37:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76bff21e719bdb925cf97d8a27b04bed27c01b67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 23:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/660476/1/check/openstack-tox-docs/a399504/html/ : SUCCESS in 2m 39s\n- openstack-tox-pypy http://logs.openstack.org/76/660476/1/check/openstack-tox-pypy/d217946/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/76/660476/1/check/openstack-tox-pep8/90725c3/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/76/660476/1/check/openstack-tox-py27/e706d69/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/76/660476/1/check/openstack-tox-py36/1d51677/ : SUCCESS in 3m 30s\n- openstack-tox-validate http://logs.openstack.org/76/660476/1/check/openstack-tox-validate/559899d/ : SUCCESS in 3m 21s\n- requirements-tox-babel http://logs.openstack.org/76/660476/1/check/requirements-tox-babel/0b89587/ : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc http://logs.openstack.org/76/660476/1/check/requirements-tox-py27-check-uc/8adcc4c/ : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc http://logs.openstack.org/76/660476/1/check/requirements-tox-py36-check-uc/470a060/ : SUCCESS in 6m 25s\n- cross-cinder-py27 http://logs.openstack.org/76/660476/1/check/cross-cinder-py27/0fe9a88/ : SUCCESS in 8m 49s\n- cross-cinder-py36 http://logs.openstack.org/76/660476/1/check/cross-cinder-py36/3f7c05d/ : SUCCESS in 8m 36s\n- cross-glance-py27 http://logs.openstack.org/76/660476/1/check/cross-glance-py27/2440087/ : SUCCESS in 3m 43s\n- cross-glance-py36 http://logs.openstack.org/76/660476/1/check/cross-glance-py36/64e9abe/ : SUCCESS in 3m 49s\n- cross-horizon-py27 http://logs.openstack.org/76/660476/1/check/cross-horizon-py27/b228321/ : SUCCESS in 10m 39s\n- cross-horizon-py36 http://logs.openstack.org/76/660476/1/check/cross-horizon-py36/e63597c/ : SUCCESS in 9m 08s\n- cross-horizon-npm http://logs.openstack.org/76/660476/1/check/cross-horizon-npm/faa5355/ : SUCCESS in 6m 15s\n- cross-keystone-py27 http://logs.openstack.org/76/660476/1/check/cross-keystone-py27/6b510ea/ : SUCCESS in 23m 56s\n- cross-keystone-py36 http://logs.openstack.org/76/660476/1/check/cross-keystone-py36/0004d35/ : SUCCESS in 20m 23s\n- cross-neutron-py27 http://logs.openstack.org/76/660476/1/check/cross-neutron-py27/e31f86b/ : SUCCESS in 23m 39s\n- cross-neutron-py36 http://logs.openstack.org/76/660476/1/check/cross-neutron-py36/426907d/ : SUCCESS in 20m 08s\n- cross-nova-py27 http://logs.openstack.org/76/660476/1/check/cross-nova-py27/454c2ff/ : SUCCESS in 12m 51s\n- cross-nova-py36 http://logs.openstack.org/76/660476/1/check/cross-nova-py36/ebe68b0/ : SUCCESS in 11m 06s\n- cross-nova-functional http://logs.openstack.org/76/660476/1/check/cross-nova-functional/da9eefb/ : SUCCESS in 18m 08s\n- cross-osc-tox-docs http://logs.openstack.org/76/660476/1/check/cross-osc-tox-docs/14bb8e6/html/ : SUCCESS in 6m 40s\n- cross-swift-py27 http://logs.openstack.org/76/660476/1/check/cross-swift-py27/fa806a9/ : SUCCESS in 7m 14s\n- cross-swift-py36 http://logs.openstack.org/76/660476/1/check/cross-swift-py36/2e319d7/ : SUCCESS in 7m 22s\n- requirements-integration http://logs.openstack.org/76/660476/1/check/requirements-integration/6b2080c/ : SUCCESS in 55m 00s\n- neutron-grenade-multinode http://logs.openstack.org/76/660476/1/check/neutron-grenade-multinode/e13ff85/ : SUCCESS in 1h 11m 26s\n- tempest-full http://logs.openstack.org/76/660476/1/check/tempest-full/1678807/ : SUCCESS in 1h 43m 15s\n- neutron-grenade http://logs.openstack.org/76/660476/1/check/neutron-grenade/22f1cae/ : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d8936665fb6f9e9ef13851ff4bfedd4548b37dbe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 05:48:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c911ad2ef2c0db15ae93177c056b822d0c98cd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:48:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26e51f9d222de63cd0636cac3081750cc63490e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:52:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ce3e9da84c90da855c02a4325c3fb1bb32713bdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:52:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/660476/1/gate/openstack-tox-docs/24ccc5d/html/ : SUCCESS in 3m 20s\n- openstack-tox-pep8 http://logs.openstack.org/76/660476/1/gate/openstack-tox-pep8/af43d14/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/76/660476/1/gate/openstack-tox-py27/1df9906/ : SUCCESS in 2m 50s\n- openstack-tox-py36 http://logs.openstack.org/76/660476/1/gate/openstack-tox-py36/bda2268/ : SUCCESS in 2m 54s\n- openstack-tox-validate http://logs.openstack.org/76/660476/1/gate/openstack-tox-validate/2172433/ : SUCCESS in 3m 12s\n- requirements-tox-babel http://logs.openstack.org/76/660476/1/gate/requirements-tox-babel/679c284/ : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/76/660476/1/gate/requirements-tox-py27-check-uc/233e5d1/ : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc http://logs.openstack.org/76/660476/1/gate/requirements-tox-py36-check-uc/4bb6baf/ : SUCCESS in 6m 02s\n- cross-cinder-py27 http://logs.openstack.org/76/660476/1/gate/cross-cinder-py27/420f9c0/ : SUCCESS in 8m 10s\n- cross-cinder-py36 http://logs.openstack.org/76/660476/1/gate/cross-cinder-py36/f8d33b8/ : SUCCESS in 8m 33s\n- cross-glance-py27 http://logs.openstack.org/76/660476/1/gate/cross-glance-py27/8d016a5/ : SUCCESS in 4m 42s\n- cross-glance-py36 http://logs.openstack.org/76/660476/1/gate/cross-glance-py36/3c610b7/ : SUCCESS in 4m 32s\n- cross-horizon-py27 http://logs.openstack.org/76/660476/1/gate/cross-horizon-py27/764373c/ : SUCCESS in 11m 34s\n- cross-horizon-py36 http://logs.openstack.org/76/660476/1/gate/cross-horizon-py36/ba2aa92/ : SUCCESS in 9m 00s\n- cross-horizon-npm http://logs.openstack.org/76/660476/1/gate/cross-horizon-npm/c55a7fc/ : SUCCESS in 6m 02s\n- cross-keystone-py27 http://logs.openstack.org/76/660476/1/gate/cross-keystone-py27/8553463/ : SUCCESS in 24m 47s\n- cross-keystone-py36 http://logs.openstack.org/76/660476/1/gate/cross-keystone-py36/13a6acf/ : SUCCESS in 24m 34s\n- cross-neutron-py27 http://logs.openstack.org/76/660476/1/gate/cross-neutron-py27/8e05eff/ : SUCCESS in 23m 28s\n- cross-neutron-py36 http://logs.openstack.org/76/660476/1/gate/cross-neutron-py36/cf71f68/ : SUCCESS in 21m 02s\n- cross-nova-py27 http://logs.openstack.org/76/660476/1/gate/cross-nova-py27/f328594/ : SUCCESS in 13m 33s\n- cross-nova-py36 http://logs.openstack.org/76/660476/1/gate/cross-nova-py36/b7820b7/ : SUCCESS in 11m 29s\n- cross-nova-functional http://logs.openstack.org/76/660476/1/gate/cross-nova-functional/3f01f59/ : SUCCESS in 20m 11s\n- cross-osc-tox-docs http://logs.openstack.org/76/660476/1/gate/cross-osc-tox-docs/0b8eb80/html/ : SUCCESS in 5m 33s\n- cross-swift-py27 http://logs.openstack.org/76/660476/1/gate/cross-swift-py27/6ca1f05/ : SUCCESS in 7m 55s\n- cross-swift-py36 http://logs.openstack.org/76/660476/1/gate/cross-swift-py36/9bf4509/ : SUCCESS in 7m 55s\n- requirements-integration http://logs.openstack.org/76/660476/1/gate/requirements-integration/d71f83a/ : SUCCESS in 52m 35s\n- neutron-grenade-multinode http://logs.openstack.org/76/660476/1/gate/neutron-grenade-multinode/0f2e4ad/ : SUCCESS in 1h 06m 42s\n- tempest-full http://logs.openstack.org/76/660476/1/gate/tempest-full/5eb7082/ : SUCCESS in 1h 42m 59s\n- neutron-grenade http://logs.openstack.org/76/660476/1/gate/neutron-grenade/087a7a5/ : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"039d743bf58d915ac5f0c378b9cc81bf26ac6a47","revisions":{"039d743bf58d915ac5f0c378b9cc81bf26ac6a47":{"kind":"REWORK","_number":1,"created":"2019-05-21 17:37:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/660476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/660476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/660476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/660476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/660476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/660476/1"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:37:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:37:34.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.17.0","message":"update constraint for ironic-lib to new release 2.17.0\n\nChange-Id: I1311aecfcf1bc69cab6558b71eac83ef73deb2c3\nmeta:version: 2.17.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Dmitry Tantsur \u003cdivius.inside@gmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdivius.inside@gmail.com\u003e\nmeta:release:Change-Id: I4e64d0cf890f3b2f62da552c5b604426b7764d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/039d743bf58d915ac5f0c378b9cc81bf26ac6a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/039d743bf58d915ac5f0c378b9cc81bf26ac6a47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
