)]}'
{"id":"openstack%2Frequirements~660480","triplet_id":"openstack%2Frequirements~master~If5d9910502f48a0ea5c7cde3a919507fcef340c6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If5d9910502f48a0ea5c7cde3a919507fcef340c6","subject":"update constraint for oslo.db to new release 5.0.0","status":"MERGED","created":"2019-05-21 17:44:14.000000000","updated":"2019-05-22 10:05:19.000000000","submitted":"2019-05-22 10:05:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660480-1558519519099-16b933f0","meta_rev_id":"174b3f9604aa3744259c429dbd77eddd7e5a31eb","_number":660480,"virtual_id_number":660480,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-22 10:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-22 05:48:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-22 06:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-22 05:48:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 05:48:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 06:55:25.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-22 10:05:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a6c14999ce70a70bc1420a1b0f27de76a7eb9f7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-21 17:44:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b8958f0e7ed4ba3837892460f690d9985946eec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 00:15:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/660480/1/check/openstack-tox-docs/f8ea715/html/ : SUCCESS in 2m 59s\n- openstack-tox-pypy http://logs.openstack.org/80/660480/1/check/openstack-tox-pypy/d61c77f/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/660480/1/check/openstack-tox-pep8/7541cbf/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/80/660480/1/check/openstack-tox-py27/8aaeadb/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/80/660480/1/check/openstack-tox-py36/e0a88af/ : SUCCESS in 2m 41s\n- openstack-tox-validate http://logs.openstack.org/80/660480/1/check/openstack-tox-validate/c2e29e9/ : SUCCESS in 2m 39s\n- requirements-tox-babel http://logs.openstack.org/80/660480/1/check/requirements-tox-babel/30c53c3/ : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/80/660480/1/check/requirements-tox-py27-check-uc/e128439/ : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/80/660480/1/check/requirements-tox-py36-check-uc/2d0a2b5/ : SUCCESS in 5m 50s\n- cross-cinder-py27 http://logs.openstack.org/80/660480/1/check/cross-cinder-py27/c97b4e1/ : SUCCESS in 8m 51s\n- cross-cinder-py36 http://logs.openstack.org/80/660480/1/check/cross-cinder-py36/397d4a3/ : SUCCESS in 8m 20s\n- cross-glance-py27 http://logs.openstack.org/80/660480/1/check/cross-glance-py27/9f73c53/ : SUCCESS in 4m 34s\n- cross-glance-py36 http://logs.openstack.org/80/660480/1/check/cross-glance-py36/bb8f73a/ : SUCCESS in 4m 09s\n- cross-horizon-py27 http://logs.openstack.org/80/660480/1/check/cross-horizon-py27/7a93a8d/ : SUCCESS in 10m 29s\n- cross-horizon-py36 http://logs.openstack.org/80/660480/1/check/cross-horizon-py36/f75a1ea/ : SUCCESS in 9m 19s\n- cross-horizon-npm http://logs.openstack.org/80/660480/1/check/cross-horizon-npm/e4774fe/ : SUCCESS in 5m 18s\n- cross-keystone-py27 http://logs.openstack.org/80/660480/1/check/cross-keystone-py27/778768b/ : SUCCESS in 19m 43s\n- cross-keystone-py36 http://logs.openstack.org/80/660480/1/check/cross-keystone-py36/01ba7ed/ : SUCCESS in 25m 31s\n- cross-neutron-py27 http://logs.openstack.org/80/660480/1/check/cross-neutron-py27/5280d7d/ : SUCCESS in 22m 08s\n- cross-neutron-py36 http://logs.openstack.org/80/660480/1/check/cross-neutron-py36/eb5e120/ : SUCCESS in 21m 37s\n- cross-nova-py27 http://logs.openstack.org/80/660480/1/check/cross-nova-py27/64a2fa9/ : SUCCESS in 12m 05s\n- cross-nova-py36 http://logs.openstack.org/80/660480/1/check/cross-nova-py36/93d77c9/ : SUCCESS in 12m 13s\n- cross-nova-functional http://logs.openstack.org/80/660480/1/check/cross-nova-functional/c53318b/ : SUCCESS in 18m 00s\n- cross-osc-tox-docs http://logs.openstack.org/80/660480/1/check/cross-osc-tox-docs/29f8c0e/html/ : SUCCESS in 4m 42s\n- cross-swift-py27 http://logs.openstack.org/80/660480/1/check/cross-swift-py27/17df79f/ : SUCCESS in 8m 02s\n- cross-swift-py36 http://logs.openstack.org/80/660480/1/check/cross-swift-py36/f78e467/ : SUCCESS in 7m 59s\n- requirements-integration http://logs.openstack.org/80/660480/1/check/requirements-integration/921ecc7/ : SUCCESS in 53m 20s\n- neutron-grenade-multinode http://logs.openstack.org/80/660480/1/check/neutron-grenade-multinode/a2f1de9/ : SUCCESS in 1h 13m 50s\n- tempest-full http://logs.openstack.org/80/660480/1/check/tempest-full/7b75136/ : SUCCESS in 1h 57m 35s\n- neutron-grenade http://logs.openstack.org/80/660480/1/check/neutron-grenade/75818f1/ : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"88cdc472c82bd4e9bc0e5167a748f285d521ee51","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 05:48:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42949d3b4220cee89696a552721714aff50532d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:49:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80c0409905fe355df50631bb60605981fb5f07e4","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-22 06:55:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f776568e81eb61cdd29a064a6360f686753de3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:05:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/660480/1/gate/openstack-tox-docs/3ef32aa/html/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/80/660480/1/gate/openstack-tox-pep8/e95dff3/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/80/660480/1/gate/openstack-tox-py27/72295e0/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/80/660480/1/gate/openstack-tox-py36/a94efc6/ : SUCCESS in 4m 57s\n- openstack-tox-validate http://logs.openstack.org/80/660480/1/gate/openstack-tox-validate/8e6cd59/ : SUCCESS in 3m 26s\n- requirements-tox-babel http://logs.openstack.org/80/660480/1/gate/requirements-tox-babel/6945ad1/ : SUCCESS in 4m 25s\n- requirements-tox-py27-check-uc http://logs.openstack.org/80/660480/1/gate/requirements-tox-py27-check-uc/af52fc9/ : SUCCESS in 7m 04s\n- requirements-tox-py36-check-uc http://logs.openstack.org/80/660480/1/gate/requirements-tox-py36-check-uc/ae24988/ : SUCCESS in 5m 31s\n- cross-cinder-py27 http://logs.openstack.org/80/660480/1/gate/cross-cinder-py27/3b086e5/ : SUCCESS in 11m 01s\n- cross-cinder-py36 http://logs.openstack.org/80/660480/1/gate/cross-cinder-py36/e9b3b41/ : SUCCESS in 10m 27s\n- cross-glance-py27 http://logs.openstack.org/80/660480/1/gate/cross-glance-py27/7d6236a/ : SUCCESS in 4m 11s\n- cross-glance-py36 http://logs.openstack.org/80/660480/1/gate/cross-glance-py36/1f9098c/ : SUCCESS in 4m 32s\n- cross-horizon-py27 http://logs.openstack.org/80/660480/1/gate/cross-horizon-py27/ebc6677/ : SUCCESS in 9m 44s\n- cross-horizon-py36 http://logs.openstack.org/80/660480/1/gate/cross-horizon-py36/ce274aa/ : SUCCESS in 9m 44s\n- cross-horizon-npm http://logs.openstack.org/80/660480/1/gate/cross-horizon-npm/590ae94/ : SUCCESS in 6m 47s\n- cross-keystone-py27 http://logs.openstack.org/80/660480/1/gate/cross-keystone-py27/5a1bcf7/ : SUCCESS in 27m 53s\n- cross-keystone-py36 http://logs.openstack.org/80/660480/1/gate/cross-keystone-py36/352d265/ : SUCCESS in 26m 36s\n- cross-neutron-py27 http://logs.openstack.org/80/660480/1/gate/cross-neutron-py27/9a806d9/ : SUCCESS in 22m 07s\n- cross-neutron-py36 http://logs.openstack.org/80/660480/1/gate/cross-neutron-py36/ff41b1d/ : SUCCESS in 23m 00s\n- cross-nova-py27 http://logs.openstack.org/80/660480/1/gate/cross-nova-py27/8337457/ : SUCCESS in 14m 11s\n- cross-nova-py36 http://logs.openstack.org/80/660480/1/gate/cross-nova-py36/77f1f1d/ : SUCCESS in 12m 06s\n- cross-nova-functional http://logs.openstack.org/80/660480/1/gate/cross-nova-functional/88fd755/ : SUCCESS in 24m 38s\n- cross-osc-tox-docs http://logs.openstack.org/80/660480/1/gate/cross-osc-tox-docs/b04ee17/html/ : SUCCESS in 5m 39s\n- cross-swift-py27 http://logs.openstack.org/80/660480/1/gate/cross-swift-py27/e672686/ : SUCCESS in 8m 15s\n- cross-swift-py36 http://logs.openstack.org/80/660480/1/gate/cross-swift-py36/04a6f12/ : SUCCESS in 8m 08s\n- requirements-integration http://logs.openstack.org/80/660480/1/gate/requirements-integration/5c69db0/ : SUCCESS in 1h 00m 01s\n- neutron-grenade-multinode http://logs.openstack.org/80/660480/1/gate/neutron-grenade-multinode/e800171/ : SUCCESS in 59m 32s\n- tempest-full http://logs.openstack.org/80/660480/1/gate/tempest-full/be5b8f4/ : SUCCESS in 2h 04m 26s\n- neutron-grenade http://logs.openstack.org/80/660480/1/gate/neutron-grenade/d0d7432/ : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":1},{"id":"362ea3e821bdcea731dafa55b3dededae460d713","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:05:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65c7c0841436b192b1a1ea10b15cec092d12a3de","revisions":{"65c7c0841436b192b1a1ea10b15cec092d12a3de":{"kind":"REWORK","_number":1,"created":"2019-05-21 17:44:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/660480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/660480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/660480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/660480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/660480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/660480/1"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:44:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-21 17:44:12.000000000","tz":0},"subject":"update constraint for oslo.db to new release 5.0.0","message":"update constraint for oslo.db to new release 5.0.0\n\nChange-Id: If5d9910502f48a0ea5c7cde3a919507fcef340c6\nmeta:version: 5.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I47c78915181ece3ecea576028edce3e90203d87e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65c7c0841436b192b1a1ea10b15cec092d12a3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65c7c0841436b192b1a1ea10b15cec092d12a3de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
