)]}'
{"id":"openstack%2Frequirements~660807","triplet_id":"openstack%2Frequirements~master~I0e70e825a34b30a62e1bd32fbdc6b35f9641eb18","project":"openstack/requirements","branch":"master","topic":"castellan-ui-retirement","hashtags":[],"change_id":"I0e70e825a34b30a62e1bd32fbdc6b35f9641eb18","subject":"Stop Requirements Sync for castellan-ui","status":"MERGED","created":"2019-05-22 19:13:07.000000000","updated":"2019-05-23 22:30:53.000000000","submitted":"2019-05-23 22:30:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660807-1558650653444-95f0d4b7","meta_rev_id":"c4e273f179300be8321d97e423d92b84d9172921","_number":660807,"virtual_id_number":660807,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-23 22:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":2,"date":"2019-05-23 15:49:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-23 06:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":1,"date":"2019-05-23 15:49:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 06:23:53.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-23 15:49:39.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-23 22:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6e7f6931041dae69918946beed82152f9bd2e0a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-05-22 19:13:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e440a04389a9ad9c898d2826c4b9a478c0e79ee8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:19:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/07/660807/1/check/openstack-tox-docs/fcd52c8/html/ : SUCCESS in 2m 56s\n- openstack-tox-pypy http://logs.openstack.org/07/660807/1/check/openstack-tox-pypy/7fa882b/ : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/660807/1/check/openstack-tox-pep8/75e283b/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/07/660807/1/check/openstack-tox-py27/60843e4/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/07/660807/1/check/openstack-tox-py36/8900368/ : SUCCESS in 3m 21s\n- openstack-tox-validate http://logs.openstack.org/07/660807/1/check/openstack-tox-validate/f4d18c1/ : SUCCESS in 2m 46s\n- requirements-tox-validate-projects http://logs.openstack.org/07/660807/1/check/requirements-tox-validate-projects/05512c3/ : FAILURE in 2m 43s (non-voting)\n- requirements-integration http://logs.openstack.org/07/660807/1/check/requirements-integration/32930d1/ : SUCCESS in 55m 06s\n- neutron-grenade-multinode http://logs.openstack.org/07/660807/1/check/neutron-grenade-multinode/ea50fdf/ : SUCCESS in 1h 06m 11s\n- tempest-full http://logs.openstack.org/07/660807/1/check/tempest-full/a0cbe4b/ : FAILURE in 1h 40m 47s\n- neutron-grenade http://logs.openstack.org/07/660807/1/check/neutron-grenade/4dc6cf1/ : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":1},{"id":"44a0c1686a742aa1d3077751965891f5e5bb1049","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:23:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"882a650141bd7a6e5e01b3bbecd2061f29a3b647","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:23:53.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks\n\nLooks good to me","accounts_in_message":[],"_revision_number":1},{"id":"ef355509c6cbf549e7c57dac4ba8f6e8792c3e1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 10:08:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/07/660807/1/check/openstack-tox-docs/160e4e6/html/ : SUCCESS in 2m 51s\n- openstack-tox-pypy http://logs.openstack.org/07/660807/1/check/openstack-tox-pypy/7ab69fb/ : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/660807/1/check/openstack-tox-pep8/8876829/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/07/660807/1/check/openstack-tox-py27/783b679/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/07/660807/1/check/openstack-tox-py36/eebada3/ : SUCCESS in 3m 35s\n- openstack-tox-validate http://logs.openstack.org/07/660807/1/check/openstack-tox-validate/c6a9c0b/ : SUCCESS in 3m 08s\n- requirements-tox-validate-projects http://logs.openstack.org/07/660807/1/check/requirements-tox-validate-projects/a2092b4/ : FAILURE in 2m 39s (non-voting)\n- requirements-integration http://logs.openstack.org/07/660807/1/check/requirements-integration/3d5ec76/ : SUCCESS in 52m 09s\n- neutron-grenade-multinode http://logs.openstack.org/07/660807/1/check/neutron-grenade-multinode/aba3f54/ : SUCCESS in 1h 00m 21s\n- tempest-full http://logs.openstack.org/07/660807/1/check/tempest-full/793da53/ : SUCCESS in 1h 38m 14s\n- neutron-grenade http://logs.openstack.org/07/660807/1/check/neutron-grenade/667b780/ : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9388498df65cc57217bc9d7cf2cb3799dc398ff0","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-23 15:49:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e544da1a3d4c9254a874efef70107e2c49981fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:50:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef40f6abf2a706beb60a0e40d6f314f943cf0fdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 22:30:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/07/660807/1/gate/openstack-tox-docs/6d7f33a/html/ : SUCCESS in 3m 08s\n- openstack-tox-pep8 http://logs.openstack.org/07/660807/1/gate/openstack-tox-pep8/df1be7f/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/07/660807/1/gate/openstack-tox-py27/722e829/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/07/660807/1/gate/openstack-tox-py36/a7b2786/ : SUCCESS in 4m 04s\n- openstack-tox-validate http://logs.openstack.org/07/660807/1/gate/openstack-tox-validate/6ab6d81/ : SUCCESS in 4m 41s\n- requirements-integration http://logs.openstack.org/07/660807/1/gate/requirements-integration/1d22c22/ : SUCCESS in 1h 00m 53s\n- neutron-grenade-multinode http://logs.openstack.org/07/660807/1/gate/neutron-grenade-multinode/ed7bc9b/ : SUCCESS in 1h 10m 45s\n- tempest-full http://logs.openstack.org/07/660807/1/gate/tempest-full/c6bf817/ : SUCCESS in 1h 41m 52s\n- neutron-grenade http://logs.openstack.org/07/660807/1/gate/neutron-grenade/b338bc5/ : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":1},{"id":"96ba688b32afed70f4fbaa52cf6c5fb7bbb31169","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 22:30:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0675456145167cbb5f9ae933bd6974ff3a830dfa","revisions":{"0675456145167cbb5f9ae933bd6974ff3a830dfa":{"kind":"REWORK","_number":1,"created":"2019-05-22 19:13:07.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/07/660807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/660807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/660807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/660807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/660807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/660807/1"}}},"commit":{"parents":[{"commit":"384b6a1143188606b6d3247db0c309a8c72b0c1a","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/384b6a1143188606b6d3247db0c309a8c72b0c1a"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2019-05-22 19:11:27.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2019-05-22 19:13:06.000000000","tz":-300},"subject":"Stop Requirements Sync for castellan-ui","message":"Stop Requirements Sync for castellan-ui\n\ncastellan-ui is being retired. [1]  This patch stops\nrequirements syncing for the project.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006509.html\n\nChange-Id: I0e70e825a34b30a62e1bd32fbdc6b35f9641eb18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0675456145167cbb5f9ae933bd6974ff3a830dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0675456145167cbb5f9ae933bd6974ff3a830dfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
