)]}'
{"id":"openstack%2Frequirements~661144","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ia61d944dbfdf494598e41e5fbefd0e32a91814ed","project":"openstack/requirements","branch":"stable/stein","topic":"u-c","hashtags":[],"change_id":"Ia61d944dbfdf494598e41e5fbefd0e32a91814ed","subject":"Fix where we use constraints","status":"MERGED","created":"2019-05-24 00:26:35.000000000","updated":"2019-05-24 16:50:34.000000000","submitted":"2019-05-24 16:50:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661144-1558716634245-2234bc6b","meta_rev_id":"92b2ba7e23d9b4407cbd06e8075714a851cbdf9c","_number":661144,"virtual_id_number":661144,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-24 16:50:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-24 03:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-24 13:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-24 13:55:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 03:44:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-24 13:55:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-24 16:50:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10c7249cf55cabe9a0a287baab22782e342672b6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-24 00:26:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5406ff5d5eeeab50c291cbc4e985051df4cd3e32","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-24 00:26:49.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 3c5cb3e854dfdcf5c223f9ed13ca74d66394bcd8","accounts_in_message":[],"_revision_number":1},{"id":"f13b262dcdeb1db219336b81bc9e92f7b3e794a4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-24 00:28:35.000000000","message":"Topic changed from u-c-stable/stein to u-c","accounts_in_message":[],"_revision_number":1},{"id":"7b9d16f212fff4d4ed9a367f9e31e766571175aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 01:59:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/44/661144/1/check/openstack-tox-pypy/80370ed/ : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/44/661144/1/check/openstack-tox-pep8/c45c736/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/44/661144/1/check/openstack-tox-py27/9a2b678/ : SUCCESS in 3m 22s\n- openstack-tox-py35 http://logs.openstack.org/44/661144/1/check/openstack-tox-py35/4c6a89f/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/44/661144/1/check/openstack-tox-py36/a4276f4/ : SUCCESS in 5m 10s\n- openstack-tox-validate http://logs.openstack.org/44/661144/1/check/openstack-tox-validate/d15ef64/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/44/661144/1/check/requirements-tox-babel/a72992e/ : SUCCESS in 2m 36s\n- requirements-integration http://logs.openstack.org/44/661144/1/check/requirements-integration/8225598/ : SUCCESS in 51m 20s\n- neutron-grenade-multinode http://logs.openstack.org/44/661144/1/check/neutron-grenade-multinode/504bfe3/ : SUCCESS in 1h 06m 57s\n- tempest-full http://logs.openstack.org/44/661144/1/check/tempest-full/0b64e73/ : FAILURE in 1h 26m 32s\n- neutron-grenade http://logs.openstack.org/44/661144/1/check/neutron-grenade/9ece4e7/ : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"66ac2df6a6012a88b5e2c045dd5f48f3dcbb2eac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 02:07:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c4e3654af8b09eee0aa1de3e3a1b8ceacab96bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 03:39:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/44/661144/1/check/openstack-tox-pypy/49a129e/ : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/44/661144/1/check/openstack-tox-pep8/bcaae8b/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/44/661144/1/check/openstack-tox-py27/10c3c65/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/44/661144/1/check/openstack-tox-py35/3ec27d7/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/44/661144/1/check/openstack-tox-py36/edba927/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/44/661144/1/check/openstack-tox-validate/ed2fcf7/ : SUCCESS in 3m 23s\n- requirements-tox-babel http://logs.openstack.org/44/661144/1/check/requirements-tox-babel/b687b1c/ : SUCCESS in 2m 52s\n- requirements-integration http://logs.openstack.org/44/661144/1/check/requirements-integration/d2015de/ : SUCCESS in 51m 19s\n- neutron-grenade-multinode http://logs.openstack.org/44/661144/1/check/neutron-grenade-multinode/28c7185/ : SUCCESS in 1h 03m 23s\n- tempest-full http://logs.openstack.org/44/661144/1/check/tempest-full/9764f67/ : SUCCESS in 1h 31m 16s\n- neutron-grenade http://logs.openstack.org/44/661144/1/check/neutron-grenade/18b99c7/ : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":1},{"id":"10f6c165077807a244a3833ec6eaadfcfa2d3282","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 03:44:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"426ace7b0ff1b5450847921544933e2317e1d595","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-24 13:55:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f99310798d13ecc6cb15b36bd86cdc3ce3b4283","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 13:56:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"651239220189e0e93e9ea37ebe51e1e8a849da9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 16:50:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/661144/1/gate/openstack-tox-pep8/f9b9f90/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/44/661144/1/gate/openstack-tox-py27/8eb09f0/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/44/661144/1/gate/openstack-tox-py35/33dc1b8/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/44/661144/1/gate/openstack-tox-py36/aae2754/ : SUCCESS in 3m 47s\n- openstack-tox-validate http://logs.openstack.org/44/661144/1/gate/openstack-tox-validate/40e57f7/ : SUCCESS in 3m 40s\n- requirements-tox-babel http://logs.openstack.org/44/661144/1/gate/requirements-tox-babel/76f966a/ : SUCCESS in 3m 24s\n- requirements-integration http://logs.openstack.org/44/661144/1/gate/requirements-integration/5555335/ : SUCCESS in 58m 27s\n- neutron-grenade-multinode http://logs.openstack.org/44/661144/1/gate/neutron-grenade-multinode/b1148ba/ : SUCCESS in 1h 13m 34s\n- tempest-full http://logs.openstack.org/44/661144/1/gate/tempest-full/dd3828d/ : SUCCESS in 1h 46m 50s\n- neutron-grenade http://logs.openstack.org/44/661144/1/gate/neutron-grenade/ecd1cef/ : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"92b2ba7e23d9b4407cbd06e8075714a851cbdf9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 16:50:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73f491ffb50fc29db552c1655639fac45b9056e1","revisions":{"73f491ffb50fc29db552c1655639fac45b9056e1":{"kind":"REWORK","_number":1,"created":"2019-05-24 00:26:35.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/44/661144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/661144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/661144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/661144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/661144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/661144/1"}}},"commit":{"parents":[{"commit":"2a837a3e58f47560c75629ccd32f870247262004","subject":"update constraint for python-ironicclient to new release 2.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a837a3e58f47560c75629ccd32f870247262004"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-16 08:23:46.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-24 00:26:35.000000000","tz":0},"subject":"Fix where we use constraints","message":"Fix where we use constraints\n\nin I2598503c2d4397d7efbb03a76df9a4e91dd0c1cf we mistakenly added\nconstraints to install_command in out based testenv.  This is incorrect\nas it means the generate env will be run in a constrained environment.\n\nWe don\u0027t use UPPER_CONSTRAINT_FILE as *right now* it\u0027s the same thing\nand this leaves us internally consistent (even if technically incorrect)\n\nChange-Id: Ia61d944dbfdf494598e41e5fbefd0e32a91814ed\n(cherry picked from commit f43647b3edf8024669f378f1943278081ce9bd85)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73f491ffb50fc29db552c1655639fac45b9056e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73f491ffb50fc29db552c1655639fac45b9056e1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
