)]}'
{"id":"openstack%2Frequirements~661149","triplet_id":"openstack%2Frequirements~master~I3e91d25cc6f9071fcc7314cfd34152c5f99a4db6","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I3e91d25cc6f9071fcc7314cfd34152c5f99a4db6","subject":"Ignore python 3.4 and 3.5 for sphinx","status":"ABANDONED","created":"2019-05-24 02:37:56.000000000","updated":"2019-06-05 05:44:04.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7b3f01a1557fa588837cbf16b12dc7ba6d435e02","_number":661149,"virtual_id_number":661149,"owner":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"date":"2019-05-28 01:38:57.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":1,"date":"2019-05-24 04:35:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":-1,"date":"2019-05-24 03:05:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2019-05-24 03:52:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 03:05:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-24 03:52:27.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-24 04:35:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c3b3ba4c2b3bfb95f1f126a08cea840915db04f","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-24 02:37:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cefa24b4920a175dec3c82491b934cd4665e4ed2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 02:59:32.000000000","message":"Patch Set 1: Code-Review-2\n\nsphinx still works on py2 though, while we are moving away from py27, we are not forcing anything yet to my knowlege","accounts_in_message":[],"_revision_number":1},{"id":"63a77514910c8d76fd33137ba273bc625fef1eae","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-24 03:01:38.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for reviewing. I didn\u0027t touch the py27 version of sphinx.","accounts_in_message":[],"_revision_number":1},{"id":"d2d1e997501dbc4d50d72840e2499abc3b69bfae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 03:05:25.000000000","message":"Patch Set 1: Code-Review-1\n\nTrue, I missed that, I would say that gr us also useful for packagers, further, some projects (thinking mainly of swift) may want to support older versions of python.","accounts_in_message":[],"_revision_number":1},{"id":"58afd073a2a910a6fa090ec11e2c1ed30d154394","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-24 03:52:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nWe\u0027re not ready for this yet but it\u0027s on the to-do list for this cycle once we have the tooling changes in place.\n\nAre you hitting an issue that this fixes?  if so can you explain it and we can perhaps find an alternate work around or make an exception for this change","accounts_in_message":[],"_revision_number":1},{"id":"9d429f08e7ae226add993315ad9c539dece63968","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 04:35:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/49/661149/1/check/openstack-tox-docs/96ed9ec/html/ : SUCCESS in 2m 49s\n- openstack-tox-pypy http://logs.openstack.org/49/661149/1/check/openstack-tox-pypy/8a1efba/ : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/661149/1/check/openstack-tox-pep8/b035d69/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/49/661149/1/check/openstack-tox-py27/6c9b9e1/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/49/661149/1/check/openstack-tox-py36/1528488/ : SUCCESS in 3m 54s\n- openstack-tox-validate http://logs.openstack.org/49/661149/1/check/openstack-tox-validate/c9de41c/ : SUCCESS in 3m 14s\n- requirements-integration http://logs.openstack.org/49/661149/1/check/requirements-integration/9cb3cb6/ : SUCCESS in 48m 18s\n- neutron-grenade-multinode http://logs.openstack.org/49/661149/1/check/neutron-grenade-multinode/540e9aa/ : SUCCESS in 1h 05m 50s\n- tempest-full http://logs.openstack.org/49/661149/1/check/tempest-full/2155c6c/ : SUCCESS in 1h 48m 22s\n- neutron-grenade http://logs.openstack.org/49/661149/1/check/neutron-grenade/efb9506/ : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":1},{"id":"70d11d7dd65cf34565dbbf9398a88d735370d27d","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-25 01:40:20.000000000","message":"Patch Set 1:\n\nThanks Matthew and Tony for the reviews. I don\u0027t have any issue with this. It\u0027s just that when trying to fix sphinx issue in Searchlight [1] by selecting python version, I think It\u0027d better now use py3.6 and above so we don\u0027t have to change it later.\n\n[1] https://review.opendev.org/#/c/659501/","accounts_in_message":[],"_revision_number":1},{"id":"002d64c68c284730d0b9ca2eb1fa651451eb7386","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-28 01:29:12.000000000","message":"Patch Set 1:\n\n\u003e Thanks Matthew and Tony for the reviews. I don\u0027t have any issue\n \u003e with this. It\u0027s just that when trying to fix sphinx issue in\n \u003e Searchlight [1] by selecting python version, I think It\u0027d better\n \u003e now use py3.6 and above so we don\u0027t have to change it later.\n \u003e \n \u003e [1] https://review.opendev.org/#/c/659501/\n\nOh :(\n\nSo cards on the table, what you want to do is the right thing.  However we (the requirements team) are\u0027t ready to make the changes needed to remove py\u003c3.6.  So we\u0027re kinda at an impasse .... 659501 can\u0027t merge yet and we can\u0027t merge this yet :(\n\nWith all the constraints work on our plate ATM I doubt we\u0027ll get to the work you need anytime soon, I\u0027m guessing it\u0027ll be around m2\n\nAre you willing to update 659501 to say \u003c3.4, and then add a patch on top that says \u003c3.6 (yes that\u0027s fixing it later but with the review there it wont be forgotten)","accounts_in_message":[],"_revision_number":1},{"id":"1d08b2c4ac00c12dc1fe702876e2fd3345f70463","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-28 01:38:57.000000000","message":"Patch Set 1:\n\nThanks Tony for the explanation. I will do the update on 659501.","accounts_in_message":[],"_revision_number":1},{"id":"5b17d1b0937af1c7985826d76aeb5d3000ede669","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-28 02:24:45.000000000","message":"Patch Set 1:\n\n\u003e Thanks Tony for the explanation. I will do the update on 659501.\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"7b3f01a1557fa588837cbf16b12dc7ba6d435e02","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-06-05 05:44:04.000000000","message":"Abandoned\n\n659501 was updated. Thanks.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f4eecc44a4a21fd0214a8391789a305dcb48f70","revisions":{"8f4eecc44a4a21fd0214a8391789a305dcb48f70":{"kind":"REWORK","_number":1,"created":"2019-05-24 02:37:56.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/49/661149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/661149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/661149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/661149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/661149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/661149/1"}}},"commit":{"parents":[{"commit":"9dc024128409b38eddf15a38ba91899fa8f89fae","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dc024128409b38eddf15a38ba91899fa8f89fae"}]}],"author":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-05-24 02:36:34.000000000","tz":540},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-05-24 02:37:53.000000000","tz":540},"subject":"Ignore python 3.4 and 3.5 for sphinx","message":"Ignore python 3.4 and 3.5 for sphinx\n\nSince the minimum supported version of python in Train is 3.6, I\nthink we can saftly ignore 3.4 and 3.5.\n\nChange-Id: I3e91d25cc6f9071fcc7314cfd34152c5f99a4db6\nSigned-off-by: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f4eecc44a4a21fd0214a8391789a305dcb48f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f4eecc44a4a21fd0214a8391789a305dcb48f70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
