)]}'
{"id":"openstack%2Frequirements~661189","triplet_id":"openstack%2Frequirements~master~I1c70279e62e205b3efb51621febe4de3e1296e00","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I1c70279e62e205b3efb51621febe4de3e1296e00","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2019-05-24 06:20:01.000000000","updated":"2019-05-26 17:08:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9705515e3a964bbf6c17eef48f5d0490afcb825e","_number":661189,"virtual_id_number":661189,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2019-05-24 22:42:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 16:27:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-24 22:42:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bbbde4528720b115d243b41b09684a139859e98d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-24 06:20:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d506c0e8c3df58b691247e4220ab9672eac951ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 08:09:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/661189/1/check/openstack-tox-docs/f915f4c/html/ : SUCCESS in 2m 46s\n- openstack-tox-pypy http://logs.openstack.org/89/661189/1/check/openstack-tox-pypy/f897fe0/ : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/661189/1/check/openstack-tox-pep8/46e4516/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/89/661189/1/check/openstack-tox-py27/59ff2b3/ : SUCCESS in 2m 48s\n- openstack-tox-py36 http://logs.openstack.org/89/661189/1/check/openstack-tox-py36/77aeabd/ : SUCCESS in 3m 19s\n- openstack-tox-validate http://logs.openstack.org/89/661189/1/check/openstack-tox-validate/866c28e/ : SUCCESS in 3m 47s\n- requirements-tox-babel http://logs.openstack.org/89/661189/1/check/requirements-tox-babel/47fa4aa/ : SUCCESS in 2m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/89/661189/1/check/requirements-tox-py27-check-uc/7cd1587/ : FAILURE in 6m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/89/661189/1/check/requirements-tox-py36-check-uc/3a21a69/ : FAILURE in 7m 01s\n- cross-cinder-py27 http://logs.openstack.org/89/661189/1/check/cross-cinder-py27/e7fe59d/ : SUCCESS in 9m 10s\n- cross-cinder-py36 http://logs.openstack.org/89/661189/1/check/cross-cinder-py36/1dc56e2/ : SUCCESS in 9m 00s\n- cross-glance-py27 http://logs.openstack.org/89/661189/1/check/cross-glance-py27/13f8bb2/ : SUCCESS in 5m 24s\n- cross-glance-py36 http://logs.openstack.org/89/661189/1/check/cross-glance-py36/a80d657/ : SUCCESS in 4m 49s\n- cross-horizon-py27 http://logs.openstack.org/89/661189/1/check/cross-horizon-py27/481e3f2/ : SUCCESS in 11m 23s\n- cross-horizon-py36 http://logs.openstack.org/89/661189/1/check/cross-horizon-py36/b34df33/ : SUCCESS in 9m 28s\n- cross-horizon-npm http://logs.openstack.org/89/661189/1/check/cross-horizon-npm/240d718/ : SUCCESS in 5m 48s\n- cross-keystone-py27 http://logs.openstack.org/89/661189/1/check/cross-keystone-py27/d44f545/ : SUCCESS in 22m 57s\n- cross-keystone-py36 http://logs.openstack.org/89/661189/1/check/cross-keystone-py36/5d8170a/ : SUCCESS in 27m 49s\n- cross-neutron-py27 http://logs.openstack.org/89/661189/1/check/cross-neutron-py27/72d6445/ : SUCCESS in 24m 01s\n- cross-neutron-py36 http://logs.openstack.org/89/661189/1/check/cross-neutron-py36/7583506/ : SUCCESS in 23m 15s\n- cross-nova-py27 http://logs.openstack.org/89/661189/1/check/cross-nova-py27/a4a1d53/ : SUCCESS in 11m 02s\n- cross-nova-py36 http://logs.openstack.org/89/661189/1/check/cross-nova-py36/7a93c96/ : SUCCESS in 11m 26s\n- cross-nova-functional http://logs.openstack.org/89/661189/1/check/cross-nova-functional/0f8fda1/ : FAILURE in 20m 17s\n- cross-osc-tox-docs http://logs.openstack.org/89/661189/1/check/cross-osc-tox-docs/5fa7900/html/ : SUCCESS in 5m 19s\n- cross-swift-py27 http://logs.openstack.org/89/661189/1/check/cross-swift-py27/ba60526/ : SUCCESS in 8m 05s\n- cross-swift-py36 http://logs.openstack.org/89/661189/1/check/cross-swift-py36/c9357f0/ : SUCCESS in 8m 01s\n- requirements-integration http://logs.openstack.org/89/661189/1/check/requirements-integration/dad01be/ : SUCCESS in 53m 24s\n- neutron-grenade-multinode http://logs.openstack.org/89/661189/1/check/neutron-grenade-multinode/e7259d8/ : SUCCESS in 1h 04m 37s\n- tempest-full http://logs.openstack.org/89/661189/1/check/tempest-full/07c6597/ : FAILURE in 1h 37m 21s\n- neutron-grenade http://logs.openstack.org/89/661189/1/check/neutron-grenade/2de2421/ : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ce9aadbb4e875c520d6dd620e803cc0237b5b362","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 16:27:32.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":1},{"id":"8eeead8526113d37c3029320d6d862ff14bc1255","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 16:27:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b3a4fd0dd4a9667d9ab226e1b6e1da97d2ada57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:27:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/661189/2/check/openstack-tox-docs/9c05249/html/ : SUCCESS in 3m 47s\n- openstack-tox-pypy http://logs.openstack.org/89/661189/2/check/openstack-tox-pypy/0c0a42c/ : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/661189/2/check/openstack-tox-pep8/7390e9f/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/89/661189/2/check/openstack-tox-py27/9f5286f/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/89/661189/2/check/openstack-tox-py36/5225636/ : SUCCESS in 4m 30s\n- openstack-tox-validate http://logs.openstack.org/89/661189/2/check/openstack-tox-validate/0ed4a37/ : SUCCESS in 4m 35s\n- requirements-tox-babel http://logs.openstack.org/89/661189/2/check/requirements-tox-babel/9307477/ : SUCCESS in 2m 53s\n- requirements-tox-py27-check-uc http://logs.openstack.org/89/661189/2/check/requirements-tox-py27-check-uc/c44e2e5/ : FAILURE in 6m 37s\n- requirements-tox-py36-check-uc http://logs.openstack.org/89/661189/2/check/requirements-tox-py36-check-uc/9134857/ : FAILURE in 7m 01s\n- cross-cinder-py27 http://logs.openstack.org/89/661189/2/check/cross-cinder-py27/cd6781d/ : SUCCESS in 7m 59s\n- cross-cinder-py36 http://logs.openstack.org/89/661189/2/check/cross-cinder-py36/cf70dbe/ : SUCCESS in 7m 05s\n- cross-glance-py27 http://logs.openstack.org/89/661189/2/check/cross-glance-py27/2ed7a41/ : SUCCESS in 4m 26s\n- cross-glance-py36 http://logs.openstack.org/89/661189/2/check/cross-glance-py36/abe647f/ : SUCCESS in 4m 00s\n- cross-horizon-py27 http://logs.openstack.org/89/661189/2/check/cross-horizon-py27/752d36a/ : SUCCESS in 10m 39s\n- cross-horizon-py36 http://logs.openstack.org/89/661189/2/check/cross-horizon-py36/dd97cb5/ : SUCCESS in 8m 53s\n- cross-horizon-npm http://logs.openstack.org/89/661189/2/check/cross-horizon-npm/acf9f0d/ : SUCCESS in 6m 21s\n- cross-keystone-py27 http://logs.openstack.org/89/661189/2/check/cross-keystone-py27/c7b283f/ : SUCCESS in 27m 09s\n- cross-keystone-py36 http://logs.openstack.org/89/661189/2/check/cross-keystone-py36/84c4dcd/ : SUCCESS in 25m 35s\n- cross-neutron-py27 http://logs.openstack.org/89/661189/2/check/cross-neutron-py27/a607607/ : SUCCESS in 26m 00s\n- cross-neutron-py36 http://logs.openstack.org/89/661189/2/check/cross-neutron-py36/630f53d/ : SUCCESS in 26m 26s\n- cross-nova-py27 http://logs.openstack.org/89/661189/2/check/cross-nova-py27/843825e/ : SUCCESS in 13m 06s\n- cross-nova-py36 http://logs.openstack.org/89/661189/2/check/cross-nova-py36/7752d02/ : SUCCESS in 14m 54s\n- cross-nova-functional http://logs.openstack.org/89/661189/2/check/cross-nova-functional/ce9d161/ : SUCCESS in 23m 17s\n- cross-osc-tox-docs http://logs.openstack.org/89/661189/2/check/cross-osc-tox-docs/a4ba30d/html/ : SUCCESS in 6m 00s\n- cross-swift-py27 http://logs.openstack.org/89/661189/2/check/cross-swift-py27/44c771a/ : SUCCESS in 11m 47s\n- cross-swift-py36 http://logs.openstack.org/89/661189/2/check/cross-swift-py36/3edefaf/ : SUCCESS in 8m 52s\n- requirements-integration http://logs.openstack.org/89/661189/2/check/requirements-integration/df4aa28/ : SUCCESS in 51m 45s\n- neutron-grenade-multinode http://logs.openstack.org/89/661189/2/check/neutron-grenade-multinode/054ff84/ : SUCCESS in 1h 11m 43s\n- tempest-full http://logs.openstack.org/89/661189/2/check/tempest-full/4b065fe/ : FAILURE in 1h 35m 37s\n- neutron-grenade http://logs.openstack.org/89/661189/2/check/neutron-grenade/417699d/ : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":2},{"id":"859050ccbc983087af18c9f4eede84bab20f405f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 20:52:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ebc409182dc6e3bece6cac2bf287fbe5a2f24118","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 22:42:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/661189/3/check/openstack-tox-docs/e4a0e1f/html/ : SUCCESS in 3m 52s\n- openstack-tox-pypy http://logs.openstack.org/89/661189/3/check/openstack-tox-pypy/f877261/ : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/661189/3/check/openstack-tox-pep8/52abab0/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/89/661189/3/check/openstack-tox-py27/21fc4d3/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/89/661189/3/check/openstack-tox-py36/ce50b0d/ : SUCCESS in 4m 33s\n- openstack-tox-validate http://logs.openstack.org/89/661189/3/check/openstack-tox-validate/3211325/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/89/661189/3/check/requirements-tox-babel/e923555/ : SUCCESS in 2m 37s\n- requirements-tox-py27-check-uc http://logs.openstack.org/89/661189/3/check/requirements-tox-py27-check-uc/81b05a2/ : FAILURE in 6m 40s\n- requirements-tox-py36-check-uc http://logs.openstack.org/89/661189/3/check/requirements-tox-py36-check-uc/eed7b31/ : FAILURE in 6m 28s\n- cross-cinder-py27 http://logs.openstack.org/89/661189/3/check/cross-cinder-py27/6682bc4/ : SUCCESS in 8m 03s\n- cross-cinder-py36 http://logs.openstack.org/89/661189/3/check/cross-cinder-py36/1c25e18/ : SUCCESS in 11m 23s\n- cross-glance-py27 http://logs.openstack.org/89/661189/3/check/cross-glance-py27/87dbdc5/ : SUCCESS in 4m 14s\n- cross-glance-py36 http://logs.openstack.org/89/661189/3/check/cross-glance-py36/5abf2af/ : SUCCESS in 3m 52s\n- cross-horizon-py27 http://logs.openstack.org/89/661189/3/check/cross-horizon-py27/187afdd/ : SUCCESS in 10m 21s\n- cross-horizon-py36 http://logs.openstack.org/89/661189/3/check/cross-horizon-py36/8a71f8d/ : SUCCESS in 8m 35s\n- cross-horizon-npm http://logs.openstack.org/89/661189/3/check/cross-horizon-npm/26e8cad/ : SUCCESS in 6m 09s\n- cross-keystone-py27 http://logs.openstack.org/89/661189/3/check/cross-keystone-py27/7fe9eef/ : SUCCESS in 22m 36s\n- cross-keystone-py36 http://logs.openstack.org/89/661189/3/check/cross-keystone-py36/0244096/ : SUCCESS in 24m 40s\n- cross-neutron-py27 http://logs.openstack.org/89/661189/3/check/cross-neutron-py27/951d445/ : SUCCESS in 25m 47s\n- cross-neutron-py36 http://logs.openstack.org/89/661189/3/check/cross-neutron-py36/8734fbc/ : SUCCESS in 25m 37s\n- cross-nova-py27 http://logs.openstack.org/89/661189/3/check/cross-nova-py27/1401cff/ : SUCCESS in 12m 48s\n- cross-nova-py36 http://logs.openstack.org/89/661189/3/check/cross-nova-py36/beeddc5/ : SUCCESS in 14m 03s\n- cross-nova-functional http://logs.openstack.org/89/661189/3/check/cross-nova-functional/1839755/ : SUCCESS in 21m 36s\n- cross-osc-tox-docs http://logs.openstack.org/89/661189/3/check/cross-osc-tox-docs/bdf7f49/html/ : SUCCESS in 5m 03s\n- cross-swift-py27 http://logs.openstack.org/89/661189/3/check/cross-swift-py27/15619ea/ : SUCCESS in 7m 59s\n- cross-swift-py36 http://logs.openstack.org/89/661189/3/check/cross-swift-py36/0b3cd47/ : SUCCESS in 9m 51s\n- requirements-integration http://logs.openstack.org/89/661189/3/check/requirements-integration/2a560b3/ : SUCCESS in 56m 49s\n- neutron-grenade-multinode http://logs.openstack.org/89/661189/3/check/neutron-grenade-multinode/32e6690/ : SUCCESS in 1h 20m 26s\n- tempest-full http://logs.openstack.org/89/661189/3/check/tempest-full/e6ee5b2/ : FAILURE in 1h 45m 51s\n- neutron-grenade http://logs.openstack.org/89/661189/3/check/neutron-grenade/a4c15bf/ : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":3},{"id":"9705515e3a964bbf6c17eef48f5d0490afcb825e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-26 17:08:06.000000000","message":"Abandoned\n\nusing the newer one","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"034d93ec71d912758ca25725e7c92e8489494bfd","revisions":{"70d67cfb351dee0779fd30b29906496cc96a9903":{"kind":"REWORK","_number":1,"created":"2019-05-24 06:20:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/661189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/661189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/661189/1"}}},"commit":{"parents":[{"commit":"f7c18048e922b260e5e27ba6a9d90d5acf208612","subject":"Merge \"Add cinder extras infinisdk\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7c18048e922b260e5e27ba6a9d90d5acf208612"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-24 06:19:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-24 06:19:59.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1c70279e62e205b3efb51621febe4de3e1296e00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70d67cfb351dee0779fd30b29906496cc96a9903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70d67cfb351dee0779fd30b29906496cc96a9903"}]},"branch":"refs/heads/master"},"cd719d32f6b08601a46d8ee4f4dd87d5a3828ee8":{"kind":"REWORK","_number":2,"created":"2019-05-24 16:27:35.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/89/661189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/661189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/661189/2"}}},"commit":{"parents":[{"commit":"f7c18048e922b260e5e27ba6a9d90d5acf208612","subject":"Merge \"Add cinder extras infinisdk\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7c18048e922b260e5e27ba6a9d90d5acf208612"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-24 06:19:59.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-24 16:27:24.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1c70279e62e205b3efb51621febe4de3e1296e00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd719d32f6b08601a46d8ee4f4dd87d5a3828ee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd719d32f6b08601a46d8ee4f4dd87d5a3828ee8"}]},"branch":"refs/heads/master"},"034d93ec71d912758ca25725e7c92e8489494bfd":{"kind":"REWORK","_number":3,"created":"2019-05-24 20:52:46.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/89/661189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/661189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/661189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/661189/3"}}},"commit":{"parents":[{"commit":"f7c18048e922b260e5e27ba6a9d90d5acf208612","subject":"Merge \"Add cinder extras infinisdk\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7c18048e922b260e5e27ba6a9d90d5acf208612"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-24 06:19:59.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-24 20:52:31.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1c70279e62e205b3efb51621febe4de3e1296e00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/034d93ec71d912758ca25725e7c92e8489494bfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/034d93ec71d912758ca25725e7c92e8489494bfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
