)]}'
{"id":"openstack%2Frequirements~661539","triplet_id":"openstack%2Frequirements~master~Ida5b084be8b68c0eaa021f548199ad477398e008","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ida5b084be8b68c0eaa021f548199ad477398e008","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-05-27 06:16:34.000000000","updated":"2019-05-30 00:23:55.000000000","submitted":"2019-05-29 04:00:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661539-1559102422891-229039fc","meta_rev_id":"b4f10aa6c6837881eb0e7022f254f65a570193ee","_number":661539,"virtual_id_number":661539,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":30388,"name":"zi tao","email":"taoxue@awcloud.com","username":"taoxue@awcloud.com"},{"value":2,"date":"2019-05-29 04:00:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-28 22:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":30388,"name":"zi tao","email":"taoxue@awcloud.com","username":"taoxue@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-29 01:13:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-28 22:01:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-05-30 00:23:55.000000000","post_submit":true,"_account_id":30388,"name":"zi tao","email":"taoxue@awcloud.com","username":"taoxue@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},{"_account_id":30388,"name":"zi tao","email":"taoxue@awcloud.com","username":"taoxue@awcloud.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 22:01:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-29 01:13:52.000000000","updated_by":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"reviewer":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"state":"REVIEWER"},{"updated":"2019-05-29 04:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 00:23:55.000000000","updated_by":{"_account_id":30388,"name":"zi tao","email":"taoxue@awcloud.com","username":"taoxue@awcloud.com"},"reviewer":{"_account_id":30388,"name":"zi tao","email":"taoxue@awcloud.com","username":"taoxue@awcloud.com"},"state":"REVIEWER"}],"messages":[{"id":"4d65bb7ad1f712e4a9a42f0719f74db10766a6cf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-27 06:16:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2a392f7505a9bfb4da4bc3cb70b2070689c3d4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 08:17:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/39/661539/1/check/openstack-tox-docs/fcf6ff4/html/ : SUCCESS in 2m 37s\n- openstack-tox-pypy http://logs.openstack.org/39/661539/1/check/openstack-tox-pypy/68d7b98/ : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/39/661539/1/check/openstack-tox-pep8/3ad0416/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/39/661539/1/check/openstack-tox-py27/f77bc55/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/39/661539/1/check/openstack-tox-py36/f92c453/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/39/661539/1/check/openstack-tox-validate/2d62ef3/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/39/661539/1/check/requirements-tox-babel/86bdb0b/ : SUCCESS in 2m 37s\n- requirements-tox-py27-check-uc http://logs.openstack.org/39/661539/1/check/requirements-tox-py27-check-uc/ac07e50/ : FAILURE in 6m 43s\n- requirements-tox-py36-check-uc http://logs.openstack.org/39/661539/1/check/requirements-tox-py36-check-uc/406078c/ : FAILURE in 7m 38s\n- cross-cinder-py27 http://logs.openstack.org/39/661539/1/check/cross-cinder-py27/1bacc32/ : SUCCESS in 8m 30s\n- cross-cinder-py36 http://logs.openstack.org/39/661539/1/check/cross-cinder-py36/95f6ac5/ : SUCCESS in 8m 39s\n- cross-glance-py27 http://logs.openstack.org/39/661539/1/check/cross-glance-py27/318301f/ : SUCCESS in 4m 31s\n- cross-glance-py36 http://logs.openstack.org/39/661539/1/check/cross-glance-py36/dea1e82/ : SUCCESS in 4m 49s\n- cross-horizon-py27 http://logs.openstack.org/39/661539/1/check/cross-horizon-py27/63d0689/ : SUCCESS in 9m 07s\n- cross-horizon-py36 http://logs.openstack.org/39/661539/1/check/cross-horizon-py36/ed213c2/ : SUCCESS in 7m 49s\n- cross-horizon-npm http://logs.openstack.org/39/661539/1/check/cross-horizon-npm/51bdc9f/ : SUCCESS in 6m 13s\n- cross-keystone-py27 http://logs.openstack.org/39/661539/1/check/cross-keystone-py27/32a0b9d/ : SUCCESS in 23m 23s\n- cross-keystone-py36 http://logs.openstack.org/39/661539/1/check/cross-keystone-py36/7fd5eba/ : SUCCESS in 27m 37s\n- cross-neutron-py27 http://logs.openstack.org/39/661539/1/check/cross-neutron-py27/02c0b6a/ : SUCCESS in 20m 57s\n- cross-neutron-py36 http://logs.openstack.org/39/661539/1/check/cross-neutron-py36/48964d5/ : SUCCESS in 23m 58s\n- cross-nova-py27 http://logs.openstack.org/39/661539/1/check/cross-nova-py27/9d6f2a0/ : SUCCESS in 12m 13s\n- cross-nova-py36 http://logs.openstack.org/39/661539/1/check/cross-nova-py36/c6d97f0/ : SUCCESS in 12m 21s\n- cross-nova-functional http://logs.openstack.org/39/661539/1/check/cross-nova-functional/4970c64/ : FAILURE in 18m 57s\n- cross-osc-tox-docs http://logs.openstack.org/39/661539/1/check/cross-osc-tox-docs/60722d6/html/ : SUCCESS in 5m 39s\n- cross-swift-py27 http://logs.openstack.org/39/661539/1/check/cross-swift-py27/4dc0f03/ : SUCCESS in 8m 13s\n- cross-swift-py36 http://logs.openstack.org/39/661539/1/check/cross-swift-py36/87cec12/ : SUCCESS in 8m 16s\n- requirements-integration http://logs.openstack.org/39/661539/1/check/requirements-integration/bb7ae2a/ : SUCCESS in 56m 55s\n- neutron-grenade-multinode http://logs.openstack.org/39/661539/1/check/neutron-grenade-multinode/05dcb96/ : SUCCESS in 1h 10m 28s\n- tempest-full http://logs.openstack.org/39/661539/1/check/tempest-full/b127828/ : FAILURE in 1h 36m 31s\n- neutron-grenade http://logs.openstack.org/39/661539/1/check/neutron-grenade/e447b84/ : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"668324ebb5f881eaf9e799f2be064b5ffaf99221","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-28 06:28:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a0e75f5d769616098d9da0e02b424734ded6d7bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:59:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/39/661539/2/check/openstack-tox-docs/3d644a3/html/ : SUCCESS in 2m 54s\n- openstack-tox-pypy http://logs.openstack.org/39/661539/2/check/openstack-tox-pypy/b9a1d32/ : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/39/661539/2/check/openstack-tox-pep8/abecb07/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/39/661539/2/check/openstack-tox-py27/bd0140d/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/39/661539/2/check/openstack-tox-py36/bba2854/ : SUCCESS in 3m 00s\n- openstack-tox-validate http://logs.openstack.org/39/661539/2/check/openstack-tox-validate/aaaf27d/ : SUCCESS in 2m 52s\n- requirements-tox-babel http://logs.openstack.org/39/661539/2/check/requirements-tox-babel/c138f25/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/39/661539/2/check/requirements-tox-py27-check-uc/9e3ab98/ : FAILURE in 11m 05s\n- requirements-tox-py36-check-uc http://logs.openstack.org/39/661539/2/check/requirements-tox-py36-check-uc/f93cb92/ : FAILURE in 11m 59s\n- cross-cinder-py27 http://logs.openstack.org/39/661539/2/check/cross-cinder-py27/ec405f6/ : SUCCESS in 8m 32s\n- cross-cinder-py36 http://logs.openstack.org/39/661539/2/check/cross-cinder-py36/2376ca0/ : SUCCESS in 7m 40s\n- cross-glance-py27 http://logs.openstack.org/39/661539/2/check/cross-glance-py27/7ceee29/ : SUCCESS in 4m 02s\n- cross-glance-py36 http://logs.openstack.org/39/661539/2/check/cross-glance-py36/3d654aa/ : SUCCESS in 3m 58s\n- cross-horizon-py27 http://logs.openstack.org/39/661539/2/check/cross-horizon-py27/60e8ffb/ : SUCCESS in 9m 57s\n- cross-horizon-py36 http://logs.openstack.org/39/661539/2/check/cross-horizon-py36/a1b1fdd/ : SUCCESS in 9m 05s\n- cross-horizon-npm http://logs.openstack.org/39/661539/2/check/cross-horizon-npm/12cd776/ : SUCCESS in 5m 35s\n- cross-keystone-py27 http://logs.openstack.org/39/661539/2/check/cross-keystone-py27/4b7b881/ : SUCCESS in 19m 53s\n- cross-keystone-py36 http://logs.openstack.org/39/661539/2/check/cross-keystone-py36/7a9c6a9/ : SUCCESS in 23m 10s\n- cross-neutron-py27 http://logs.openstack.org/39/661539/2/check/cross-neutron-py27/8bae7c0/ : SUCCESS in 23m 27s\n- cross-neutron-py36 http://logs.openstack.org/39/661539/2/check/cross-neutron-py36/565b02d/ : SUCCESS in 22m 59s\n- cross-nova-py27 http://logs.openstack.org/39/661539/2/check/cross-nova-py27/46b3155/ : SUCCESS in 11m 24s\n- cross-nova-py36 http://logs.openstack.org/39/661539/2/check/cross-nova-py36/8f35b70/ : SUCCESS in 11m 01s\n- cross-nova-functional http://logs.openstack.org/39/661539/2/check/cross-nova-functional/2510020/ : FAILURE in 18m 18s\n- cross-osc-tox-docs http://logs.openstack.org/39/661539/2/check/cross-osc-tox-docs/6d4e6c8/html/ : SUCCESS in 4m 58s\n- cross-swift-py27 http://logs.openstack.org/39/661539/2/check/cross-swift-py27/e604855/ : SUCCESS in 7m 43s\n- cross-swift-py36 http://logs.openstack.org/39/661539/2/check/cross-swift-py36/d62007d/ : SUCCESS in 10m 07s\n- requirements-integration http://logs.openstack.org/39/661539/2/check/requirements-integration/048bb45/ : SUCCESS in 1h 04m 23s\n- neutron-grenade-multinode http://logs.openstack.org/39/661539/2/check/neutron-grenade-multinode/a1c20e7/ : SUCCESS in 1h 03m 53s\n- tempest-full http://logs.openstack.org/39/661539/2/check/tempest-full/cbce9e2/ : FAILURE in 1h 24m 42s\n- neutron-grenade http://logs.openstack.org/39/661539/2/check/neutron-grenade/8001f1c/ : SUCCESS in 1h 03m 46s","accounts_in_message":[],"_revision_number":2},{"id":"22ba314198498e7c7790c954bc27c7ce6d1118a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-28 14:49:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5050657a288d1fb72103639d0c15dece366bb050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:10:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/39/661539/3/check/openstack-tox-docs/6aa1d06/html/ : SUCCESS in 3m 34s\n- openstack-tox-pypy http://logs.openstack.org/39/661539/3/check/openstack-tox-pypy/10eefb9/ : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/39/661539/3/check/openstack-tox-pep8/56bcf84/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/39/661539/3/check/openstack-tox-py27/7a844b3/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/39/661539/3/check/openstack-tox-py36/89774ef/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/39/661539/3/check/openstack-tox-validate/c9e8a6d/ : SUCCESS in 2m 37s\n- requirements-tox-babel http://logs.openstack.org/39/661539/3/check/requirements-tox-babel/aedd8a5/ : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/39/661539/3/check/requirements-tox-py27-check-uc/f551557/ : FAILURE in 14m 21s\n- requirements-tox-py36-check-uc http://logs.openstack.org/39/661539/3/check/requirements-tox-py36-check-uc/26dbf4c/ : FAILURE in 11m 18s\n- cross-cinder-py27 http://logs.openstack.org/39/661539/3/check/cross-cinder-py27/a66f446/ : SUCCESS in 9m 20s\n- cross-cinder-py36 http://logs.openstack.org/39/661539/3/check/cross-cinder-py36/4f664e1/ : SUCCESS in 7m 23s\n- cross-glance-py27 http://logs.openstack.org/39/661539/3/check/cross-glance-py27/582a2db/ : SUCCESS in 3m 48s\n- cross-glance-py36 http://logs.openstack.org/39/661539/3/check/cross-glance-py36/f4ee501/ : SUCCESS in 4m 46s\n- cross-horizon-py27 http://logs.openstack.org/39/661539/3/check/cross-horizon-py27/f9a2810/ : SUCCESS in 12m 00s\n- cross-horizon-py36 http://logs.openstack.org/39/661539/3/check/cross-horizon-py36/556aeaa/ : SUCCESS in 9m 43s\n- cross-horizon-npm http://logs.openstack.org/39/661539/3/check/cross-horizon-npm/daa25ed/ : SUCCESS in 6m 14s\n- cross-keystone-py27 http://logs.openstack.org/39/661539/3/check/cross-keystone-py27/7f15c40/ : SUCCESS in 19m 39s\n- cross-keystone-py36 http://logs.openstack.org/39/661539/3/check/cross-keystone-py36/0c9a6fe/ : SUCCESS in 28m 29s\n- cross-neutron-py27 http://logs.openstack.org/39/661539/3/check/cross-neutron-py27/20fb122/ : SUCCESS in 28m 35s\n- cross-neutron-py36 http://logs.openstack.org/39/661539/3/check/cross-neutron-py36/f31492e/ : SUCCESS in 28m 15s\n- cross-nova-py27 http://logs.openstack.org/39/661539/3/check/cross-nova-py27/a31cdbb/ : SUCCESS in 18m 11s\n- cross-nova-py36 http://logs.openstack.org/39/661539/3/check/cross-nova-py36/6275ae0/ : SUCCESS in 11m 49s\n- cross-nova-functional http://logs.openstack.org/39/661539/3/check/cross-nova-functional/484f72e/ : SUCCESS in 20m 57s\n- cross-osc-tox-docs http://logs.openstack.org/39/661539/3/check/cross-osc-tox-docs/895c5aa/html/ : SUCCESS in 5m 20s\n- cross-swift-py27 http://logs.openstack.org/39/661539/3/check/cross-swift-py27/1a342ea/ : SUCCESS in 7m 59s\n- cross-swift-py36 http://logs.openstack.org/39/661539/3/check/cross-swift-py36/ac0292a/ : SUCCESS in 8m 26s\n- requirements-integration http://logs.openstack.org/39/661539/3/check/requirements-integration/fc789d9/ : SUCCESS in 1h 18m 23s\n- neutron-grenade-multinode http://logs.openstack.org/39/661539/3/check/neutron-grenade-multinode/8ee3e26/ : SUCCESS in 1h 07m 41s\n- tempest-full http://logs.openstack.org/39/661539/3/check/tempest-full/64d042f/ : FAILURE in 1h 41m 41s\n- neutron-grenade http://logs.openstack.org/39/661539/3/check/neutron-grenade/532e899/ : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":3},{"id":"5fb8e2646b1b8250e3bfc75e5c7463bd6b873ecf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-28 17:11:51.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"d5bf1d71c6698671e21d6f0bf80c782099ba96b6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-28 17:19:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"30ea36d02374b066b3bf17d62638440c254c2140","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 19:30:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/39/661539/4/check/openstack-tox-docs/19074ee/html/ : SUCCESS in 3m 10s\n- openstack-tox-pypy http://logs.openstack.org/39/661539/4/check/openstack-tox-pypy/257604a/ : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/39/661539/4/check/openstack-tox-pep8/6d7ff25/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/39/661539/4/check/openstack-tox-py27/5b86a82/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/39/661539/4/check/openstack-tox-py36/a62136d/ : SUCCESS in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/39/661539/4/check/openstack-tox-validate/01a2218/ : SUCCESS in 3m 00s\n- requirements-tox-babel http://logs.openstack.org/39/661539/4/check/requirements-tox-babel/338ce00/ : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/39/661539/4/check/requirements-tox-py27-check-uc/62de5e6/ : SUCCESS in 11m 22s\n- requirements-tox-py36-check-uc http://logs.openstack.org/39/661539/4/check/requirements-tox-py36-check-uc/fc11560/ : SUCCESS in 12m 28s\n- cross-cinder-py27 http://logs.openstack.org/39/661539/4/check/cross-cinder-py27/9086353/ : SUCCESS in 8m 48s\n- cross-cinder-py36 http://logs.openstack.org/39/661539/4/check/cross-cinder-py36/1e14956/ : SUCCESS in 8m 12s\n- cross-glance-py27 http://logs.openstack.org/39/661539/4/check/cross-glance-py27/f0290b0/ : SUCCESS in 4m 20s\n- cross-glance-py36 http://logs.openstack.org/39/661539/4/check/cross-glance-py36/e65a78b/ : SUCCESS in 4m 07s\n- cross-horizon-py27 http://logs.openstack.org/39/661539/4/check/cross-horizon-py27/0b4b8f3/ : SUCCESS in 10m 26s\n- cross-horizon-py36 http://logs.openstack.org/39/661539/4/check/cross-horizon-py36/332e466/ : SUCCESS in 9m 33s\n- cross-horizon-npm http://logs.openstack.org/39/661539/4/check/cross-horizon-npm/21f0efe/ : SUCCESS in 6m 05s\n- cross-keystone-py27 http://logs.openstack.org/39/661539/4/check/cross-keystone-py27/8f62218/ : SUCCESS in 21m 41s\n- cross-keystone-py36 http://logs.openstack.org/39/661539/4/check/cross-keystone-py36/8d9d823/ : SUCCESS in 22m 30s\n- cross-neutron-py27 http://logs.openstack.org/39/661539/4/check/cross-neutron-py27/68db23e/ : SUCCESS in 23m 29s\n- cross-neutron-py36 http://logs.openstack.org/39/661539/4/check/cross-neutron-py36/ace334b/ : SUCCESS in 22m 31s\n- cross-nova-py27 http://logs.openstack.org/39/661539/4/check/cross-nova-py27/1361f45/ : SUCCESS in 13m 05s\n- cross-nova-py36 http://logs.openstack.org/39/661539/4/check/cross-nova-py36/c59a808/ : SUCCESS in 11m 30s\n- cross-nova-functional http://logs.openstack.org/39/661539/4/check/cross-nova-functional/685e67d/ : SUCCESS in 19m 05s\n- cross-osc-tox-docs http://logs.openstack.org/39/661539/4/check/cross-osc-tox-docs/4fcf16d/html/ : SUCCESS in 5m 04s\n- cross-swift-py27 http://logs.openstack.org/39/661539/4/check/cross-swift-py27/371a16f/ : SUCCESS in 7m 41s\n- cross-swift-py36 http://logs.openstack.org/39/661539/4/check/cross-swift-py36/e04a482/ : SUCCESS in 8m 40s\n- requirements-integration http://logs.openstack.org/39/661539/4/check/requirements-integration/afaf7a5/ : SUCCESS in 1h 03m 44s\n- neutron-grenade-multinode http://logs.openstack.org/39/661539/4/check/neutron-grenade-multinode/0fa925e/ : SUCCESS in 1h 18m 23s\n- tempest-full http://logs.openstack.org/39/661539/4/check/tempest-full/10e2dc9/ : FAILURE in 1h 30m 09s\n- neutron-grenade http://logs.openstack.org/39/661539/4/check/neutron-grenade/1cfbd9e/ : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":4},{"id":"1cc2d84dfb056fc773577d2f3407ed939494cdd8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-28 19:58:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"017782e1c35c4c366adcea690b2d731f6a92b121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 21:55:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/39/661539/5/check/openstack-tox-docs/4f2f837/html/ : SUCCESS in 2m 50s\n- openstack-tox-pypy http://logs.openstack.org/39/661539/5/check/openstack-tox-pypy/f839aab/ : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/39/661539/5/check/openstack-tox-pep8/08b97da/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/39/661539/5/check/openstack-tox-py27/fb4f1b7/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/39/661539/5/check/openstack-tox-py36/5d3b72a/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/39/661539/5/check/openstack-tox-validate/6f98c3e/ : SUCCESS in 5m 11s\n- requirements-tox-babel http://logs.openstack.org/39/661539/5/check/requirements-tox-babel/0abffee/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/39/661539/5/check/requirements-tox-py27-check-uc/9f6d862/ : SUCCESS in 10m 32s\n- requirements-tox-py36-check-uc http://logs.openstack.org/39/661539/5/check/requirements-tox-py36-check-uc/4770e01/ : SUCCESS in 14m 26s\n- cross-cinder-py27 http://logs.openstack.org/39/661539/5/check/cross-cinder-py27/db908c9/ : SUCCESS in 9m 12s\n- cross-cinder-py36 http://logs.openstack.org/39/661539/5/check/cross-cinder-py36/ac95e8d/ : SUCCESS in 7m 18s\n- cross-glance-py27 http://logs.openstack.org/39/661539/5/check/cross-glance-py27/6ef88e3/ : SUCCESS in 4m 40s\n- cross-glance-py36 http://logs.openstack.org/39/661539/5/check/cross-glance-py36/765dfbd/ : SUCCESS in 4m 23s\n- cross-horizon-py27 http://logs.openstack.org/39/661539/5/check/cross-horizon-py27/88d0960/ : SUCCESS in 10m 11s\n- cross-horizon-py36 http://logs.openstack.org/39/661539/5/check/cross-horizon-py36/dc08f90/ : SUCCESS in 10m 19s\n- cross-horizon-npm http://logs.openstack.org/39/661539/5/check/cross-horizon-npm/5d78378/ : SUCCESS in 6m 15s\n- cross-keystone-py27 http://logs.openstack.org/39/661539/5/check/cross-keystone-py27/48c1fe9/ : SUCCESS in 25m 14s\n- cross-keystone-py36 http://logs.openstack.org/39/661539/5/check/cross-keystone-py36/d81b8d7/ : SUCCESS in 22m 47s\n- cross-neutron-py27 http://logs.openstack.org/39/661539/5/check/cross-neutron-py27/b2ed775/ : SUCCESS in 22m 28s\n- cross-neutron-py36 http://logs.openstack.org/39/661539/5/check/cross-neutron-py36/d75d501/ : SUCCESS in 25m 08s\n- cross-nova-py27 http://logs.openstack.org/39/661539/5/check/cross-nova-py27/bf7534a/ : SUCCESS in 12m 24s\n- cross-nova-py36 http://logs.openstack.org/39/661539/5/check/cross-nova-py36/3741f0c/ : SUCCESS in 11m 38s\n- cross-nova-functional http://logs.openstack.org/39/661539/5/check/cross-nova-functional/4ac2e6d/ : SUCCESS in 16m 08s\n- cross-osc-tox-docs http://logs.openstack.org/39/661539/5/check/cross-osc-tox-docs/7204a42/html/ : SUCCESS in 5m 26s\n- cross-swift-py27 http://logs.openstack.org/39/661539/5/check/cross-swift-py27/c944f4f/ : SUCCESS in 8m 08s\n- cross-swift-py36 http://logs.openstack.org/39/661539/5/check/cross-swift-py36/e84ddca/ : SUCCESS in 8m 08s\n- requirements-integration http://logs.openstack.org/39/661539/5/check/requirements-integration/26df7db/ : SUCCESS in 58m 31s\n- neutron-grenade-multinode http://logs.openstack.org/39/661539/5/check/neutron-grenade-multinode/7fb7056/ : SUCCESS in 1h 12m 30s\n- tempest-full http://logs.openstack.org/39/661539/5/check/tempest-full/37ddf1f/ : SUCCESS in 1h 50m 44s\n- neutron-grenade http://logs.openstack.org/39/661539/5/check/neutron-grenade/57531a8/ : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":5},{"id":"818d10c6b0f8399ab033a8d50ad2661b77ea84f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-28 22:01:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"68f57da270cc0b6ca136c3fedc578105d328e2e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 22:38:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ad8505be06515ddcdd119a7c7f5b83ff787d85de","author":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"date":"2019-05-29 01:13:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a9cb6b2a8be8c426a9ba88801ed6ae52bf131d11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 04:00:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/39/661539/5/gate/openstack-tox-docs/9cc1aca/html/ : SUCCESS in 3m 03s\n- openstack-tox-pep8 http://logs.openstack.org/39/661539/5/gate/openstack-tox-pep8/262f7d5/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/39/661539/5/gate/openstack-tox-py27/5be854f/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/39/661539/5/gate/openstack-tox-py36/6cedcb2/ : SUCCESS in 4m 07s\n- openstack-tox-validate http://logs.openstack.org/39/661539/5/gate/openstack-tox-validate/abd0cae/ : SUCCESS in 3m 23s\n- requirements-tox-babel http://logs.openstack.org/39/661539/5/gate/requirements-tox-babel/8b8a03b/ : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc http://logs.openstack.org/39/661539/5/gate/requirements-tox-py27-check-uc/117a3c5/ : SUCCESS in 12m 33s\n- requirements-tox-py36-check-uc http://logs.openstack.org/39/661539/5/gate/requirements-tox-py36-check-uc/23488d1/ : SUCCESS in 13m 15s\n- cross-cinder-py27 http://logs.openstack.org/39/661539/5/gate/cross-cinder-py27/f6afbd1/ : SUCCESS in 8m 24s\n- cross-cinder-py36 http://logs.openstack.org/39/661539/5/gate/cross-cinder-py36/6499afd/ : SUCCESS in 12m 10s\n- cross-glance-py27 http://logs.openstack.org/39/661539/5/gate/cross-glance-py27/1c744bc/ : SUCCESS in 4m 34s\n- cross-glance-py36 http://logs.openstack.org/39/661539/5/gate/cross-glance-py36/438f5bd/ : SUCCESS in 5m 02s\n- cross-horizon-py27 http://logs.openstack.org/39/661539/5/gate/cross-horizon-py27/386b2cb/ : SUCCESS in 11m 09s\n- cross-horizon-py36 http://logs.openstack.org/39/661539/5/gate/cross-horizon-py36/e53da16/ : SUCCESS in 9m 58s\n- cross-horizon-npm http://logs.openstack.org/39/661539/5/gate/cross-horizon-npm/b8e54e5/ : SUCCESS in 6m 10s\n- cross-keystone-py27 http://logs.openstack.org/39/661539/5/gate/cross-keystone-py27/facd982/ : SUCCESS in 28m 48s\n- cross-keystone-py36 http://logs.openstack.org/39/661539/5/gate/cross-keystone-py36/3e09d71/ : SUCCESS in 24m 35s\n- cross-neutron-py27 http://logs.openstack.org/39/661539/5/gate/cross-neutron-py27/7eeb8bd/ : SUCCESS in 26m 13s\n- cross-neutron-py36 http://logs.openstack.org/39/661539/5/gate/cross-neutron-py36/ac76bc0/ : SUCCESS in 20m 51s\n- cross-nova-py27 http://logs.openstack.org/39/661539/5/gate/cross-nova-py27/706be05/ : SUCCESS in 12m 53s\n- cross-nova-py36 http://logs.openstack.org/39/661539/5/gate/cross-nova-py36/cb5aabf/ : SUCCESS in 14m 22s\n- cross-nova-functional http://logs.openstack.org/39/661539/5/gate/cross-nova-functional/42104be/ : SUCCESS in 18m 53s\n- cross-osc-tox-docs http://logs.openstack.org/39/661539/5/gate/cross-osc-tox-docs/a6b90bb/html/ : SUCCESS in 5m 43s\n- cross-swift-py27 http://logs.openstack.org/39/661539/5/gate/cross-swift-py27/aed05bd/ : SUCCESS in 8m 02s\n- cross-swift-py36 http://logs.openstack.org/39/661539/5/gate/cross-swift-py36/e77e6f1/ : SUCCESS in 8m 13s\n- requirements-integration http://logs.openstack.org/39/661539/5/gate/requirements-integration/0438bbc/ : SUCCESS in 1h 07m 38s\n- neutron-grenade-multinode http://logs.openstack.org/39/661539/5/gate/neutron-grenade-multinode/7fdc75e/ : SUCCESS in 1h 07m 51s\n- tempest-full http://logs.openstack.org/39/661539/5/gate/tempest-full/374885f/ : SUCCESS in 1h 47m 34s\n- neutron-grenade http://logs.openstack.org/39/661539/5/gate/neutron-grenade/b9bc822/ : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":5},{"id":"33383e0646ff56a36ccb11bd407ce420dcc4f4a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 04:00:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"16fd3c7bab7f6bb6add42ac5a60c1fd6aedaff29","revisions":{"5bd2592ee91517eec0133e6522a69ccfb7636096":{"kind":"REWORK","_number":1,"created":"2019-05-27 06:16:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/661539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/661539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/661539/1"}}},"commit":{"parents":[{"commit":"614885c53811f74a33d825cbb9d000b938ecdecf","subject":"Merge \"update constraint for osprofiler to new release 2.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/614885c53811f74a33d825cbb9d000b938ecdecf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-27 06:16:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-27 06:16:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ida5b084be8b68c0eaa021f548199ad477398e008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bd2592ee91517eec0133e6522a69ccfb7636096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bd2592ee91517eec0133e6522a69ccfb7636096"}]},"branch":"refs/heads/master"},"8ed102075139c5ebf91c69516e83e2f7b20728d1":{"kind":"REWORK","_number":2,"created":"2019-05-28 06:28:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/661539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/661539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/661539/2"}}},"commit":{"parents":[{"commit":"3fd78e5eb1dda2d33a73a89a6704b3f29c910847","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fd78e5eb1dda2d33a73a89a6704b3f29c910847"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-28 06:28:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-28 06:28:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ida5b084be8b68c0eaa021f548199ad477398e008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ed102075139c5ebf91c69516e83e2f7b20728d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ed102075139c5ebf91c69516e83e2f7b20728d1"}]},"branch":"refs/heads/master"},"932e6c05c2190bf5b40dd6272df0b0074ab287d6":{"kind":"REWORK","_number":3,"created":"2019-05-28 14:49:55.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/39/661539/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/661539/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/661539/3"}}},"commit":{"parents":[{"commit":"3fd78e5eb1dda2d33a73a89a6704b3f29c910847","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fd78e5eb1dda2d33a73a89a6704b3f29c910847"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-28 06:28:34.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-28 14:35:05.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\n-grpcio\u003d\u003d\u003d1.21.1\n+grpcio\u003d\u003d\u003d1.15.0\ntaskflow email, needs a release as well to unblock\n\n-pypowervm\u003d\u003d\u003d1.1.21\n+pypowervm\u003d\u003d\u003d1.1.20\nhttps://github.com/powervm/pypowervm/issues/12\n  - will mask the bad version\n\n-tornado\u003d\u003d\u003d5.1.1;python_version\u003d\u003d\u00272.7\u0027\n+tornado\u003d\u003d\u003d4.5.3;python_version\u003d\u003d\u00272.7\u0027\npy27 keeps pulling in too new versions, will need to either maintain a\nsecondary requirements list for for py27 caps in constraints only\nlibraries (those not tracked by global-requirements.txt).\n\n-kubernetes\u003d\u003d\u003d9.0.0\n+kubernetes\u003d\u003d\u003d8.0.1\nopenshift is capping things, they don\u0027t seem to want to change\nperhaps we should request that the only two projects using openshift\nto not use it anymore.  openstack/kuryr-tempest-plugin x/flame\n\n-jsonschema\u003d\u003d\u003d3.0.1\n+jsonschema\u003d\u003d\u003d2.6.0\nbunch of stuff in openstack still has a cap\n\n-fasteners\u003d\u003d\u003d0.15\n+fasteners\u003d\u003d\u003d0.14.1\nhttps://github.com/harlowja/fasteners/issues/36\n\nChange-Id: Ida5b084be8b68c0eaa021f548199ad477398e008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/932e6c05c2190bf5b40dd6272df0b0074ab287d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/932e6c05c2190bf5b40dd6272df0b0074ab287d6"}]},"branch":"refs/heads/master"},"e1769272d2b70b7323fe9af55afc70ab6b0ab608":{"kind":"REWORK","_number":4,"created":"2019-05-28 17:19:03.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/39/661539/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/661539/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/661539/4"}}},"commit":{"parents":[{"commit":"3fd78e5eb1dda2d33a73a89a6704b3f29c910847","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fd78e5eb1dda2d33a73a89a6704b3f29c910847"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-28 06:28:34.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-28 17:17:46.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\n-grpcio\u003d\u003d\u003d1.21.1\n+grpcio\u003d\u003d\u003d1.15.0\ntaskflow email, needs a release as well to unblock\n\n-pypowervm\u003d\u003d\u003d1.1.21\n+pypowervm\u003d\u003d\u003d1.1.20\nhttps://github.com/powervm/pypowervm/issues/12\n  - will mask the bad version\n\n-tornado\u003d\u003d\u003d5.1.1;python_version\u003d\u003d\u00272.7\u0027\n+tornado\u003d\u003d\u003d4.5.3;python_version\u003d\u003d\u00272.7\u0027\npy27 keeps pulling in too new versions, will need to either maintain a\nsecondary requirements list for for py27 caps in constraints only\nlibraries (those not tracked by global-requirements.txt).\n\n-urllib3\u003d\u003d\u003d1.25.3\n+urllib3\u003d\u003d\u003d1.24.3\n-kubernetes\u003d\u003d\u003d9.0.0\n+kubernetes\u003d\u003d\u003d8.0.1\nopenshift is capping things, they don\u0027t seem to want to change\nperhaps we should request that the only two projects using openshift\nto not use it anymore.  openstack/kuryr-tempest-plugin x/flame\n\n-jsonschema\u003d\u003d\u003d3.0.1\n+jsonschema\u003d\u003d\u003d2.6.0\nbunch of stuff in openstack still has a cap\n\n-fasteners\u003d\u003d\u003d0.15\n+fasteners\u003d\u003d\u003d0.14.1\nhttps://github.com/harlowja/fasteners/issues/36\n\nChange-Id: Ida5b084be8b68c0eaa021f548199ad477398e008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1769272d2b70b7323fe9af55afc70ab6b0ab608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1769272d2b70b7323fe9af55afc70ab6b0ab608"}]},"branch":"refs/heads/master"},"16fd3c7bab7f6bb6add42ac5a60c1fd6aedaff29":{"kind":"REWORK","_number":5,"created":"2019-05-28 19:58:47.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/39/661539/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/661539/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/661539/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/661539/5"}}},"commit":{"parents":[{"commit":"3fd78e5eb1dda2d33a73a89a6704b3f29c910847","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fd78e5eb1dda2d33a73a89a6704b3f29c910847"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-28 06:28:34.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-28 19:58:27.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\n-grpcio\u003d\u003d\u003d1.21.1\n+grpcio\u003d\u003d\u003d1.15.0\ntaskflow email, needs a release as well to unblock\n\n-pypowervm\u003d\u003d\u003d1.1.21\n+pypowervm\u003d\u003d\u003d1.1.20\nhttps://github.com/powervm/pypowervm/issues/12\n  - will mask the bad version\n\n-tornado\u003d\u003d\u003d5.1.1;python_version\u003d\u003d\u00272.7\u0027\n+tornado\u003d\u003d\u003d4.5.3;python_version\u003d\u003d\u00272.7\u0027\npy27 keeps pulling in too new versions, will need to either maintain a\nsecondary requirements list for for py27 caps in constraints only\nlibraries (those not tracked by global-requirements.txt).\n\n-urllib3\u003d\u003d\u003d1.25.3\n+urllib3\u003d\u003d\u003d1.24.3\n-kubernetes\u003d\u003d\u003d9.0.0\n+kubernetes\u003d\u003d\u003d8.0.1\nopenshift is capping things, they don\u0027t seem to want to change\nperhaps we should request that the only two projects using openshift\nto not use it anymore.  openstack/kuryr-tempest-plugin x/flame\n\n-jsonschema\u003d\u003d\u003d3.0.1\n+jsonschema\u003d\u003d\u003d2.6.0\nbunch of stuff in openstack still has a cap\n\n-fasteners\u003d\u003d\u003d0.15\n+fasteners\u003d\u003d\u003d0.14.1\nhttps://github.com/harlowja/fasteners/issues/36\n\nChange-Id: Ida5b084be8b68c0eaa021f548199ad477398e008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16fd3c7bab7f6bb6add42ac5a60c1fd6aedaff29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16fd3c7bab7f6bb6add42ac5a60c1fd6aedaff29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
