)]}'
{"id":"openstack%2Frequirements~662008","triplet_id":"openstack%2Frequirements~master~I4216714cad8df18a79f7cd0b8b25d6d4b29a9c3c","project":"openstack/requirements","branch":"master","topic":"disallow-fasteners-0.15","hashtags":[],"change_id":"I4216714cad8df18a79f7cd0b8b25d6d4b29a9c3c","subject":"disallow fasteners 0.15","status":"MERGED","created":"2019-05-29 14:47:41.000000000","updated":"2019-05-29 22:41:56.000000000","submitted":"2019-05-29 22:41:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662008-1559169716711-65a39237","meta_rev_id":"e5797d45ddd9ed093514e36f02e97bc56c5564d5","_number":662008,"virtual_id_number":662008,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-29 22:41:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-29 14:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-29 20:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-29 20:30:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 14:52:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-29 20:30:44.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-29 22:41:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af3f3f9429862fa10174552815a1082e2ee46dd3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-29 14:47:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4ef9dd24b07e55e0abcb17613db25a6c39695d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-29 14:52:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7cdedb08e49c0ba3de4b4bb4ea2fa75553108bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 19:46:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/662008/1/check/openstack-tox-docs/af12932/html/ : SUCCESS in 2m 47s\n- openstack-tox-pypy http://logs.openstack.org/08/662008/1/check/openstack-tox-pypy/fb9eb7d/ : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/08/662008/1/check/openstack-tox-pep8/1c86ba1/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/08/662008/1/check/openstack-tox-py27/080960a/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/08/662008/1/check/openstack-tox-py36/e6d3b2a/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/08/662008/1/check/openstack-tox-validate/8b0c811/ : SUCCESS in 3m 04s\n- requirements-integration http://logs.openstack.org/08/662008/1/check/requirements-integration/310e0af/ : SUCCESS in 45m 35s\n- neutron-grenade-multinode http://logs.openstack.org/08/662008/1/check/neutron-grenade-multinode/591b7e6/ : SUCCESS in 1h 08m 32s\n- tempest-full http://logs.openstack.org/08/662008/1/check/tempest-full/6b318db/ : SUCCESS in 1h 56m 54s\n- neutron-grenade http://logs.openstack.org/08/662008/1/check/neutron-grenade/d338777/ : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":1},{"id":"52c990645535b864554f0dd1de20f5fa62f4f067","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-29 20:30:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ac8e04432a0e6b35d113151196ef9728a4c7612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 20:31:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c43c707c9864315d1b28e780026ce6445968284c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 22:41:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/662008/1/gate/openstack-tox-docs/d589032/html/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/08/662008/1/gate/openstack-tox-pep8/6e3db64/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/08/662008/1/gate/openstack-tox-py27/33901c7/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/08/662008/1/gate/openstack-tox-py36/99a6834/ : SUCCESS in 3m 30s\n- openstack-tox-validate http://logs.openstack.org/08/662008/1/gate/openstack-tox-validate/e1fae2c/ : SUCCESS in 3m 10s\n- requirements-integration http://logs.openstack.org/08/662008/1/gate/requirements-integration/6b5146a/ : SUCCESS in 58m 01s\n- neutron-grenade-multinode http://logs.openstack.org/08/662008/1/gate/neutron-grenade-multinode/a769be4/ : SUCCESS in 1h 12m 46s\n- tempest-full http://logs.openstack.org/08/662008/1/gate/tempest-full/76d99b2/ : SUCCESS in 1h 52m 27s\n- neutron-grenade http://logs.openstack.org/08/662008/1/gate/neutron-grenade/140f697/ : SUCCESS in 1h 04m 46s","accounts_in_message":[],"_revision_number":1},{"id":"eec481a5e3f3233c8fc828922adcf09b2d99ea6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 22:41:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3495c06c2a52e2a99fc841f5af027408d0d24679","revisions":{"3495c06c2a52e2a99fc841f5af027408d0d24679":{"kind":"REWORK","_number":1,"created":"2019-05-29 14:47:41.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/08/662008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/662008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/662008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/662008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/662008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/662008/1"}}},"commit":{"parents":[{"commit":"c3b2aaf055e6e1e10327dd3b1654045a40ea1d3b","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3b2aaf055e6e1e10327dd3b1654045a40ea1d3b"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-29 14:41:56.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-29 14:41:56.000000000","tz":-300},"subject":"disallow fasteners 0.15","message":"disallow fasteners 0.15\n\nThere is apparently still a race somewhere with eventlet and we are\nwaiting on a revert of the \u0027fix\u0027 before we can upgrade fasteners.\n\nhttps://github.com/harlowja/fasteners/pull/37\n\nChange-Id: I4216714cad8df18a79f7cd0b8b25d6d4b29a9c3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3495c06c2a52e2a99fc841f5af027408d0d24679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3495c06c2a52e2a99fc841f5af027408d0d24679"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
