)]}'
{"id":"openstack%2Frequirements~662121","triplet_id":"openstack%2Frequirements~master~I75677aed784be0c096ae7be335f6ec8f69658246","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I75677aed784be0c096ae7be335f6ec8f69658246","subject":"update constraint for pbr to new release 5.2.1","status":"MERGED","created":"2019-05-30 02:12:24.000000000","updated":"2019-05-30 21:59:20.000000000","submitted":"2019-05-30 21:59:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662121-1559253560368-ac2212ed","meta_rev_id":"71079d5bb609edfdcf63c9c24d07971f1d6303eb","_number":662121,"virtual_id_number":662121,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-30 21:59:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-30 19:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-30 19:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-30 19:35:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-30 21:59:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d3666c0f14115e7e03841859540ac5acc881310","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-30 02:12:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7fb2ddfd6a556ab918a98ff0357a3f3bcbda608","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 04:58:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/21/662121/1/check/openstack-tox-docs/2b5223b/html/ : SUCCESS in 3m 17s\n- openstack-tox-pypy http://logs.openstack.org/21/662121/1/check/openstack-tox-pypy/e3c34dd/ : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/662121/1/check/openstack-tox-pep8/d341c24/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/21/662121/1/check/openstack-tox-py27/551b54a/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/21/662121/1/check/openstack-tox-py36/d8f0438/ : SUCCESS in 3m 39s\n- openstack-tox-validate http://logs.openstack.org/21/662121/1/check/openstack-tox-validate/6b9054c/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/21/662121/1/check/requirements-tox-babel/fec370f/ : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/21/662121/1/check/requirements-tox-py27-check-uc/3841883/ : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc http://logs.openstack.org/21/662121/1/check/requirements-tox-py36-check-uc/9e61c0c/ : SUCCESS in 6m 12s\n- cross-cinder-py27 http://logs.openstack.org/21/662121/1/check/cross-cinder-py27/b9164f1/ : SUCCESS in 8m 23s\n- cross-cinder-py36 http://logs.openstack.org/21/662121/1/check/cross-cinder-py36/a3c60d7/ : SUCCESS in 7m 38s\n- cross-glance-py27 http://logs.openstack.org/21/662121/1/check/cross-glance-py27/40b794b/ : SUCCESS in 4m 22s\n- cross-glance-py36 http://logs.openstack.org/21/662121/1/check/cross-glance-py36/a36c81f/ : SUCCESS in 5m 10s\n- cross-horizon-py27 http://logs.openstack.org/21/662121/1/check/cross-horizon-py27/cb621ec/ : SUCCESS in 10m 19s\n- cross-horizon-py36 http://logs.openstack.org/21/662121/1/check/cross-horizon-py36/5c665f1/ : RETRY_LIMIT in 3m 50s\n- cross-horizon-npm http://logs.openstack.org/21/662121/1/check/cross-horizon-npm/6d07524/ : SUCCESS in 5m 26s\n- cross-keystone-py27 http://logs.openstack.org/21/662121/1/check/cross-keystone-py27/d95e25e/ : RETRY_LIMIT in 2m 38s\n- cross-keystone-py36 http://logs.openstack.org/21/662121/1/check/cross-keystone-py36/1e3872a/ : SUCCESS in 23m 43s\n- cross-neutron-py27 http://logs.openstack.org/21/662121/1/check/cross-neutron-py27/c612e14/ : SUCCESS in 27m 38s\n- cross-neutron-py36 http://logs.openstack.org/21/662121/1/check/cross-neutron-py36/d8bd90d/ : SUCCESS in 19m 28s\n- cross-nova-py27 http://logs.openstack.org/21/662121/1/check/cross-nova-py27/d83d6d6/ : SUCCESS in 15m 05s\n- cross-nova-py36 http://logs.openstack.org/21/662121/1/check/cross-nova-py36/470a7ac/ : SUCCESS in 12m 01s\n- cross-nova-functional http://logs.openstack.org/21/662121/1/check/cross-nova-functional/e62e4b3/ : SUCCESS in 18m 27s\n- cross-osc-tox-docs http://logs.openstack.org/21/662121/1/check/cross-osc-tox-docs/4d2bf3a/html/ : SUCCESS in 5m 39s\n- cross-swift-py27 http://logs.openstack.org/21/662121/1/check/cross-swift-py27/73ec75f/ : SUCCESS in 8m 41s\n- cross-swift-py36 http://logs.openstack.org/21/662121/1/check/cross-swift-py36/d8cfa20/ : SUCCESS in 8m 24s\n- requirements-integration http://logs.openstack.org/21/662121/1/check/requirements-integration/1267c0d/ : SUCCESS in 50m 43s\n- neutron-grenade-multinode http://logs.openstack.org/21/662121/1/check/neutron-grenade-multinode/9083277/ : SUCCESS in 1h 08m 05s\n- tempest-full http://logs.openstack.org/21/662121/1/check/tempest-full/f62d942/ : SUCCESS in 1h 36m 51s\n- neutron-grenade http://logs.openstack.org/21/662121/1/check/neutron-grenade/406c61a/ : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ef0fbb9bb18df599f249c3deff5b40dde5703398","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-30 14:27:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a036b3f29cae9a6e05bb3c4faa64efa78395970","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 18:14:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/21/662121/1/check/openstack-tox-docs/ccf657b/html/ : SUCCESS in 2m 45s\n- openstack-tox-pypy http://logs.openstack.org/21/662121/1/check/openstack-tox-pypy/56273c6/ : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/662121/1/check/openstack-tox-pep8/6c2bdf9/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/21/662121/1/check/openstack-tox-py27/38005d7/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/21/662121/1/check/openstack-tox-py36/e975217/ : SUCCESS in 3m 35s\n- openstack-tox-validate http://logs.openstack.org/21/662121/1/check/openstack-tox-validate/7b62ed6/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/21/662121/1/check/requirements-tox-babel/34b607c/ : SUCCESS in 4m 37s\n- requirements-tox-py27-check-uc http://logs.openstack.org/21/662121/1/check/requirements-tox-py27-check-uc/2a41710/ : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc http://logs.openstack.org/21/662121/1/check/requirements-tox-py36-check-uc/1fc4632/ : SUCCESS in 6m 21s\n- cross-cinder-py27 http://logs.openstack.org/21/662121/1/check/cross-cinder-py27/e07c59f/ : SUCCESS in 7m 54s\n- cross-cinder-py36 http://logs.openstack.org/21/662121/1/check/cross-cinder-py36/e57b3d6/ : SUCCESS in 8m 08s\n- cross-glance-py27 http://logs.openstack.org/21/662121/1/check/cross-glance-py27/b935a7c/ : SUCCESS in 4m 21s\n- cross-glance-py36 http://logs.openstack.org/21/662121/1/check/cross-glance-py36/05e58d3/ : SUCCESS in 3m 58s\n- cross-horizon-py27 http://logs.openstack.org/21/662121/1/check/cross-horizon-py27/a791610/ : SUCCESS in 10m 21s\n- cross-horizon-py36 http://logs.openstack.org/21/662121/1/check/cross-horizon-py36/92e06a7/ : SUCCESS in 9m 53s\n- cross-horizon-npm http://logs.openstack.org/21/662121/1/check/cross-horizon-npm/fc268cf/ : SUCCESS in 6m 23s\n- cross-keystone-py27 http://logs.openstack.org/21/662121/1/check/cross-keystone-py27/4f27073/ : SUCCESS in 24m 11s\n- cross-keystone-py36 http://logs.openstack.org/21/662121/1/check/cross-keystone-py36/b1385e8/ : SUCCESS in 23m 59s\n- cross-neutron-py27 http://logs.openstack.org/21/662121/1/check/cross-neutron-py27/cdb1365/ : SUCCESS in 30m 13s\n- cross-neutron-py36 http://logs.openstack.org/21/662121/1/check/cross-neutron-py36/e3b644e/ : SUCCESS in 29m 12s\n- cross-nova-py27 http://logs.openstack.org/21/662121/1/check/cross-nova-py27/a730494/ : SUCCESS in 18m 40s\n- cross-nova-py36 http://logs.openstack.org/21/662121/1/check/cross-nova-py36/7686153/ : SUCCESS in 10m 50s\n- cross-nova-functional http://logs.openstack.org/21/662121/1/check/cross-nova-functional/88e32ef/ : SUCCESS in 17m 32s\n- cross-osc-tox-docs http://logs.openstack.org/21/662121/1/check/cross-osc-tox-docs/4f9cf98/html/ : SUCCESS in 5m 25s\n- cross-swift-py27 http://logs.openstack.org/21/662121/1/check/cross-swift-py27/ca0ae1c/ : SUCCESS in 7m 44s\n- cross-swift-py36 http://logs.openstack.org/21/662121/1/check/cross-swift-py36/9e36f52/ : SUCCESS in 8m 01s\n- requirements-integration http://logs.openstack.org/21/662121/1/check/requirements-integration/8431461/ : SUCCESS in 59m 56s\n- neutron-grenade-multinode http://logs.openstack.org/21/662121/1/check/neutron-grenade-multinode/44848c5/ : SUCCESS in 1h 22m 43s\n- tempest-full http://logs.openstack.org/21/662121/1/check/tempest-full/f0cebb9/ : SUCCESS in 1h 34m 50s\n- neutron-grenade http://logs.openstack.org/21/662121/1/check/neutron-grenade/701c29a/ : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9a58124761b7ce0e67db56f8f39d4adb3cfa52c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-30 19:35:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"665361a70bcc4d7a9d951b8e1542a4439e5c7049","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 19:36:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9aaf5f4e5b2029a724bfbcf3ee13cc6f89c43ac2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 21:59:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/21/662121/1/gate/openstack-tox-docs/cff5f2f/html/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/21/662121/1/gate/openstack-tox-pep8/caab6ab/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/21/662121/1/gate/openstack-tox-py27/ba121f8/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/21/662121/1/gate/openstack-tox-py36/89b381e/ : SUCCESS in 3m 43s\n- openstack-tox-validate http://logs.openstack.org/21/662121/1/gate/openstack-tox-validate/3d5ee7e/ : SUCCESS in 5m 00s\n- requirements-tox-babel http://logs.openstack.org/21/662121/1/gate/requirements-tox-babel/ba5e512/ : SUCCESS in 3m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/21/662121/1/gate/requirements-tox-py27-check-uc/e352561/ : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc http://logs.openstack.org/21/662121/1/gate/requirements-tox-py36-check-uc/59b337f/ : SUCCESS in 6m 41s\n- cross-cinder-py27 http://logs.openstack.org/21/662121/1/gate/cross-cinder-py27/34bf14a/ : SUCCESS in 8m 20s\n- cross-cinder-py36 http://logs.openstack.org/21/662121/1/gate/cross-cinder-py36/01df2bb/ : SUCCESS in 10m 47s\n- cross-glance-py27 http://logs.openstack.org/21/662121/1/gate/cross-glance-py27/95edcf1/ : SUCCESS in 4m 18s\n- cross-glance-py36 http://logs.openstack.org/21/662121/1/gate/cross-glance-py36/44422cb/ : SUCCESS in 4m 18s\n- cross-horizon-py27 http://logs.openstack.org/21/662121/1/gate/cross-horizon-py27/3a63dfc/ : SUCCESS in 10m 50s\n- cross-horizon-py36 http://logs.openstack.org/21/662121/1/gate/cross-horizon-py36/b87dba9/ : SUCCESS in 7m 32s\n- cross-horizon-npm http://logs.openstack.org/21/662121/1/gate/cross-horizon-npm/3e15ede/ : SUCCESS in 7m 36s\n- cross-keystone-py27 http://logs.openstack.org/21/662121/1/gate/cross-keystone-py27/5132a54/ : SUCCESS in 24m 01s\n- cross-keystone-py36 http://logs.openstack.org/21/662121/1/gate/cross-keystone-py36/632e7e3/ : SUCCESS in 22m 50s\n- cross-neutron-py27 http://logs.openstack.org/21/662121/1/gate/cross-neutron-py27/a4b2760/ : SUCCESS in 30m 24s\n- cross-neutron-py36 http://logs.openstack.org/21/662121/1/gate/cross-neutron-py36/51dab8b/ : SUCCESS in 22m 36s\n- cross-nova-py27 http://logs.openstack.org/21/662121/1/gate/cross-nova-py27/7fbf947/ : SUCCESS in 16m 44s\n- cross-nova-py36 http://logs.openstack.org/21/662121/1/gate/cross-nova-py36/b78a388/ : SUCCESS in 13m 02s\n- cross-nova-functional http://logs.openstack.org/21/662121/1/gate/cross-nova-functional/265db4c/ : SUCCESS in 17m 50s\n- cross-osc-tox-docs http://logs.openstack.org/21/662121/1/gate/cross-osc-tox-docs/4fe31fa/html/ : SUCCESS in 5m 23s\n- cross-swift-py27 http://logs.openstack.org/21/662121/1/gate/cross-swift-py27/f57c0dc/ : SUCCESS in 8m 08s\n- cross-swift-py36 http://logs.openstack.org/21/662121/1/gate/cross-swift-py36/0bbcc48/ : SUCCESS in 8m 56s\n- requirements-integration http://logs.openstack.org/21/662121/1/gate/requirements-integration/742155c/ : SUCCESS in 51m 07s\n- neutron-grenade-multinode http://logs.openstack.org/21/662121/1/gate/neutron-grenade-multinode/57bfb69/ : SUCCESS in 1h 06m 36s\n- tempest-full http://logs.openstack.org/21/662121/1/gate/tempest-full/c553349/ : SUCCESS in 1h 38m 30s\n- neutron-grenade http://logs.openstack.org/21/662121/1/gate/neutron-grenade/fd7e222/ : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":1},{"id":"590d37413a294944ca2b59a1c76a29224c1fa374","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 21:59:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd8620b55a1a7e2680b8693ebb9393b8036569b3","revisions":{"fd8620b55a1a7e2680b8693ebb9393b8036569b3":{"kind":"REWORK","_number":1,"created":"2019-05-30 02:12:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/662121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/662121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/662121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/662121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/662121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/662121/1"}}},"commit":{"parents":[{"commit":"d44eb063be2fb76d74f94655b6dd42d305ef13ae","subject":"Merge \"disallow fasteners 0.15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d44eb063be2fb76d74f94655b6dd42d305ef13ae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-30 02:12:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-30 02:12:22.000000000","tz":0},"subject":"update constraint for pbr to new release 5.2.1","message":"update constraint for pbr to new release 5.2.1\n\nChange-Id: I75677aed784be0c096ae7be335f6ec8f69658246\nmeta:version: 5.2.1\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Iba1df2119592ce5a468d56958019c08692b6b245\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd8620b55a1a7e2680b8693ebb9393b8036569b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd8620b55a1a7e2680b8693ebb9393b8036569b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
