)]}'
{"id":"openstack%2Frequirements~662420","triplet_id":"openstack%2Frequirements~master~Ia33f82cb6561fd8e0ffad7a51c28c3ac922d0c97","project":"openstack/requirements","branch":"master","topic":"add_support_for_responses","hashtags":[],"change_id":"Ia33f82cb6561fd8e0ffad7a51c28c3ac922d0c97","subject":"Add responses library for tacker test cases","status":"ABANDONED","created":"2019-05-31 09:37:15.000000000","updated":"2019-06-07 08:35:56.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"003aaf7a5209709a2f62661ccfd9c8a3e19c74ba","_number":662420,"virtual_id_number":662420,"owner":{"_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2019-06-01 05:33:26.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2019-06-07 08:35:10.000000000","_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"},{"value":1,"date":"2019-05-31 11:24:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2019-06-03 21:05:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 11:24:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-01 05:33:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-03 21:05:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"1885bb1f39b5567720ee1f25bacaa4027317b044","author":{"_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"},"date":"2019-05-31 09:37:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cc3d689be4443c86a5f37d1607a1171c859c045","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 11:24:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/20/662420/1/check/openstack-tox-docs/fd8e138/html/ : SUCCESS in 2m 49s\n- openstack-tox-pypy http://logs.openstack.org/20/662420/1/check/openstack-tox-pypy/7f44ea1/ : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/662420/1/check/openstack-tox-pep8/6c19b41/ : SUCCESS in 2m 34s\n- openstack-tox-py27 http://logs.openstack.org/20/662420/1/check/openstack-tox-py27/6531ee0/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/20/662420/1/check/openstack-tox-py36/842ebb8/ : SUCCESS in 5m 02s\n- openstack-tox-validate http://logs.openstack.org/20/662420/1/check/openstack-tox-validate/c0179fa/ : SUCCESS in 2m 44s\n- requirements-tox-babel http://logs.openstack.org/20/662420/1/check/requirements-tox-babel/58bab56/ : SUCCESS in 2m 29s\n- requirements-tox-py27-check-uc http://logs.openstack.org/20/662420/1/check/requirements-tox-py27-check-uc/bcec45a/ : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc http://logs.openstack.org/20/662420/1/check/requirements-tox-py36-check-uc/d999916/ : SUCCESS in 8m 33s\n- cross-cinder-py27 http://logs.openstack.org/20/662420/1/check/cross-cinder-py27/d438320/ : SUCCESS in 8m 59s\n- cross-cinder-py36 http://logs.openstack.org/20/662420/1/check/cross-cinder-py36/e80fa9e/ : SUCCESS in 7m 20s\n- cross-glance-py27 http://logs.openstack.org/20/662420/1/check/cross-glance-py27/589875d/ : SUCCESS in 3m 50s\n- cross-glance-py36 http://logs.openstack.org/20/662420/1/check/cross-glance-py36/8a2ff69/ : SUCCESS in 3m 45s\n- cross-horizon-py27 http://logs.openstack.org/20/662420/1/check/cross-horizon-py27/2cc178c/ : SUCCESS in 10m 40s\n- cross-horizon-py36 http://logs.openstack.org/20/662420/1/check/cross-horizon-py36/3b0990d/ : SUCCESS in 8m 18s\n- cross-horizon-npm http://logs.openstack.org/20/662420/1/check/cross-horizon-npm/4e95d03/ : SUCCESS in 5m 26s\n- cross-keystone-py27 http://logs.openstack.org/20/662420/1/check/cross-keystone-py27/69253cf/ : SUCCESS in 21m 35s\n- cross-keystone-py36 http://logs.openstack.org/20/662420/1/check/cross-keystone-py36/e19c6fc/ : SUCCESS in 25m 49s\n- cross-neutron-py27 http://logs.openstack.org/20/662420/1/check/cross-neutron-py27/8ed3be8/ : SUCCESS in 21m 13s\n- cross-neutron-py36 http://logs.openstack.org/20/662420/1/check/cross-neutron-py36/c409302/ : SUCCESS in 22m 40s\n- cross-nova-py27 http://logs.openstack.org/20/662420/1/check/cross-nova-py27/2d5f67d/ : SUCCESS in 11m 26s\n- cross-nova-py36 http://logs.openstack.org/20/662420/1/check/cross-nova-py36/f89f836/ : SUCCESS in 10m 43s\n- cross-nova-functional http://logs.openstack.org/20/662420/1/check/cross-nova-functional/fd5e70d/ : SUCCESS in 28m 04s\n- cross-osc-tox-docs http://logs.openstack.org/20/662420/1/check/cross-osc-tox-docs/5706343/html/ : SUCCESS in 5m 15s\n- cross-swift-py27 http://logs.openstack.org/20/662420/1/check/cross-swift-py27/02f1030/ : SUCCESS in 6m 59s\n- cross-swift-py36 http://logs.openstack.org/20/662420/1/check/cross-swift-py36/6d169e2/ : SUCCESS in 10m 13s\n- requirements-integration http://logs.openstack.org/20/662420/1/check/requirements-integration/5c68469/ : SUCCESS in 52m 39s\n- neutron-grenade-multinode http://logs.openstack.org/20/662420/1/check/neutron-grenade-multinode/afd267f/ : SUCCESS in 1h 05m 28s\n- tempest-full http://logs.openstack.org/20/662420/1/check/tempest-full/7b1c12b/ : SUCCESS in 1h 43m 33s\n- neutron-grenade http://logs.openstack.org/20/662420/1/check/neutron-grenade/0a612e7/ : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0133e46853288f2260efecbce7b2af7a8f3e97dd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-31 13:06:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07cbfc068225188e80c0fe6495611e5dae1ddcd3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 21:05:55.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"dc7a0068c0d4992f954258246f125f5444de7781","author":{"_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"},"date":"2019-06-07 08:35:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ddd62f5e5a6ff16dcc2d3982e1298b0b8cefa2dd","author":{"_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"},"date":"2019-06-07 08:35:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09aa3afb89708728e2f26db157a9971625d2019e","revisions":{"09aa3afb89708728e2f26db157a9971625d2019e":{"kind":"REWORK","_number":1,"created":"2019-05-31 09:37:15.000000000","uploader":{"_account_id":29236,"name":"Shubham Potale","email":"Shubham.Potale@nttdata.com","username":"shubham.potale"},"ref":"refs/changes/20/662420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/662420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/662420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/662420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/662420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/662420/1"}}},"commit":{"parents":[{"commit":"8908582fee4cb8f978a5cecfc2e1a6db9a46c5bf","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8908582fee4cb8f978a5cecfc2e1a6db9a46c5bf"}]}],"author":{"name":"Shubham Potale","email":"shubham.potale@nttdata.com","date":"2019-05-28 11:09:45.000000000","tz":330},"committer":{"name":"Shubham Potale","email":"shubham.potale@nttdata.com","date":"2019-05-31 09:34:07.000000000","tz":330},"subject":"Add responses library for tacker test cases","message":"Add responses library for tacker test cases\n\nThis patch adds the responses library for writing test cases in tacker\nto mock the responses of python-heatclient.\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: tacker\nmanaged by openstack: no\n\nChange-Id: Ia33f82cb6561fd8e0ffad7a51c28c3ac922d0c97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09aa3afb89708728e2f26db157a9971625d2019e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09aa3afb89708728e2f26db157a9971625d2019e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
