)]}'
{"id":"openstack%2Frequirements~662486","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I0cfdbb0f57a99eb7a813b5241b33ed59cbea6fc0","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I0cfdbb0f57a99eb7a813b5241b33ed59cbea6fc0","subject":"update constraint for oslo.messaging to new release 5.35.5","status":"MERGED","created":"2019-05-31 14:34:46.000000000","updated":"2019-06-04 06:36:17.000000000","submitted":"2019-06-04 06:36:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662486-1559630177396-cb0ce384","meta_rev_id":"d4b92065bc33348bb953f12e4a0a7b20437e008a","_number":662486,"virtual_id_number":662486,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-04 06:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-31 19:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-03 21:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-06-03 21:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 19:20:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-03 21:06:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-04 06:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39d7465a0b4ea06c98c3316330eddeb7331a7e38","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-31 14:34:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"336862724a651048ec0ea6670f3bac2745bcd27a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 19:11:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/86/662486/1/check/openstack-tox-pypy/3a89e4f/ : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/86/662486/1/check/openstack-tox-pep8/541555f/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/86/662486/1/check/openstack-tox-py27/db218ee/ : SUCCESS in 3m 10s\n- openstack-tox-py35 http://logs.openstack.org/86/662486/1/check/openstack-tox-py35/f55df11/ : SUCCESS in 2m 53s\n- openstack-tox-validate http://logs.openstack.org/86/662486/1/check/openstack-tox-validate/0209509/ : SUCCESS in 2m 42s\n- requirements-tox-babel http://logs.openstack.org/86/662486/1/check/requirements-tox-babel/ab6c5ac/ : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/86/662486/1/check/requirements-tox-py27-check-uc/8783d9b/ : SUCCESS in 6m 23s\n- requirements-tox-py35-check-uc http://logs.openstack.org/86/662486/1/check/requirements-tox-py35-check-uc/3ec9290/ : SUCCESS in 6m 36s\n- cross-cinder-py27 http://logs.openstack.org/86/662486/1/check/cross-cinder-py27/0a99ea3/ : SUCCESS in 7m 45s\n- cross-cinder-py35 http://logs.openstack.org/86/662486/1/check/cross-cinder-py35/a473acf/ : SUCCESS in 9m 05s\n- cross-glance-py27 http://logs.openstack.org/86/662486/1/check/cross-glance-py27/03bb04e/ : SUCCESS in 4m 24s\n- cross-glance-py35 http://logs.openstack.org/86/662486/1/check/cross-glance-py35/ae50da0/ : SUCCESS in 5m 29s\n- cross-horizon-py27 http://logs.openstack.org/86/662486/1/check/cross-horizon-py27/90ad234/ : SUCCESS in 11m 09s\n- cross-horizon-py35 http://logs.openstack.org/86/662486/1/check/cross-horizon-py35/3d7cfe9/ : SUCCESS in 11m 13s\n- cross-keystone-py27 http://logs.openstack.org/86/662486/1/check/cross-keystone-py27/ea7ac69/ : SUCCESS in 14m 09s\n- cross-keystone-py35 http://logs.openstack.org/86/662486/1/check/cross-keystone-py35/1f74020/ : SUCCESS in 24m 39s\n- cross-neutron-py27 http://logs.openstack.org/86/662486/1/check/cross-neutron-py27/88e3c13/ : SUCCESS in 24m 12s\n- cross-neutron-py35 http://logs.openstack.org/86/662486/1/check/cross-neutron-py35/fd16923/ : SUCCESS in 25m 32s\n- cross-nova-py27 http://logs.openstack.org/86/662486/1/check/cross-nova-py27/c1aa119/ : SUCCESS in 13m 08s\n- cross-nova-py35 http://logs.openstack.org/86/662486/1/check/cross-nova-py35/b45421e/ : SUCCESS in 13m 01s\n- cross-nova-functional http://logs.openstack.org/86/662486/1/check/cross-nova-functional/d935018/ : SUCCESS in 20m 19s\n- cross-swift-py27 http://logs.openstack.org/86/662486/1/check/cross-swift-py27/ca89044/ : SUCCESS in 5m 19s\n- cross-swift-py35 http://logs.openstack.org/86/662486/1/check/cross-swift-py35/51291f7/ : SUCCESS in 3m 02s\n- requirements-integration http://logs.openstack.org/86/662486/1/check/requirements-integration/f9245a2/ : SUCCESS in 50m 44s\n- neutron-grenade-multinode http://logs.openstack.org/86/662486/1/check/neutron-grenade-multinode/b5d9c0d/ : SUCCESS in 1h 00m 52s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/86/662486/1/check/legacy-tempest-dsvm-neutron-full/6304a0a/ : SUCCESS in 1h 24m 26s\n- neutron-grenade http://logs.openstack.org/86/662486/1/check/neutron-grenade/5d6ec61/ : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5ce7737a81858182f7685a911217f345dcefa7c5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-31 19:20:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2176eca000032d81edc67266879aa00290a4dbf9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 21:06:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcedfc2334422c15fb4aa4fd06bc727e9403f286","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:07:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25350d39d92014f8ea656976b51acd68e424262b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 00:54:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/86/662486/1/gate/openstack-tox-pep8/7996076/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/86/662486/1/gate/openstack-tox-py27/572a195/ : SUCCESS in 3m 16s\n- openstack-tox-py35 http://logs.openstack.org/86/662486/1/gate/openstack-tox-py35/0269ec3/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/86/662486/1/gate/openstack-tox-validate/24a285e/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/86/662486/1/gate/requirements-tox-babel/92fadca/ : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/86/662486/1/gate/requirements-tox-py27-check-uc/b0c1407/ : SUCCESS in 7m 47s\n- requirements-tox-py35-check-uc http://logs.openstack.org/86/662486/1/gate/requirements-tox-py35-check-uc/a41d9eb/ : SUCCESS in 6m 36s\n- cross-cinder-py27 http://logs.openstack.org/86/662486/1/gate/cross-cinder-py27/add0299/ : SUCCESS in 8m 52s\n- cross-cinder-py35 http://logs.openstack.org/86/662486/1/gate/cross-cinder-py35/8863fb6/ : SUCCESS in 8m 13s\n- cross-glance-py27 http://logs.openstack.org/86/662486/1/gate/cross-glance-py27/dd23404/ : SUCCESS in 5m 11s\n- cross-glance-py35 http://logs.openstack.org/86/662486/1/gate/cross-glance-py35/8e43268/ : SUCCESS in 4m 44s\n- cross-horizon-py27 http://logs.openstack.org/86/662486/1/gate/cross-horizon-py27/a1c9a4d/ : SUCCESS in 13m 02s\n- cross-horizon-py35 http://logs.openstack.org/86/662486/1/gate/cross-horizon-py35/6df883e/ : SUCCESS in 13m 30s\n- cross-keystone-py27 http://logs.openstack.org/86/662486/1/gate/cross-keystone-py27/5a30c61/ : TIMED_OUT in 40m 46s\n- cross-keystone-py35 http://logs.openstack.org/86/662486/1/gate/cross-keystone-py35/8bba6d1/ : SUCCESS in 31m 34s\n- cross-neutron-py27 http://logs.openstack.org/86/662486/1/gate/cross-neutron-py27/875418e/ : SUCCESS in 27m 30s\n- cross-neutron-py35 http://logs.openstack.org/86/662486/1/gate/cross-neutron-py35/95d4de2/ : SUCCESS in 25m 08s\n- cross-nova-py27 http://logs.openstack.org/86/662486/1/gate/cross-nova-py27/495b22f/ : SUCCESS in 14m 22s\n- cross-nova-py35 http://logs.openstack.org/86/662486/1/gate/cross-nova-py35/db92435/ : SUCCESS in 12m 07s\n- cross-nova-functional http://logs.openstack.org/86/662486/1/gate/cross-nova-functional/33f415b/ : SUCCESS in 21m 37s\n- cross-swift-py27 http://logs.openstack.org/86/662486/1/gate/cross-swift-py27/41397c9/ : SUCCESS in 6m 15s\n- cross-swift-py35 http://logs.openstack.org/86/662486/1/gate/cross-swift-py35/5d000f2/ : SUCCESS in 3m 50s\n- requirements-integration http://logs.openstack.org/86/662486/1/gate/requirements-integration/4456d6a/ : SUCCESS in 51m 23s\n- neutron-grenade-multinode http://logs.openstack.org/86/662486/1/gate/neutron-grenade-multinode/a76cac7/ : SUCCESS in 1h 04m 52s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/86/662486/1/gate/legacy-tempest-dsvm-neutron-full/4a07e59/ : SUCCESS in 1h 25m 37s\n- neutron-grenade http://logs.openstack.org/86/662486/1/gate/neutron-grenade/34ed73b/ : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fa10e018ba05ccb21d8d31a53dafb02c89c0b38e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-04 03:23:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"abdb5272a4d1b993792fce6953e45795ac64efaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 04:48:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/86/662486/1/check/openstack-tox-pypy/b564bd8/ : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/86/662486/1/check/openstack-tox-pep8/b0283bc/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/86/662486/1/check/openstack-tox-py27/e1ac125/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/86/662486/1/check/openstack-tox-py35/6c8e4a5/ : SUCCESS in 6m 00s\n- openstack-tox-validate http://logs.openstack.org/86/662486/1/check/openstack-tox-validate/bc69ce9/ : SUCCESS in 3m 15s\n- requirements-tox-babel http://logs.openstack.org/86/662486/1/check/requirements-tox-babel/aadce60/ : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/86/662486/1/check/requirements-tox-py27-check-uc/e1c0902/ : SUCCESS in 6m 13s\n- requirements-tox-py35-check-uc http://logs.openstack.org/86/662486/1/check/requirements-tox-py35-check-uc/c3083cc/ : SUCCESS in 7m 23s\n- cross-cinder-py27 http://logs.openstack.org/86/662486/1/check/cross-cinder-py27/19cf1f2/ : SUCCESS in 9m 54s\n- cross-cinder-py35 http://logs.openstack.org/86/662486/1/check/cross-cinder-py35/903a74b/ : SUCCESS in 8m 46s\n- cross-glance-py27 http://logs.openstack.org/86/662486/1/check/cross-glance-py27/12273aa/ : SUCCESS in 4m 38s\n- cross-glance-py35 http://logs.openstack.org/86/662486/1/check/cross-glance-py35/3773806/ : SUCCESS in 6m 07s\n- cross-horizon-py27 http://logs.openstack.org/86/662486/1/check/cross-horizon-py27/69f679c/ : SUCCESS in 11m 27s\n- cross-horizon-py35 http://logs.openstack.org/86/662486/1/check/cross-horizon-py35/36a7069/ : SUCCESS in 10m 55s\n- cross-keystone-py27 http://logs.openstack.org/86/662486/1/check/cross-keystone-py27/97d1ba8/ : SUCCESS in 12m 39s\n- cross-keystone-py35 http://logs.openstack.org/86/662486/1/check/cross-keystone-py35/937e284/ : SUCCESS in 12m 27s\n- cross-neutron-py27 http://logs.openstack.org/86/662486/1/check/cross-neutron-py27/8aa5a64/ : SUCCESS in 24m 07s\n- cross-neutron-py35 http://logs.openstack.org/86/662486/1/check/cross-neutron-py35/46faabd/ : SUCCESS in 28m 17s\n- cross-nova-py27 http://logs.openstack.org/86/662486/1/check/cross-nova-py27/e510959/ : SUCCESS in 13m 31s\n- cross-nova-py35 http://logs.openstack.org/86/662486/1/check/cross-nova-py35/9890e7a/ : SUCCESS in 14m 02s\n- cross-nova-functional http://logs.openstack.org/86/662486/1/check/cross-nova-functional/c3079a5/ : SUCCESS in 21m 43s\n- cross-swift-py27 http://logs.openstack.org/86/662486/1/check/cross-swift-py27/da4ff4e/ : SUCCESS in 7m 40s\n- cross-swift-py35 http://logs.openstack.org/86/662486/1/check/cross-swift-py35/f4f999b/ : SUCCESS in 3m 31s\n- requirements-integration http://logs.openstack.org/86/662486/1/check/requirements-integration/fb17b77/ : SUCCESS in 49m 54s\n- neutron-grenade-multinode http://logs.openstack.org/86/662486/1/check/neutron-grenade-multinode/51cd33e/ : SUCCESS in 59m 50s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/86/662486/1/check/legacy-tempest-dsvm-neutron-full/0753c10/ : SUCCESS in 1h 17m 59s\n- neutron-grenade http://logs.openstack.org/86/662486/1/check/neutron-grenade/efa971f/ : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":1},{"id":"19b02aac34de5bfc15a719d9cfcbdc6203108697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 04:48:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba5a239fffd281bf42ea3cdb69991579dd9d8053","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 06:36:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/86/662486/1/gate/openstack-tox-pep8/2cbd417/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/86/662486/1/gate/openstack-tox-py27/40d71cf/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/86/662486/1/gate/openstack-tox-py35/fec1365/ : SUCCESS in 5m 01s\n- openstack-tox-validate http://logs.openstack.org/86/662486/1/gate/openstack-tox-validate/ed844a0/ : SUCCESS in 5m 13s\n- requirements-tox-babel http://logs.openstack.org/86/662486/1/gate/requirements-tox-babel/5cc2c06/ : SUCCESS in 5m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/86/662486/1/gate/requirements-tox-py27-check-uc/ea3e4c3/ : SUCCESS in 8m 32s\n- requirements-tox-py35-check-uc http://logs.openstack.org/86/662486/1/gate/requirements-tox-py35-check-uc/d02b17f/ : SUCCESS in 6m 27s\n- cross-cinder-py27 http://logs.openstack.org/86/662486/1/gate/cross-cinder-py27/a0086bf/ : SUCCESS in 10m 56s\n- cross-cinder-py35 http://logs.openstack.org/86/662486/1/gate/cross-cinder-py35/c134d2a/ : SUCCESS in 8m 01s\n- cross-glance-py27 http://logs.openstack.org/86/662486/1/gate/cross-glance-py27/f32edb0/ : SUCCESS in 7m 17s\n- cross-glance-py35 http://logs.openstack.org/86/662486/1/gate/cross-glance-py35/6fcb762/ : SUCCESS in 6m 47s\n- cross-horizon-py27 http://logs.openstack.org/86/662486/1/gate/cross-horizon-py27/e9f5228/ : SUCCESS in 12m 09s\n- cross-horizon-py35 http://logs.openstack.org/86/662486/1/gate/cross-horizon-py35/66c76b7/ : SUCCESS in 11m 56s\n- cross-keystone-py27 http://logs.openstack.org/86/662486/1/gate/cross-keystone-py27/d78cbe2/ : SUCCESS in 15m 16s\n- cross-keystone-py35 http://logs.openstack.org/86/662486/1/gate/cross-keystone-py35/fc0aa5c/ : SUCCESS in 37m 30s\n- cross-neutron-py27 http://logs.openstack.org/86/662486/1/gate/cross-neutron-py27/a5ee908/ : SUCCESS in 22m 00s\n- cross-neutron-py35 http://logs.openstack.org/86/662486/1/gate/cross-neutron-py35/cde33ef/ : SUCCESS in 26m 50s\n- cross-nova-py27 http://logs.openstack.org/86/662486/1/gate/cross-nova-py27/768cec5/ : SUCCESS in 15m 21s\n- cross-nova-py35 http://logs.openstack.org/86/662486/1/gate/cross-nova-py35/723d37a/ : SUCCESS in 14m 24s\n- cross-nova-functional http://logs.openstack.org/86/662486/1/gate/cross-nova-functional/6365a26/ : SUCCESS in 20m 03s\n- cross-swift-py27 http://logs.openstack.org/86/662486/1/gate/cross-swift-py27/deec42e/ : SUCCESS in 6m 44s\n- cross-swift-py35 http://logs.openstack.org/86/662486/1/gate/cross-swift-py35/7df604e/ : SUCCESS in 3m 06s\n- requirements-integration http://logs.openstack.org/86/662486/1/gate/requirements-integration/822a4dc/ : SUCCESS in 1h 14m 53s\n- neutron-grenade-multinode http://logs.openstack.org/86/662486/1/gate/neutron-grenade-multinode/a420bd1/ : SUCCESS in 1h 02m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/86/662486/1/gate/legacy-tempest-dsvm-neutron-full/da019f3/ : SUCCESS in 1h 13m 05s\n- neutron-grenade http://logs.openstack.org/86/662486/1/gate/neutron-grenade/9bf4a89/ : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6d9cde950380e587fb3f0bb4081b2de5599a4324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 06:36:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc9fb39c76474b79c7d1c9e83d732e5b77d066cf","revisions":{"fc9fb39c76474b79c7d1c9e83d732e5b77d066cf":{"kind":"REWORK","_number":1,"created":"2019-05-31 14:34:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/662486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/662486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/662486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/662486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/662486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/662486/1"}}},"commit":{"parents":[{"commit":"6a92e89c14e68c42a149b719d93742979d241c5b","subject":"Fix where we use constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a92e89c14e68c42a149b719d93742979d241c5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-31 14:34:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-31 14:34:45.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 5.35.5","message":"update constraint for oslo.messaging to new release 5.35.5\n\nChange-Id: I0cfdbb0f57a99eb7a813b5241b33ed59cbea6fc0\nmeta:version: 5.35.5\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I0a0efa9f2f9f353696ab07abbb7174cb3c94a027\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc9fb39c76474b79c7d1c9e83d732e5b77d066cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc9fb39c76474b79c7d1c9e83d732e5b77d066cf"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
