)]}'
{"id":"openstack%2Frequirements~662489","triplet_id":"openstack%2Frequirements~stable%2Fqueens~Iad21263a75c7c653ba319617e208ea8eb992b6be","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"Iad21263a75c7c653ba319617e208ea8eb992b6be","subject":"update constraint for oslo.db to new release 4.33.4","status":"MERGED","created":"2019-05-31 14:43:43.000000000","updated":"2019-06-04 00:54:14.000000000","submitted":"2019-06-04 00:54:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662489-1559609654498-576f606b","meta_rev_id":"4527d87aa9e62de4649aecbd582b77f0c28cfc9f","_number":662489,"virtual_id_number":662489,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-04 00:54:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-01 01:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-03 21:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-06-03 21:06:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-01 01:28:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-03 21:06:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-04 00:54:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe5005a8e7567bf630ac69500340b3c47aecd4d8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-31 14:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f74a231e64e8202f579a703f4a89c5dac6e8dc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 19:39:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/89/662489/1/check/openstack-tox-pypy/3352dcd/ : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/662489/1/check/openstack-tox-pep8/ed43e12/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/89/662489/1/check/openstack-tox-py27/8d3f9b6/ : SUCCESS in 3m 11s\n- openstack-tox-py35 http://logs.openstack.org/89/662489/1/check/openstack-tox-py35/bf56e39/ : SUCCESS in 3m 09s\n- openstack-tox-validate http://logs.openstack.org/89/662489/1/check/openstack-tox-validate/b09810e/ : SUCCESS in 3m 54s\n- requirements-tox-babel http://logs.openstack.org/89/662489/1/check/requirements-tox-babel/27a7f9b/ : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/89/662489/1/check/requirements-tox-py27-check-uc/b39beaa/ : SUCCESS in 5m 57s\n- requirements-tox-py35-check-uc http://logs.openstack.org/89/662489/1/check/requirements-tox-py35-check-uc/26be31c/ : SUCCESS in 5m 44s\n- cross-cinder-py27 http://logs.openstack.org/89/662489/1/check/cross-cinder-py27/9494631/ : SUCCESS in 8m 11s\n- cross-cinder-py35 http://logs.openstack.org/89/662489/1/check/cross-cinder-py35/678b8a4/ : SUCCESS in 7m 20s\n- cross-glance-py27 http://logs.openstack.org/89/662489/1/check/cross-glance-py27/a15d6f1/ : SUCCESS in 5m 14s\n- cross-glance-py35 http://logs.openstack.org/89/662489/1/check/cross-glance-py35/a175f2c/ : SUCCESS in 5m 04s\n- cross-horizon-py27 http://logs.openstack.org/89/662489/1/check/cross-horizon-py27/8957132/ : SUCCESS in 12m 11s\n- cross-horizon-py35 http://logs.openstack.org/89/662489/1/check/cross-horizon-py35/64e2cf7/ : SUCCESS in 10m 59s\n- cross-keystone-py27 http://logs.openstack.org/89/662489/1/check/cross-keystone-py27/fed5fdc/ : SUCCESS in 11m 50s\n- cross-keystone-py35 http://logs.openstack.org/89/662489/1/check/cross-keystone-py35/dced533/ : SUCCESS in 13m 24s\n- cross-neutron-py27 http://logs.openstack.org/89/662489/1/check/cross-neutron-py27/da1b12e/ : SUCCESS in 34m 11s\n- cross-neutron-py35 http://logs.openstack.org/89/662489/1/check/cross-neutron-py35/8baf069/ : SUCCESS in 32m 23s\n- cross-nova-py27 http://logs.openstack.org/89/662489/1/check/cross-nova-py27/5b49101/ : SUCCESS in 13m 39s\n- cross-nova-py35 http://logs.openstack.org/89/662489/1/check/cross-nova-py35/8ade9eb/ : FAILURE in 27m 55s\n- cross-nova-functional http://logs.openstack.org/89/662489/1/check/cross-nova-functional/c65e1b9/ : SUCCESS in 22m 59s\n- cross-swift-py27 http://logs.openstack.org/89/662489/1/check/cross-swift-py27/2a07617/ : SUCCESS in 5m 13s\n- cross-swift-py35 http://logs.openstack.org/89/662489/1/check/cross-swift-py35/f37407a/ : SUCCESS in 3m 24s\n- requirements-integration http://logs.openstack.org/89/662489/1/check/requirements-integration/c86e2aa/ : SUCCESS in 51m 19s\n- neutron-grenade-multinode http://logs.openstack.org/89/662489/1/check/neutron-grenade-multinode/3ec01e8/ : SUCCESS in 1h 19m 35s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/662489/1/check/legacy-tempest-dsvm-neutron-full/fb79ea5/ : SUCCESS in 1h 32m 56s\n- neutron-grenade http://logs.openstack.org/89/662489/1/check/neutron-grenade/76c2b67/ : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ac2a4aa7e1b872ffcc6d4dc2d9b456a6a61b4088","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-31 20:35:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b1e812d68550adfd045baf1c76c34f5862bf248","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 01:25:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/89/662489/1/check/openstack-tox-pypy/c5da178/ : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/662489/1/check/openstack-tox-pep8/51d4fb1/ : SUCCESS in 2m 40s\n- openstack-tox-py27 http://logs.openstack.org/89/662489/1/check/openstack-tox-py27/2adfda5/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/89/662489/1/check/openstack-tox-py35/110b7aa/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/89/662489/1/check/openstack-tox-validate/d44e323/ : SUCCESS in 3m 53s\n- requirements-tox-babel http://logs.openstack.org/89/662489/1/check/requirements-tox-babel/2d60df0/ : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/89/662489/1/check/requirements-tox-py27-check-uc/2ce73ee/ : SUCCESS in 6m 48s\n- requirements-tox-py35-check-uc http://logs.openstack.org/89/662489/1/check/requirements-tox-py35-check-uc/5daeb04/ : SUCCESS in 6m 40s\n- cross-cinder-py27 http://logs.openstack.org/89/662489/1/check/cross-cinder-py27/8684f1c/ : SUCCESS in 8m 36s\n- cross-cinder-py35 http://logs.openstack.org/89/662489/1/check/cross-cinder-py35/e4610ff/ : SUCCESS in 7m 22s\n- cross-glance-py27 http://logs.openstack.org/89/662489/1/check/cross-glance-py27/0fab378/ : SUCCESS in 4m 16s\n- cross-glance-py35 http://logs.openstack.org/89/662489/1/check/cross-glance-py35/05e54d0/ : SUCCESS in 4m 37s\n- cross-horizon-py27 http://logs.openstack.org/89/662489/1/check/cross-horizon-py27/ca7d367/ : SUCCESS in 11m 22s\n- cross-horizon-py35 http://logs.openstack.org/89/662489/1/check/cross-horizon-py35/f79b6fe/ : SUCCESS in 11m 22s\n- cross-keystone-py27 http://logs.openstack.org/89/662489/1/check/cross-keystone-py27/d26fbbe/ : TIMED_OUT in 40m 54s\n- cross-keystone-py35 http://logs.openstack.org/89/662489/1/check/cross-keystone-py35/36ebc76/ : SUCCESS in 14m 32s\n- cross-neutron-py27 http://logs.openstack.org/89/662489/1/check/cross-neutron-py27/a07b3e2/ : SUCCESS in 22m 16s\n- cross-neutron-py35 http://logs.openstack.org/89/662489/1/check/cross-neutron-py35/f55ea1c/ : SUCCESS in 24m 24s\n- cross-nova-py27 http://logs.openstack.org/89/662489/1/check/cross-nova-py27/4355bf5/ : SUCCESS in 16m 37s\n- cross-nova-py35 http://logs.openstack.org/89/662489/1/check/cross-nova-py35/9be4457/ : SUCCESS in 14m 21s\n- cross-nova-functional http://logs.openstack.org/89/662489/1/check/cross-nova-functional/ac11b77/ : SUCCESS in 20m 12s\n- cross-swift-py27 http://logs.openstack.org/89/662489/1/check/cross-swift-py27/833dca3/ : SUCCESS in 5m 15s\n- cross-swift-py35 http://logs.openstack.org/89/662489/1/check/cross-swift-py35/8a18eee/ : SUCCESS in 3m 25s\n- requirements-integration http://logs.openstack.org/89/662489/1/check/requirements-integration/eca8fa5/ : SUCCESS in 52m 06s\n- neutron-grenade-multinode http://logs.openstack.org/89/662489/1/check/neutron-grenade-multinode/25955d7/ : SUCCESS in 1h 02m 29s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/662489/1/check/legacy-tempest-dsvm-neutron-full/67d0fbf/ : SUCCESS in 1h 21m 09s\n- neutron-grenade http://logs.openstack.org/89/662489/1/check/neutron-grenade/8b269f4/ : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4316c807e4caa7c4a8c10b7d26cea9689432a28a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-01 01:28:31.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b877c39da1238ad521e467d7539259c9fee6210","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 02:56:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/89/662489/1/check/openstack-tox-pypy/f4d934f/ : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/662489/1/check/openstack-tox-pep8/6822959/ : SUCCESS in 2m 25s\n- openstack-tox-py27 http://logs.openstack.org/89/662489/1/check/openstack-tox-py27/0bce6b2/ : SUCCESS in 3m 13s\n- openstack-tox-py35 http://logs.openstack.org/89/662489/1/check/openstack-tox-py35/f89dbc2/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/89/662489/1/check/openstack-tox-validate/7a5e2fa/ : SUCCESS in 3m 15s\n- requirements-tox-babel http://logs.openstack.org/89/662489/1/check/requirements-tox-babel/8a67120/ : SUCCESS in 3m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/89/662489/1/check/requirements-tox-py27-check-uc/9e489db/ : SUCCESS in 6m 02s\n- requirements-tox-py35-check-uc http://logs.openstack.org/89/662489/1/check/requirements-tox-py35-check-uc/7aaeb88/ : SUCCESS in 6m 25s\n- cross-cinder-py27 http://logs.openstack.org/89/662489/1/check/cross-cinder-py27/cbe8508/ : FAILURE in 9m 42s\n- cross-cinder-py35 http://logs.openstack.org/89/662489/1/check/cross-cinder-py35/4fff2c2/ : SUCCESS in 7m 35s\n- cross-glance-py27 http://logs.openstack.org/89/662489/1/check/cross-glance-py27/3008254/ : SUCCESS in 4m 41s\n- cross-glance-py35 http://logs.openstack.org/89/662489/1/check/cross-glance-py35/67f989a/ : SUCCESS in 5m 01s\n- cross-horizon-py27 http://logs.openstack.org/89/662489/1/check/cross-horizon-py27/2e16f35/ : SUCCESS in 13m 15s\n- cross-horizon-py35 http://logs.openstack.org/89/662489/1/check/cross-horizon-py35/31b7603/ : SUCCESS in 9m 59s\n- cross-keystone-py27 http://logs.openstack.org/89/662489/1/check/cross-keystone-py27/1ba0050/ : SUCCESS in 37m 26s\n- cross-keystone-py35 http://logs.openstack.org/89/662489/1/check/cross-keystone-py35/28e2ca3/ : SUCCESS in 14m 17s\n- cross-neutron-py27 http://logs.openstack.org/89/662489/1/check/cross-neutron-py27/2a2e68d/ : SUCCESS in 25m 29s\n- cross-neutron-py35 http://logs.openstack.org/89/662489/1/check/cross-neutron-py35/17ed61d/ : SUCCESS in 26m 18s\n- cross-nova-py27 http://logs.openstack.org/89/662489/1/check/cross-nova-py27/b84485e/ : SUCCESS in 11m 29s\n- cross-nova-py35 http://logs.openstack.org/89/662489/1/check/cross-nova-py35/d034db6/ : SUCCESS in 15m 29s\n- cross-nova-functional http://logs.openstack.org/89/662489/1/check/cross-nova-functional/0eb70c1/ : SUCCESS in 20m 26s\n- cross-swift-py27 http://logs.openstack.org/89/662489/1/check/cross-swift-py27/ecd853b/ : SUCCESS in 5m 32s\n- cross-swift-py35 http://logs.openstack.org/89/662489/1/check/cross-swift-py35/e3ed30d/ : SUCCESS in 4m 22s\n- requirements-integration http://logs.openstack.org/89/662489/1/check/requirements-integration/54e5656/ : SUCCESS in 54m 41s\n- neutron-grenade-multinode http://logs.openstack.org/89/662489/1/check/neutron-grenade-multinode/fd08bde/ : SUCCESS in 50m 40s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/662489/1/check/legacy-tempest-dsvm-neutron-full/1f697c1/ : SUCCESS in 1h 21m 41s\n- neutron-grenade http://logs.openstack.org/89/662489/1/check/neutron-grenade/15e3996/ : SUCCESS in 44m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e8412961f3a91e6cffe57afb1fda7e094ba455d6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-01 03:00:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1aad8d3855cf7bd909733cb25ad4c772ecd572c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 04:46:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/89/662489/1/check/openstack-tox-pypy/e13ea01/ : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/662489/1/check/openstack-tox-pep8/283ca98/ : SUCCESS in 2m 41s\n- openstack-tox-py27 http://logs.openstack.org/89/662489/1/check/openstack-tox-py27/e9492a0/ : SUCCESS in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/89/662489/1/check/openstack-tox-py35/b753ab1/ : SUCCESS in 3m 02s\n- openstack-tox-validate http://logs.openstack.org/89/662489/1/check/openstack-tox-validate/2263f40/ : SUCCESS in 2m 35s\n- requirements-tox-babel http://logs.openstack.org/89/662489/1/check/requirements-tox-babel/86df04e/ : SUCCESS in 2m 16s\n- requirements-tox-py27-check-uc http://logs.openstack.org/89/662489/1/check/requirements-tox-py27-check-uc/b5ec716/ : SUCCESS in 6m 22s\n- requirements-tox-py35-check-uc http://logs.openstack.org/89/662489/1/check/requirements-tox-py35-check-uc/97ee8f4/ : SUCCESS in 6m 56s\n- cross-cinder-py27 http://logs.openstack.org/89/662489/1/check/cross-cinder-py27/789d968/ : SUCCESS in 7m 54s\n- cross-cinder-py35 http://logs.openstack.org/89/662489/1/check/cross-cinder-py35/aebf447/ : SUCCESS in 7m 44s\n- cross-glance-py27 http://logs.openstack.org/89/662489/1/check/cross-glance-py27/69c0cf8/ : SUCCESS in 4m 18s\n- cross-glance-py35 http://logs.openstack.org/89/662489/1/check/cross-glance-py35/8f7e929/ : SUCCESS in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/89/662489/1/check/cross-horizon-py27/26c5470/ : SUCCESS in 11m 06s\n- cross-horizon-py35 http://logs.openstack.org/89/662489/1/check/cross-horizon-py35/34c9c6a/ : SUCCESS in 10m 28s\n- cross-keystone-py27 http://logs.openstack.org/89/662489/1/check/cross-keystone-py27/0aba4f7/ : SUCCESS in 12m 39s\n- cross-keystone-py35 http://logs.openstack.org/89/662489/1/check/cross-keystone-py35/46051d6/ : SUCCESS in 11m 28s\n- cross-neutron-py27 http://logs.openstack.org/89/662489/1/check/cross-neutron-py27/7ac2d88/ : SUCCESS in 27m 38s\n- cross-neutron-py35 http://logs.openstack.org/89/662489/1/check/cross-neutron-py35/a1ff494/ : SUCCESS in 23m 03s\n- cross-nova-py27 http://logs.openstack.org/89/662489/1/check/cross-nova-py27/d1884c6/ : SUCCESS in 11m 36s\n- cross-nova-py35 http://logs.openstack.org/89/662489/1/check/cross-nova-py35/287e4b5/ : SUCCESS in 11m 40s\n- cross-nova-functional http://logs.openstack.org/89/662489/1/check/cross-nova-functional/4f9c657/ : SUCCESS in 18m 43s\n- cross-swift-py27 http://logs.openstack.org/89/662489/1/check/cross-swift-py27/b757407/ : SUCCESS in 5m 08s\n- cross-swift-py35 http://logs.openstack.org/89/662489/1/check/cross-swift-py35/3bc52bb/ : SUCCESS in 3m 08s\n- requirements-integration http://logs.openstack.org/89/662489/1/check/requirements-integration/8ad6f63/ : SUCCESS in 50m 28s\n- neutron-grenade-multinode http://logs.openstack.org/89/662489/1/check/neutron-grenade-multinode/6dd267b/ : SUCCESS in 1h 03m 06s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/662489/1/check/legacy-tempest-dsvm-neutron-full/972c1d2/ : SUCCESS in 1h 20m 46s\n- neutron-grenade http://logs.openstack.org/89/662489/1/check/neutron-grenade/aafb9e0/ : SUCCESS in 1h 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"52ec46bdf62c83862c7876e49ac8d6c79516668d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 21:06:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf38bc94bc40097d061fa4d056b586acbb41dd78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:06:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"392b330ecee8a069ed5b703561537b95550c7699","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 00:54:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c96efaceaf8ab22d437c809737dfc35b7649d54f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 00:54:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/89/662489/1/gate/openstack-tox-pep8/3b80401/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/89/662489/1/gate/openstack-tox-py27/22f7c8c/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/89/662489/1/gate/openstack-tox-py35/70d68d7/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/89/662489/1/gate/openstack-tox-validate/b3819df/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/89/662489/1/gate/requirements-tox-babel/a3938f6/ : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc http://logs.openstack.org/89/662489/1/gate/requirements-tox-py27-check-uc/451d42b/ : SUCCESS in 6m 32s\n- requirements-tox-py35-check-uc http://logs.openstack.org/89/662489/1/gate/requirements-tox-py35-check-uc/5f3d0e0/ : SUCCESS in 6m 57s\n- cross-cinder-py27 http://logs.openstack.org/89/662489/1/gate/cross-cinder-py27/d7696d4/ : SUCCESS in 8m 14s\n- cross-cinder-py35 http://logs.openstack.org/89/662489/1/gate/cross-cinder-py35/a4dcb45/ : SUCCESS in 8m 25s\n- cross-glance-py27 http://logs.openstack.org/89/662489/1/gate/cross-glance-py27/04df596/ : SUCCESS in 4m 42s\n- cross-glance-py35 http://logs.openstack.org/89/662489/1/gate/cross-glance-py35/7e83c27/ : SUCCESS in 5m 06s\n- cross-horizon-py27 http://logs.openstack.org/89/662489/1/gate/cross-horizon-py27/8998ed9/ : SUCCESS in 11m 47s\n- cross-horizon-py35 http://logs.openstack.org/89/662489/1/gate/cross-horizon-py35/6a04e31/ : SUCCESS in 11m 12s\n- cross-keystone-py27 http://logs.openstack.org/89/662489/1/gate/cross-keystone-py27/ab61c84/ : SUCCESS in 27m 42s\n- cross-keystone-py35 http://logs.openstack.org/89/662489/1/gate/cross-keystone-py35/d99e0b0/ : SUCCESS in 27m 40s\n- cross-neutron-py27 http://logs.openstack.org/89/662489/1/gate/cross-neutron-py27/c0630bb/ : SUCCESS in 25m 59s\n- cross-neutron-py35 http://logs.openstack.org/89/662489/1/gate/cross-neutron-py35/0ee168f/ : SUCCESS in 35m 26s\n- cross-nova-py27 http://logs.openstack.org/89/662489/1/gate/cross-nova-py27/b3d659d/ : SUCCESS in 15m 19s\n- cross-nova-py35 http://logs.openstack.org/89/662489/1/gate/cross-nova-py35/2a272da/ : SUCCESS in 19m 46s\n- cross-nova-functional http://logs.openstack.org/89/662489/1/gate/cross-nova-functional/bd8d58a/ : SUCCESS in 30m 44s\n- cross-swift-py27 http://logs.openstack.org/89/662489/1/gate/cross-swift-py27/2138e59/ : SUCCESS in 5m 33s\n- cross-swift-py35 http://logs.openstack.org/89/662489/1/gate/cross-swift-py35/bac4748/ : SUCCESS in 3m 47s\n- requirements-integration http://logs.openstack.org/89/662489/1/gate/requirements-integration/b77e90e/ : SUCCESS in 50m 15s\n- neutron-grenade-multinode http://logs.openstack.org/89/662489/1/gate/neutron-grenade-multinode/bcec233/ : SUCCESS in 1h 02m 31s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/89/662489/1/gate/legacy-tempest-dsvm-neutron-full/30cf497/ : SUCCESS in 1h 21m 35s\n- neutron-grenade http://logs.openstack.org/89/662489/1/gate/neutron-grenade/5d2fa67/ : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fa8ffc1b2a61643a0599f9e14b4ba0b5f813641","revisions":{"0fa8ffc1b2a61643a0599f9e14b4ba0b5f813641":{"kind":"REWORK","_number":1,"created":"2019-05-31 14:43:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/662489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/662489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/662489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/662489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/662489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/662489/1"}}},"commit":{"parents":[{"commit":"6a92e89c14e68c42a149b719d93742979d241c5b","subject":"Fix where we use constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a92e89c14e68c42a149b719d93742979d241c5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-31 14:43:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-31 14:43:41.000000000","tz":0},"subject":"update constraint for oslo.db to new release 4.33.4","message":"update constraint for oslo.db to new release 4.33.4\n\nChange-Id: Iad21263a75c7c653ba319617e208ea8eb992b6be\nmeta:version: 4.33.4\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I0a0efa9f2f9f353696ab07abbb7174cb3c94a027\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fa8ffc1b2a61643a0599f9e14b4ba0b5f813641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fa8ffc1b2a61643a0599f9e14b4ba0b5f813641"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
