)]}'
{"id":"openstack%2Frequirements~662490","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I86ed11b56eb25f51dc31ac17e361381980bd917f","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I86ed11b56eb25f51dc31ac17e361381980bd917f","subject":"update constraint for oslo.cache to new release 1.28.1","status":"MERGED","created":"2019-05-31 14:46:40.000000000","updated":"2019-06-04 00:39:36.000000000","submitted":"2019-06-04 00:39:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662490-1559608776705-576f606b","meta_rev_id":"372b82d0bd4205939a35f8c5e1f5feceb495e136","_number":662490,"virtual_id_number":662490,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-04 00:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-31 19:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-03 21:05:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-03 01:21:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-06-03 21:05:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 19:36:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-03 01:21:27.000000000","updated_by":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"reviewer":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"state":"REVIEWER"},{"updated":"2019-06-03 21:05:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-04 00:39:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c647f170f7fd85900e7a4030f939066f1a7f41c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-31 14:46:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28ddc24db05b1dd6940f91241b889ac0fb047bf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 19:24:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/90/662490/1/check/openstack-tox-pypy/4803002/ : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/90/662490/1/check/openstack-tox-pep8/800b4a1/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/90/662490/1/check/openstack-tox-py27/a550a44/ : SUCCESS in 4m 22s\n- openstack-tox-py35 http://logs.openstack.org/90/662490/1/check/openstack-tox-py35/8aafa6c/ : SUCCESS in 2m 52s\n- openstack-tox-validate http://logs.openstack.org/90/662490/1/check/openstack-tox-validate/9c23081/ : SUCCESS in 4m 45s\n- requirements-tox-babel http://logs.openstack.org/90/662490/1/check/requirements-tox-babel/14e4179/ : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc http://logs.openstack.org/90/662490/1/check/requirements-tox-py27-check-uc/cec3ad3/ : SUCCESS in 6m 47s\n- requirements-tox-py35-check-uc http://logs.openstack.org/90/662490/1/check/requirements-tox-py35-check-uc/f5d5dfd/ : SUCCESS in 6m 30s\n- cross-cinder-py27 http://logs.openstack.org/90/662490/1/check/cross-cinder-py27/97c0663/ : SUCCESS in 7m 37s\n- cross-cinder-py35 http://logs.openstack.org/90/662490/1/check/cross-cinder-py35/4035dd6/ : SUCCESS in 8m 19s\n- cross-glance-py27 http://logs.openstack.org/90/662490/1/check/cross-glance-py27/fc3c64c/ : SUCCESS in 4m 29s\n- cross-glance-py35 http://logs.openstack.org/90/662490/1/check/cross-glance-py35/c924540/ : SUCCESS in 6m 00s\n- cross-horizon-py27 http://logs.openstack.org/90/662490/1/check/cross-horizon-py27/c58cf77/ : SUCCESS in 11m 07s\n- cross-horizon-py35 http://logs.openstack.org/90/662490/1/check/cross-horizon-py35/a62f259/ : SUCCESS in 11m 09s\n- cross-keystone-py27 http://logs.openstack.org/90/662490/1/check/cross-keystone-py27/20a45ea/ : SUCCESS in 19m 33s\n- cross-keystone-py35 http://logs.openstack.org/90/662490/1/check/cross-keystone-py35/d12a764/ : SUCCESS in 11m 46s\n- cross-neutron-py27 http://logs.openstack.org/90/662490/1/check/cross-neutron-py27/56a9a20/ : SUCCESS in 24m 55s\n- cross-neutron-py35 http://logs.openstack.org/90/662490/1/check/cross-neutron-py35/3700c42/ : SUCCESS in 25m 18s\n- cross-nova-py27 http://logs.openstack.org/90/662490/1/check/cross-nova-py27/4f0d651/ : SUCCESS in 12m 49s\n- cross-nova-py35 http://logs.openstack.org/90/662490/1/check/cross-nova-py35/ee69e1e/ : SUCCESS in 13m 23s\n- cross-nova-functional http://logs.openstack.org/90/662490/1/check/cross-nova-functional/0144ab8/ : SUCCESS in 18m 28s\n- cross-swift-py27 http://logs.openstack.org/90/662490/1/check/cross-swift-py27/c25b9a1/ : SUCCESS in 6m 14s\n- cross-swift-py35 http://logs.openstack.org/90/662490/1/check/cross-swift-py35/5391bff/ : SUCCESS in 3m 26s\n- requirements-integration http://logs.openstack.org/90/662490/1/check/requirements-integration/f42a48b/ : SUCCESS in 56m 59s\n- neutron-grenade-multinode http://logs.openstack.org/90/662490/1/check/neutron-grenade-multinode/168c48f/ : SUCCESS in 53m 47s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/90/662490/1/check/legacy-tempest-dsvm-neutron-full/52b73a1/ : SUCCESS in 1h 18m 05s\n- neutron-grenade http://logs.openstack.org/90/662490/1/check/neutron-grenade/74a3388/ : SUCCESS in 47m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7c9125235d7c0674e326fcf856ce8edf1864ff37","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-31 19:36:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc4e28d30e54afa22ad9771c6bb7f3a162f4ea8d","author":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"date":"2019-06-03 01:21:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b0638689bff1ed62928fbdc884e258e929b388da","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 21:05:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8265c1a21a7fe06aad7196bea838312dc5a2067","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:06:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"996c74cbb6efd9687edb999c000f3b7907747f72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 00:39:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/662490/1/gate/openstack-tox-pep8/a57bb17/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/90/662490/1/gate/openstack-tox-py27/a0e0c8b/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/90/662490/1/gate/openstack-tox-py35/a4f99ad/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/90/662490/1/gate/openstack-tox-validate/6ae73da/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/90/662490/1/gate/requirements-tox-babel/e9a102f/ : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/90/662490/1/gate/requirements-tox-py27-check-uc/f4806b5/ : SUCCESS in 6m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/90/662490/1/gate/requirements-tox-py35-check-uc/946023a/ : SUCCESS in 6m 58s\n- cross-cinder-py27 http://logs.openstack.org/90/662490/1/gate/cross-cinder-py27/8c435b9/ : SUCCESS in 8m 26s\n- cross-cinder-py35 http://logs.openstack.org/90/662490/1/gate/cross-cinder-py35/6c8e0c9/ : SUCCESS in 8m 31s\n- cross-glance-py27 http://logs.openstack.org/90/662490/1/gate/cross-glance-py27/9d965ab/ : SUCCESS in 4m 14s\n- cross-glance-py35 http://logs.openstack.org/90/662490/1/gate/cross-glance-py35/5b3ce4f/ : SUCCESS in 4m 57s\n- cross-horizon-py27 http://logs.openstack.org/90/662490/1/gate/cross-horizon-py27/3c7d710/ : SUCCESS in 11m 27s\n- cross-horizon-py35 http://logs.openstack.org/90/662490/1/gate/cross-horizon-py35/2dbcfd8/ : SUCCESS in 10m 51s\n- cross-keystone-py27 http://logs.openstack.org/90/662490/1/gate/cross-keystone-py27/e952dba/ : SUCCESS in 12m 43s\n- cross-keystone-py35 http://logs.openstack.org/90/662490/1/gate/cross-keystone-py35/9b891b9/ : SUCCESS in 12m 51s\n- cross-neutron-py27 http://logs.openstack.org/90/662490/1/gate/cross-neutron-py27/e2265e5/ : SUCCESS in 26m 16s\n- cross-neutron-py35 http://logs.openstack.org/90/662490/1/gate/cross-neutron-py35/1f3e427/ : SUCCESS in 24m 25s\n- cross-nova-py27 http://logs.openstack.org/90/662490/1/gate/cross-nova-py27/abbecc6/ : SUCCESS in 13m 07s\n- cross-nova-py35 http://logs.openstack.org/90/662490/1/gate/cross-nova-py35/05b5812/ : SUCCESS in 12m 57s\n- cross-nova-functional http://logs.openstack.org/90/662490/1/gate/cross-nova-functional/2543a4b/ : SUCCESS in 19m 53s\n- cross-swift-py27 http://logs.openstack.org/90/662490/1/gate/cross-swift-py27/e2fb8d1/ : SUCCESS in 5m 56s\n- cross-swift-py35 http://logs.openstack.org/90/662490/1/gate/cross-swift-py35/44cea27/ : SUCCESS in 3m 42s\n- requirements-integration http://logs.openstack.org/90/662490/1/gate/requirements-integration/97aad0d/ : SUCCESS in 53m 23s\n- neutron-grenade-multinode http://logs.openstack.org/90/662490/1/gate/neutron-grenade-multinode/4bcffa4/ : SUCCESS in 1h 06m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/90/662490/1/gate/legacy-tempest-dsvm-neutron-full/3cc0ed0/ : SUCCESS in 1h 14m 40s\n- neutron-grenade http://logs.openstack.org/90/662490/1/gate/neutron-grenade/9b9580e/ : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e1b6d5853184c4860109f7b719c9199fed9f6dcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 00:39:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bac8cad8021ecc165f52bd482f5189a59041328","revisions":{"8bac8cad8021ecc165f52bd482f5189a59041328":{"kind":"REWORK","_number":1,"created":"2019-05-31 14:46:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/662490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/662490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/662490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/662490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/662490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/662490/1"}}},"commit":{"parents":[{"commit":"6a92e89c14e68c42a149b719d93742979d241c5b","subject":"Fix where we use constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a92e89c14e68c42a149b719d93742979d241c5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-31 14:46:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-31 14:46:37.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 1.28.1","message":"update constraint for oslo.cache to new release 1.28.1\n\nChange-Id: I86ed11b56eb25f51dc31ac17e361381980bd917f\nmeta:version: 1.28.1\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I0a0efa9f2f9f353696ab07abbb7174cb3c94a027\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8bac8cad8021ecc165f52bd482f5189a59041328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8bac8cad8021ecc165f52bd482f5189a59041328"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
