)]}'
{"id":"openstack%2Frequirements~662598","triplet_id":"openstack%2Frequirements~master~I7b2ae14f14326463c34bf227dc759dfa3c038bd3","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I7b2ae14f14326463c34bf227dc759dfa3c038bd3","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-06-01 06:13:25.000000000","updated":"2019-06-05 22:48:20.000000000","submitted":"2019-06-05 22:48:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662598-1559774900389-c7054fac","meta_rev_id":"37fe8c8c53c75edc653f826a88239e13745d02e3","_number":662598,"virtual_id_number":662598,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-05 22:48:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-04 22:30:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-04 22:30:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 21:09:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-04 22:30:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-05 22:48:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df2b7ed7923b2b2b1494bd11dbab05855cd9bd87","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-01 06:13:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e844e88b061459fe3200af7f3fd7135f5609aa9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 07:40:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/1/check/openstack-tox-docs/dc7bc19/html/ : SUCCESS in 3m 24s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/1/check/openstack-tox-pypy/0cac19e/ : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/1/check/openstack-tox-pep8/83c572b/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/1/check/openstack-tox-py27/968051b/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/1/check/openstack-tox-py36/6261e27/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/98/662598/1/check/openstack-tox-validate/2c95a6e/ : SUCCESS in 3m 00s\n- requirements-tox-babel http://logs.openstack.org/98/662598/1/check/requirements-tox-babel/40001ec/ : SUCCESS in 2m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/1/check/requirements-tox-py27-check-uc/654ac9d/ : FAILURE in 7m 12s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/1/check/requirements-tox-py36-check-uc/4cebf36/ : FAILURE in 9m 58s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/1/check/cross-cinder-py27/658766e/ : SUCCESS in 7m 45s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/1/check/cross-cinder-py36/2b3936a/ : SUCCESS in 10m 04s\n- cross-glance-py27 http://logs.openstack.org/98/662598/1/check/cross-glance-py27/7a51bd3/ : SUCCESS in 6m 20s\n- cross-glance-py36 http://logs.openstack.org/98/662598/1/check/cross-glance-py36/3899d1d/ : SUCCESS in 5m 31s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/1/check/cross-horizon-py27/7dce598/ : SUCCESS in 10m 54s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/1/check/cross-horizon-py36/6d775f2/ : SUCCESS in 8m 59s\n- cross-horizon-npm http://logs.openstack.org/98/662598/1/check/cross-horizon-npm/d610253/ : SUCCESS in 5m 42s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/1/check/cross-keystone-py27/e1c3d96/ : SUCCESS in 30m 16s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/1/check/cross-keystone-py36/1ba4170/ : SUCCESS in 21m 24s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/1/check/cross-neutron-py27/44e913b/ : SUCCESS in 22m 10s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/1/check/cross-neutron-py36/72c9c78/ : SUCCESS in 30m 07s\n- cross-nova-py27 http://logs.openstack.org/98/662598/1/check/cross-nova-py27/a9349b4/ : SUCCESS in 11m 13s\n- cross-nova-py36 http://logs.openstack.org/98/662598/1/check/cross-nova-py36/69cf948/ : SUCCESS in 11m 34s\n- cross-nova-functional http://logs.openstack.org/98/662598/1/check/cross-nova-functional/4986572/ : SUCCESS in 17m 56s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/1/check/cross-osc-tox-docs/4fc2ff9/html/ : SUCCESS in 5m 09s\n- cross-swift-py27 http://logs.openstack.org/98/662598/1/check/cross-swift-py27/947b506/ : SUCCESS in 7m 33s\n- cross-swift-py36 http://logs.openstack.org/98/662598/1/check/cross-swift-py36/ca4d7c2/ : SUCCESS in 8m 40s\n- requirements-integration http://logs.openstack.org/98/662598/1/check/requirements-integration/227884e/ : SUCCESS in 54m 48s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/1/check/neutron-grenade-multinode/a7c7ce2/ : SUCCESS in 1h 09m 51s\n- tempest-full http://logs.openstack.org/98/662598/1/check/tempest-full/67dc4d1/ : FAILURE in 1h 19m 14s\n- neutron-grenade http://logs.openstack.org/98/662598/1/check/neutron-grenade/3c3b7d8/ : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"21c3560ba8e1dc890395b8e3a1f27bce802feee1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-02 06:10:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8d27722537e8053e15890d9f2dd0db15a9ee196","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 07:43:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/2/check/openstack-tox-docs/5ad9657/html/ : SUCCESS in 2m 46s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/2/check/openstack-tox-pypy/596b4a5/ : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/2/check/openstack-tox-pep8/39c181d/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/2/check/openstack-tox-py27/5fd1fbc/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/2/check/openstack-tox-py36/8b945f7/ : SUCCESS in 4m 26s\n- openstack-tox-validate http://logs.openstack.org/98/662598/2/check/openstack-tox-validate/35ffa94/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/98/662598/2/check/requirements-tox-babel/d04c7c2/ : SUCCESS in 2m 47s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/2/check/requirements-tox-py27-check-uc/afd56ab/ : FAILURE in 7m 02s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/2/check/requirements-tox-py36-check-uc/0585a17/ : FAILURE in 6m 45s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/2/check/cross-cinder-py27/d0022ca/ : SUCCESS in 7m 09s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/2/check/cross-cinder-py36/0b1fe64/ : SUCCESS in 8m 18s\n- cross-glance-py27 http://logs.openstack.org/98/662598/2/check/cross-glance-py27/1fefdac/ : SUCCESS in 4m 17s\n- cross-glance-py36 http://logs.openstack.org/98/662598/2/check/cross-glance-py36/99b33e3/ : SUCCESS in 5m 11s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/2/check/cross-horizon-py27/b3aaa49/ : SUCCESS in 8m 20s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/2/check/cross-horizon-py36/2b13ebf/ : SUCCESS in 8m 45s\n- cross-horizon-npm http://logs.openstack.org/98/662598/2/check/cross-horizon-npm/36342fa/ : SUCCESS in 5m 29s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/2/check/cross-keystone-py27/b72975f/ : SUCCESS in 24m 57s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/2/check/cross-keystone-py36/fb37874/ : SUCCESS in 22m 45s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/2/check/cross-neutron-py27/e666603/ : SUCCESS in 20m 01s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/2/check/cross-neutron-py36/cf7064d/ : SUCCESS in 24m 38s\n- cross-nova-py27 http://logs.openstack.org/98/662598/2/check/cross-nova-py27/e2286fe/ : SUCCESS in 11m 34s\n- cross-nova-py36 http://logs.openstack.org/98/662598/2/check/cross-nova-py36/4976101/ : SUCCESS in 10m 14s\n- cross-nova-functional http://logs.openstack.org/98/662598/2/check/cross-nova-functional/3322e50/ : SUCCESS in 18m 04s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/2/check/cross-osc-tox-docs/c6ff1a9/html/ : SUCCESS in 5m 12s\n- cross-swift-py27 http://logs.openstack.org/98/662598/2/check/cross-swift-py27/f53fc7a/ : SUCCESS in 9m 09s\n- cross-swift-py36 http://logs.openstack.org/98/662598/2/check/cross-swift-py36/3984065/ : SUCCESS in 8m 44s\n- requirements-integration http://logs.openstack.org/98/662598/2/check/requirements-integration/1baa5b3/ : SUCCESS in 54m 57s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/2/check/neutron-grenade-multinode/51e7b60/ : SUCCESS in 1h 03m 19s\n- tempest-full http://logs.openstack.org/98/662598/2/check/tempest-full/9a57f7e/ : FAILURE in 1h 24m 50s\n- neutron-grenade http://logs.openstack.org/98/662598/2/check/neutron-grenade/6531295/ : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":2},{"id":"00f76f85e4ae72a1d08d84137d4f65a685d694b9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-03 06:31:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5e59ba4e5ae1d7ef615035abafbf06b3b15edc78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 07:57:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/3/check/openstack-tox-docs/4db5a38/html/ : SUCCESS in 3m 04s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/3/check/openstack-tox-pypy/bc1041c/ : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/3/check/openstack-tox-pep8/c70b849/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/3/check/openstack-tox-py27/564f7dd/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/3/check/openstack-tox-py36/db47ff2/ : SUCCESS in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/98/662598/3/check/openstack-tox-validate/9f8f742/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/98/662598/3/check/requirements-tox-babel/bd66fb4/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/3/check/requirements-tox-py27-check-uc/6cc36fe/ : FAILURE in 6m 01s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/3/check/requirements-tox-py36-check-uc/948086d/ : FAILURE in 6m 08s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/3/check/cross-cinder-py27/d115edd/ : SUCCESS in 7m 57s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/3/check/cross-cinder-py36/52051a6/ : SUCCESS in 8m 19s\n- cross-glance-py27 http://logs.openstack.org/98/662598/3/check/cross-glance-py27/96c6f6e/ : SUCCESS in 4m 13s\n- cross-glance-py36 http://logs.openstack.org/98/662598/3/check/cross-glance-py36/915a17f/ : SUCCESS in 4m 24s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/3/check/cross-horizon-py27/e7cc0f4/ : SUCCESS in 11m 00s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/3/check/cross-horizon-py36/92b0353/ : SUCCESS in 8m 26s\n- cross-horizon-npm http://logs.openstack.org/98/662598/3/check/cross-horizon-npm/7659a65/ : SUCCESS in 5m 22s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/3/check/cross-keystone-py27/a6061ec/ : SUCCESS in 21m 59s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/3/check/cross-keystone-py36/7c15f56/ : SUCCESS in 30m 29s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/3/check/cross-neutron-py27/54736f9/ : SUCCESS in 23m 09s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/3/check/cross-neutron-py36/3a92c74/ : SUCCESS in 25m 47s\n- cross-nova-py27 http://logs.openstack.org/98/662598/3/check/cross-nova-py27/ebf2825/ : SUCCESS in 13m 54s\n- cross-nova-py36 http://logs.openstack.org/98/662598/3/check/cross-nova-py36/8493d07/ : SUCCESS in 11m 57s\n- cross-nova-functional http://logs.openstack.org/98/662598/3/check/cross-nova-functional/58f3450/ : SUCCESS in 17m 48s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/3/check/cross-osc-tox-docs/363a7a6/ : FAILURE in 4m 40s\n- cross-swift-py27 http://logs.openstack.org/98/662598/3/check/cross-swift-py27/d8f61ca/ : SUCCESS in 7m 22s\n- cross-swift-py36 http://logs.openstack.org/98/662598/3/check/cross-swift-py36/8134584/ : SUCCESS in 7m 39s\n- requirements-integration http://logs.openstack.org/98/662598/3/check/requirements-integration/675edc2/ : SUCCESS in 44m 50s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/3/check/neutron-grenade-multinode/6fa7f80/ : SUCCESS in 1h 07m 14s\n- tempest-full http://logs.openstack.org/98/662598/3/check/tempest-full/f9fb974/ : FAILURE in 1h 23m 52s\n- neutron-grenade http://logs.openstack.org/98/662598/3/check/neutron-grenade/646c334/ : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":3},{"id":"46ec2ebefe1e0e038bd4d0b89b6d4a32fa0af38d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-03 15:06:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8cb3c005ebd5d49d948e93ff3a91d331cf945c73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 19:28:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/4/check/openstack-tox-docs/6c34522/html/ : SUCCESS in 2m 51s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/4/check/openstack-tox-pypy/3fd1b69/ : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/4/check/openstack-tox-pep8/02b0804/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/4/check/openstack-tox-py27/658c4ec/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/4/check/openstack-tox-py36/f270704/ : SUCCESS in 3m 12s\n- openstack-tox-validate http://logs.openstack.org/98/662598/4/check/openstack-tox-validate/058ad5e/ : SUCCESS in 2m 53s\n- requirements-tox-babel http://logs.openstack.org/98/662598/4/check/requirements-tox-babel/3500cd9/ : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/4/check/requirements-tox-py27-check-uc/662671e/ : SUCCESS in 7m 31s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/4/check/requirements-tox-py36-check-uc/be799dd/ : SUCCESS in 6m 54s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/4/check/cross-cinder-py27/d4b4b9b/ : SUCCESS in 8m 40s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/4/check/cross-cinder-py36/c6962f9/ : SUCCESS in 8m 16s\n- cross-glance-py27 http://logs.openstack.org/98/662598/4/check/cross-glance-py27/436ff59/ : SUCCESS in 5m 00s\n- cross-glance-py36 http://logs.openstack.org/98/662598/4/check/cross-glance-py36/cd9486e/ : SUCCESS in 3m 57s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/4/check/cross-horizon-py27/1508b45/ : SUCCESS in 10m 33s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/4/check/cross-horizon-py36/8bdac7f/ : SUCCESS in 9m 14s\n- cross-horizon-npm http://logs.openstack.org/98/662598/4/check/cross-horizon-npm/7a273b4/ : SUCCESS in 6m 10s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/4/check/cross-keystone-py27/c32b498/ : SUCCESS in 19m 10s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/4/check/cross-keystone-py36/7ae0dc5/ : SUCCESS in 28m 13s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/4/check/cross-neutron-py27/e5eacbc/ : SUCCESS in 23m 58s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/4/check/cross-neutron-py36/f8f4de8/ : SUCCESS in 20m 40s\n- cross-nova-py27 http://logs.openstack.org/98/662598/4/check/cross-nova-py27/69a982a/ : SUCCESS in 15m 00s\n- cross-nova-py36 http://logs.openstack.org/98/662598/4/check/cross-nova-py36/1702c42/ : SUCCESS in 11m 43s\n- cross-nova-functional http://logs.openstack.org/98/662598/4/check/cross-nova-functional/bce6453/ : SUCCESS in 18m 22s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/4/check/cross-osc-tox-docs/bc02444/ : FAILURE in 6m 27s\n- cross-swift-py27 http://logs.openstack.org/98/662598/4/check/cross-swift-py27/46a2845/ : SUCCESS in 7m 37s\n- cross-swift-py36 http://logs.openstack.org/98/662598/4/check/cross-swift-py36/f7ec380/ : SUCCESS in 8m 11s\n- requirements-integration http://logs.openstack.org/98/662598/4/check/requirements-integration/68f6ca1/ : SUCCESS in 45m 29s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/4/check/neutron-grenade-multinode/b2221cd/ : SUCCESS in 1h 10m 32s\n- tempest-full http://logs.openstack.org/98/662598/4/check/tempest-full/2a3725d/ : SUCCESS in 1h 55m 51s\n- neutron-grenade http://logs.openstack.org/98/662598/4/check/neutron-grenade/cbf2163/ : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":4},{"id":"6869654455eb6e52087268fbac69f78cfb7e4f4d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-03 20:10:41.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"fb8725d350068840c80629a9b77e5b6a36cf708a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-03 20:10:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0e97d4bc0e61bc6ad55a8dbb3588e0d281f3f2ea","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 21:09:04.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6c34263cc97a3567342df62c55f198b4ec24c8fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 00:57:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/4/check/openstack-tox-docs/73a0319/html/ : SUCCESS in 2m 59s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/4/check/openstack-tox-pypy/83d12e8/ : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/4/check/openstack-tox-pep8/ec9ab4b/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/4/check/openstack-tox-py27/478fe51/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/4/check/openstack-tox-py36/2084296/ : SUCCESS in 2m 57s\n- openstack-tox-validate http://logs.openstack.org/98/662598/4/check/openstack-tox-validate/2ef7ec3/ : SUCCESS in 2m 27s\n- requirements-tox-babel http://logs.openstack.org/98/662598/4/check/requirements-tox-babel/60d9868/ : SUCCESS in 2m 44s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/4/check/requirements-tox-py27-check-uc/5f482ea/ : SUCCESS in 7m 32s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/4/check/requirements-tox-py36-check-uc/fe14054/ : SUCCESS in 6m 38s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/4/check/cross-cinder-py27/adbc2cf/ : SUCCESS in 9m 28s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/4/check/cross-cinder-py36/45aaa59/ : SUCCESS in 7m 55s\n- cross-glance-py27 http://logs.openstack.org/98/662598/4/check/cross-glance-py27/6a5197f/ : SUCCESS in 4m 07s\n- cross-glance-py36 http://logs.openstack.org/98/662598/4/check/cross-glance-py36/aee5d0e/ : SUCCESS in 4m 10s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/4/check/cross-horizon-py27/46e459a/ : SUCCESS in 10m 11s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/4/check/cross-horizon-py36/510a4ba/ : SUCCESS in 9m 11s\n- cross-horizon-npm http://logs.openstack.org/98/662598/4/check/cross-horizon-npm/2b8ba46/ : SUCCESS in 5m 47s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/4/check/cross-keystone-py27/aae5300/ : SUCCESS in 23m 14s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/4/check/cross-keystone-py36/019305f/ : SUCCESS in 23m 49s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/4/check/cross-neutron-py27/28b1bdf/ : SUCCESS in 26m 23s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/4/check/cross-neutron-py36/9fdd440/ : SUCCESS in 22m 23s\n- cross-nova-py27 http://logs.openstack.org/98/662598/4/check/cross-nova-py27/672d178/ : SUCCESS in 13m 06s\n- cross-nova-py36 http://logs.openstack.org/98/662598/4/check/cross-nova-py36/db1c0e7/ : SUCCESS in 10m 57s\n- cross-nova-functional http://logs.openstack.org/98/662598/4/check/cross-nova-functional/815246d/ : SUCCESS in 18m 50s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/4/check/cross-osc-tox-docs/7ae7c6f/ : FAILURE in 4m 45s\n- cross-swift-py27 http://logs.openstack.org/98/662598/4/check/cross-swift-py27/45bb718/ : SUCCESS in 20m 14s\n- cross-swift-py36 http://logs.openstack.org/98/662598/4/check/cross-swift-py36/34b20f4/ : SUCCESS in 10m 47s\n- requirements-integration http://logs.openstack.org/98/662598/4/check/requirements-integration/f9b7cec/ : SUCCESS in 55m 23s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/4/check/neutron-grenade-multinode/497bced/ : SUCCESS in 1h 13m 35s\n- tempest-full http://logs.openstack.org/98/662598/4/check/tempest-full/6b32d4a/ : SUCCESS in 1h 47m 47s\n- neutron-grenade http://logs.openstack.org/98/662598/4/check/neutron-grenade/395a71d/ : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":4},{"id":"83f3c344f5cff28884c5f46275cd5bc0727a40fb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-04 14:36:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1664a4269ae2d562876ba36f2cf180226cabc093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 22:07:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/5/check/openstack-tox-docs/1728748/html/ : SUCCESS in 2m 53s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/5/check/openstack-tox-pypy/55ece9c/ : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/5/check/openstack-tox-pep8/99103c8/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/5/check/openstack-tox-py27/f18c565/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/5/check/openstack-tox-py36/73fb3d1/ : SUCCESS in 4m 57s\n- openstack-tox-validate http://logs.openstack.org/98/662598/5/check/openstack-tox-validate/d1f86f9/ : SUCCESS in 2m 30s\n- requirements-tox-babel http://logs.openstack.org/98/662598/5/check/requirements-tox-babel/0c52154/ : SUCCESS in 2m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/5/check/requirements-tox-py27-check-uc/cc03f7d/ : SUCCESS in 9m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/5/check/requirements-tox-py36-check-uc/4b05713/ : SUCCESS in 6m 15s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/5/check/cross-cinder-py27/9f28bcc/ : SUCCESS in 8m 27s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/5/check/cross-cinder-py36/66a6b68/ : SUCCESS in 7m 10s\n- cross-glance-py27 http://logs.openstack.org/98/662598/5/check/cross-glance-py27/c030d3c/ : SUCCESS in 4m 10s\n- cross-glance-py36 http://logs.openstack.org/98/662598/5/check/cross-glance-py36/3d023ea/ : SUCCESS in 6m 08s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/5/check/cross-horizon-py27/ee6dc25/ : SUCCESS in 10m 15s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/5/check/cross-horizon-py36/2da53d7/ : SUCCESS in 8m 58s\n- cross-horizon-npm http://logs.openstack.org/98/662598/5/check/cross-horizon-npm/2d8bf0a/ : SUCCESS in 5m 32s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/5/check/cross-keystone-py27/33dad03/ : SUCCESS in 21m 09s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/5/check/cross-keystone-py36/2708bd6/ : SUCCESS in 23m 40s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/5/check/cross-neutron-py27/38168c2/ : SUCCESS in 21m 28s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/5/check/cross-neutron-py36/d719693/ : SUCCESS in 26m 15s\n- cross-nova-py27 http://logs.openstack.org/98/662598/5/check/cross-nova-py27/7109d84/ : SUCCESS in 11m 38s\n- cross-nova-py36 http://logs.openstack.org/98/662598/5/check/cross-nova-py36/6a10e90/ : SUCCESS in 10m 19s\n- cross-nova-functional http://logs.openstack.org/98/662598/5/check/cross-nova-functional/ccca6e2/ : SUCCESS in 16m 28s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/5/check/cross-osc-tox-docs/8c0f9c5/html/ : SUCCESS in 5m 46s\n- cross-swift-py27 http://logs.openstack.org/98/662598/5/check/cross-swift-py27/9e8a1de/ : SUCCESS in 7m 54s\n- cross-swift-py36 http://logs.openstack.org/98/662598/5/check/cross-swift-py36/8f46c37/ : SUCCESS in 7m 51s\n- requirements-integration http://logs.openstack.org/98/662598/5/check/requirements-integration/6ebe3e3/ : SUCCESS in 58m 26s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/5/check/neutron-grenade-multinode/0579ab1/ : SUCCESS in 1h 07m 50s\n- tempest-full http://logs.openstack.org/98/662598/5/check/tempest-full/ab5f335/ : SUCCESS in 1h 48m 57s\n- neutron-grenade http://logs.openstack.org/98/662598/5/check/neutron-grenade/5d935a7/ : SUCCESS in 1h 29m 47s","accounts_in_message":[],"_revision_number":5},{"id":"20b694226d312e70a836976c0f0adfea8fc56c66","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-04 22:30:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4262246ece47a649c7faf5450c1eedd89e69fa3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 22:31:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ab7dca32ee1d4436ac7fc0bcc4d4cbb40ccba7d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 01:56:22.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/5/gate/openstack-tox-docs/b50bd05/html/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/5/gate/openstack-tox-pep8/73645e9/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/5/gate/openstack-tox-py27/8b5ecf5/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/5/gate/openstack-tox-py36/e39748b/ : SUCCESS in 3m 28s\n- openstack-tox-validate http://logs.openstack.org/98/662598/5/gate/openstack-tox-validate/21c217c/ : SUCCESS in 2m 52s\n- requirements-tox-babel http://logs.openstack.org/98/662598/5/gate/requirements-tox-babel/c399098/ : SUCCESS in 4m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/5/gate/requirements-tox-py27-check-uc/54bcf8f/ : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/5/gate/requirements-tox-py36-check-uc/05bcce9/ : SUCCESS in 7m 01s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/5/gate/cross-cinder-py27/b299c31/ : SUCCESS in 8m 25s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/5/gate/cross-cinder-py36/a5fee45/ : SUCCESS in 8m 17s\n- cross-glance-py27 http://logs.openstack.org/98/662598/5/gate/cross-glance-py27/e74d0f5/ : SUCCESS in 5m 44s\n- cross-glance-py36 http://logs.openstack.org/98/662598/5/gate/cross-glance-py36/9e561ae/ : SUCCESS in 4m 32s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/5/gate/cross-horizon-py27/986618b/ : SUCCESS in 10m 14s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/5/gate/cross-horizon-py36/bb8e94a/ : SUCCESS in 8m 35s\n- cross-horizon-npm http://logs.openstack.org/98/662598/5/gate/cross-horizon-npm/f96d5ba/ : SUCCESS in 6m 53s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/5/gate/cross-keystone-py27/1a739a6/ : SUCCESS in 20m 55s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/5/gate/cross-keystone-py36/0f84356/ : SUCCESS in 20m 41s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/5/gate/cross-neutron-py27/d0655a2/ : SUCCESS in 20m 09s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/5/gate/cross-neutron-py36/3ab15ff/ : SUCCESS in 22m 01s\n- cross-nova-py27 http://logs.openstack.org/98/662598/5/gate/cross-nova-py27/20206c8/ : SUCCESS in 12m 59s\n- cross-nova-py36 http://logs.openstack.org/98/662598/5/gate/cross-nova-py36/f4c7e4a/ : SUCCESS in 11m 55s\n- cross-nova-functional http://logs.openstack.org/98/662598/5/gate/cross-nova-functional/e965ba1/ : SUCCESS in 19m 43s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/5/gate/cross-osc-tox-docs/8e4b2e3/html/ : SUCCESS in 5m 38s\n- cross-swift-py27 http://logs.openstack.org/98/662598/5/gate/cross-swift-py27/5715742/ : SUCCESS in 7m 58s\n- cross-swift-py36 http://logs.openstack.org/98/662598/5/gate/cross-swift-py36/c057252/ : SUCCESS in 10m 06s\n- requirements-integration http://logs.openstack.org/98/662598/5/gate/requirements-integration/b641aaa/ : SUCCESS in 56m 01s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/5/gate/neutron-grenade-multinode/0b85b73/ : SUCCESS in 1h 11m 08s\n- tempest-full http://logs.openstack.org/98/662598/5/gate/tempest-full/60333c4/ : TIMED_OUT in 2h 06m 36s\n- neutron-grenade http://logs.openstack.org/98/662598/5/gate/neutron-grenade/929cf7f/ : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":5},{"id":"6d9415b057f4fd24c0109530d08a7e9b83d6e014","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-05 02:53:28.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c3cda42f9f7eaf3cdcf6a91b9c51d9478bedde85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 04:53:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/5/check/openstack-tox-docs/122a548/html/ : SUCCESS in 2m 55s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/5/check/openstack-tox-pypy/7fc3ea7/ : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/5/check/openstack-tox-pep8/c15827e/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/5/check/openstack-tox-py27/6859eba/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/5/check/openstack-tox-py36/fc62752/ : SUCCESS in 3m 17s\n- openstack-tox-validate http://logs.openstack.org/98/662598/5/check/openstack-tox-validate/246fea9/ : SUCCESS in 2m 47s\n- requirements-tox-babel http://logs.openstack.org/98/662598/5/check/requirements-tox-babel/14a39fa/ : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/5/check/requirements-tox-py27-check-uc/08bc4c2/ : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/5/check/requirements-tox-py36-check-uc/493f223/ : SUCCESS in 6m 14s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/5/check/cross-cinder-py27/95255d1/ : SUCCESS in 8m 42s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/5/check/cross-cinder-py36/bff1223/ : SUCCESS in 8m 04s\n- cross-glance-py27 http://logs.openstack.org/98/662598/5/check/cross-glance-py27/2ae4621/ : SUCCESS in 4m 24s\n- cross-glance-py36 http://logs.openstack.org/98/662598/5/check/cross-glance-py36/1150655/ : SUCCESS in 4m 00s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/5/check/cross-horizon-py27/a519b52/ : SUCCESS in 10m 04s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/5/check/cross-horizon-py36/490e29c/ : SUCCESS in 7m 22s\n- cross-horizon-npm http://logs.openstack.org/98/662598/5/check/cross-horizon-npm/ce6f5ae/ : SUCCESS in 5m 49s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/5/check/cross-keystone-py27/91858e9/ : SUCCESS in 21m 47s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/5/check/cross-keystone-py36/737cf3d/ : SUCCESS in 25m 28s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/5/check/cross-neutron-py27/b947ecb/ : SUCCESS in 21m 21s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/5/check/cross-neutron-py36/34de1ab/ : SUCCESS in 22m 31s\n- cross-nova-py27 http://logs.openstack.org/98/662598/5/check/cross-nova-py27/510bb29/ : RETRY_LIMIT in 1m 34s\n- cross-nova-py36 http://logs.openstack.org/98/662598/5/check/cross-nova-py36/433c228/ : SUCCESS in 12m 19s\n- cross-nova-functional http://logs.openstack.org/98/662598/5/check/cross-nova-functional/f716cda/ : SUCCESS in 23m 28s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/5/check/cross-osc-tox-docs/23eaf10/html/ : SUCCESS in 4m 56s\n- cross-swift-py27 http://logs.openstack.org/98/662598/5/check/cross-swift-py27/3a27ac6/ : SUCCESS in 7m 56s\n- cross-swift-py36 http://logs.openstack.org/98/662598/5/check/cross-swift-py36/2449cc7/ : SUCCESS in 7m 57s\n- requirements-integration http://logs.openstack.org/98/662598/5/check/requirements-integration/22641c1/ : SUCCESS in 49m 22s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/5/check/neutron-grenade-multinode/ad63c25/ : SUCCESS in 1h 04m 37s\n- tempest-full http://logs.openstack.org/98/662598/5/check/tempest-full/e63a61c/ : SUCCESS in 1h 53m 22s\n- neutron-grenade http://logs.openstack.org/98/662598/5/check/neutron-grenade/4351d39/ : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":5},{"id":"33eb90907d5af896a481b472832fa66f277d8a75","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-05 05:01:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"28ac708ff6a06fee0e1a4efce70d3681664257c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 06:55:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/5/check/openstack-tox-docs/5bcd396/html/ : SUCCESS in 4m 12s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/5/check/openstack-tox-pypy/1035df2/ : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/5/check/openstack-tox-pep8/f2f8604/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/5/check/openstack-tox-py27/ddbe5d1/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/5/check/openstack-tox-py36/81116cd/ : SUCCESS in 3m 16s\n- openstack-tox-validate http://logs.openstack.org/98/662598/5/check/openstack-tox-validate/78b0fb1/ : SUCCESS in 3m 31s\n- requirements-tox-babel http://logs.openstack.org/98/662598/5/check/requirements-tox-babel/6ef1f0e/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/5/check/requirements-tox-py27-check-uc/c8a043f/ : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/5/check/requirements-tox-py36-check-uc/a4a8337/ : SUCCESS in 6m 41s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/5/check/cross-cinder-py27/ea09e9a/ : SUCCESS in 8m 10s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/5/check/cross-cinder-py36/1e91837/ : SUCCESS in 14m 04s\n- cross-glance-py27 http://logs.openstack.org/98/662598/5/check/cross-glance-py27/502fedf/ : SUCCESS in 3m 53s\n- cross-glance-py36 http://logs.openstack.org/98/662598/5/check/cross-glance-py36/1107638/ : SUCCESS in 4m 21s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/5/check/cross-horizon-py27/208b64f/ : SUCCESS in 10m 55s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/5/check/cross-horizon-py36/bd025d2/ : SUCCESS in 8m 56s\n- cross-horizon-npm http://logs.openstack.org/98/662598/5/check/cross-horizon-npm/a27ccce/ : SUCCESS in 7m 15s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/5/check/cross-keystone-py27/4340efc/ : SUCCESS in 21m 45s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/5/check/cross-keystone-py36/969665e/ : SUCCESS in 22m 21s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/5/check/cross-neutron-py27/47e865c/ : SUCCESS in 22m 27s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/5/check/cross-neutron-py36/c710eb1/ : SUCCESS in 21m 28s\n- cross-nova-py27 http://logs.openstack.org/98/662598/5/check/cross-nova-py27/b073596/ : SUCCESS in 11m 47s\n- cross-nova-py36 http://logs.openstack.org/98/662598/5/check/cross-nova-py36/ef38013/ : SUCCESS in 12m 28s\n- cross-nova-functional http://logs.openstack.org/98/662598/5/check/cross-nova-functional/1813572/ : SUCCESS in 19m 32s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/5/check/cross-osc-tox-docs/e7b960e/html/ : SUCCESS in 4m 51s\n- cross-swift-py27 http://logs.openstack.org/98/662598/5/check/cross-swift-py27/78b1416/ : SUCCESS in 8m 20s\n- cross-swift-py36 http://logs.openstack.org/98/662598/5/check/cross-swift-py36/2926d6a/ : SUCCESS in 8m 12s\n- requirements-integration http://logs.openstack.org/98/662598/5/check/requirements-integration/1037028/ : SUCCESS in 53m 22s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/5/check/neutron-grenade-multinode/87f2498/ : SUCCESS in 1h 08m 52s\n- tempest-full http://logs.openstack.org/98/662598/5/check/tempest-full/4feb22c/ : SUCCESS in 1h 48m 04s\n- neutron-grenade http://logs.openstack.org/98/662598/5/check/neutron-grenade/cf84b93/ : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":5},{"id":"c3ef837c9517eb57a89a11bd7d77676ef661eb5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 06:55:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1c71ae4a1a2596c7b2c5436d8ac71e91cf65f445","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 09:48:48.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/5/gate/openstack-tox-docs/233e5a7/html/ : SUCCESS in 3m 16s\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/5/gate/openstack-tox-pep8/195b0b8/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/5/gate/openstack-tox-py27/f7d6411/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/5/gate/openstack-tox-py36/c4321c5/ : SUCCESS in 4m 56s\n- openstack-tox-validate http://logs.openstack.org/98/662598/5/gate/openstack-tox-validate/c185e26/ : SUCCESS in 3m 41s\n- requirements-tox-babel http://logs.openstack.org/98/662598/5/gate/requirements-tox-babel/ff16e00/ : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/5/gate/requirements-tox-py27-check-uc/c05dde4/ : SUCCESS in 7m 51s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/5/gate/requirements-tox-py36-check-uc/a2e5385/ : SUCCESS in 7m 06s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/5/gate/cross-cinder-py27/19dc6a8/ : SUCCESS in 7m 37s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/5/gate/cross-cinder-py36/3b348ac/ : SUCCESS in 9m 39s\n- cross-glance-py27 http://logs.openstack.org/98/662598/5/gate/cross-glance-py27/62eaa9a/ : SUCCESS in 4m 58s\n- cross-glance-py36 http://logs.openstack.org/98/662598/5/gate/cross-glance-py36/760e543/ : SUCCESS in 4m 39s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/5/gate/cross-horizon-py27/782ff9a/ : SUCCESS in 10m 30s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/5/gate/cross-horizon-py36/7b3d54f/ : SUCCESS in 8m 31s\n- cross-horizon-npm http://logs.openstack.org/98/662598/5/gate/cross-horizon-npm/f33179b/ : SUCCESS in 6m 17s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/5/gate/cross-keystone-py27/f0fe7b3/ : SUCCESS in 21m 01s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/5/gate/cross-keystone-py36/b9ea7be/ : SUCCESS in 24m 19s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/5/gate/cross-neutron-py27/f07e0e2/ : SUCCESS in 30m 54s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/5/gate/cross-neutron-py36/a62f7c8/ : SUCCESS in 29m 06s\n- cross-nova-py27 http://logs.openstack.org/98/662598/5/gate/cross-nova-py27/34cb837/ : SUCCESS in 15m 57s\n- cross-nova-py36 http://logs.openstack.org/98/662598/5/gate/cross-nova-py36/f5c51cc/ : SUCCESS in 11m 17s\n- cross-nova-functional http://logs.openstack.org/98/662598/5/gate/cross-nova-functional/a6aea3c/ : SUCCESS in 21m 07s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/5/gate/cross-osc-tox-docs/568a544/html/ : SUCCESS in 5m 40s\n- cross-swift-py27 http://logs.openstack.org/98/662598/5/gate/cross-swift-py27/c9aacca/ : FAILURE in 7m 50s\n- cross-swift-py36 http://logs.openstack.org/98/662598/5/gate/cross-swift-py36/aa7395c/ : SUCCESS in 8m 50s\n- requirements-integration http://logs.openstack.org/98/662598/5/gate/requirements-integration/ffb8482/ : SUCCESS in 51m 57s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/5/gate/neutron-grenade-multinode/678d6e0/ : SUCCESS in 1h 07m 25s\n- tempest-full http://logs.openstack.org/98/662598/5/gate/tempest-full/7f30ca1/ : SUCCESS in 2h 00m 57s\n- neutron-grenade http://logs.openstack.org/98/662598/5/gate/neutron-grenade/1354422/ : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":5},{"id":"ecad54e59471ffac85f4e9d09c42deff51016cfa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-05 15:00:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ba51920e39932f736476407debde3c9a56135ecf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 20:47:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/5/check/openstack-tox-docs/c0b8c30/html/ : SUCCESS in 2m 33s\n- openstack-tox-pypy http://logs.openstack.org/98/662598/5/check/openstack-tox-pypy/b7f0c7b/ : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/5/check/openstack-tox-pep8/e07eaab/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/5/check/openstack-tox-py27/412d176/ : SUCCESS in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/5/check/openstack-tox-py36/7e7b747/ : SUCCESS in 3m 06s\n- openstack-tox-validate http://logs.openstack.org/98/662598/5/check/openstack-tox-validate/fb574dc/ : SUCCESS in 3m 19s\n- requirements-tox-babel http://logs.openstack.org/98/662598/5/check/requirements-tox-babel/b2943fa/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/5/check/requirements-tox-py27-check-uc/15ababf/ : SUCCESS in 7m 26s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/5/check/requirements-tox-py36-check-uc/81edb00/ : SUCCESS in 6m 40s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/5/check/cross-cinder-py27/d94a14e/ : SUCCESS in 8m 47s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/5/check/cross-cinder-py36/c8ae7d9/ : SUCCESS in 7m 26s\n- cross-glance-py27 http://logs.openstack.org/98/662598/5/check/cross-glance-py27/a508579/ : SUCCESS in 4m 08s\n- cross-glance-py36 http://logs.openstack.org/98/662598/5/check/cross-glance-py36/ba0faad/ : SUCCESS in 4m 21s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/5/check/cross-horizon-py27/c8fed5b/ : SUCCESS in 10m 24s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/5/check/cross-horizon-py36/2072161/ : SUCCESS in 7m 16s\n- cross-horizon-npm http://logs.openstack.org/98/662598/5/check/cross-horizon-npm/b328ef1/ : SUCCESS in 5m 46s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/5/check/cross-keystone-py27/85b91f0/ : SUCCESS in 21m 04s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/5/check/cross-keystone-py36/aab2ccc/ : SUCCESS in 25m 54s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/5/check/cross-neutron-py27/5f7fe33/ : SUCCESS in 29m 02s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/5/check/cross-neutron-py36/8125152/ : SUCCESS in 22m 56s\n- cross-nova-py27 http://logs.openstack.org/98/662598/5/check/cross-nova-py27/116a949/ : SUCCESS in 11m 24s\n- cross-nova-py36 http://logs.openstack.org/98/662598/5/check/cross-nova-py36/63d73ac/ : SUCCESS in 11m 07s\n- cross-nova-functional http://logs.openstack.org/98/662598/5/check/cross-nova-functional/774ace5/ : SUCCESS in 19m 37s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/5/check/cross-osc-tox-docs/b88faea/html/ : SUCCESS in 5m 50s\n- cross-swift-py27 http://logs.openstack.org/98/662598/5/check/cross-swift-py27/feaa63c/ : SUCCESS in 7m 35s\n- cross-swift-py36 http://logs.openstack.org/98/662598/5/check/cross-swift-py36/2e9f9b6/ : SUCCESS in 7m 18s\n- requirements-integration http://logs.openstack.org/98/662598/5/check/requirements-integration/8ba7e77/ : SUCCESS in 52m 28s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/5/check/neutron-grenade-multinode/38f59f5/ : SUCCESS in 1h 01m 45s\n- tempest-full http://logs.openstack.org/98/662598/5/check/tempest-full/4c6578e/ : SUCCESS in 1h 47m 36s\n- neutron-grenade http://logs.openstack.org/98/662598/5/check/neutron-grenade/4541e27/ : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":5},{"id":"14d7240f54d009720a41964fd761b9c86b020cbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 20:47:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7c3dcf13b6a7b53ac97666034d22877dcde386d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 22:48:20.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/662598/5/gate/openstack-tox-docs/8f90bfe/html/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/98/662598/5/gate/openstack-tox-pep8/1300460/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/98/662598/5/gate/openstack-tox-py27/9873d3c/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/98/662598/5/gate/openstack-tox-py36/b773c7f/ : SUCCESS in 4m 57s\n- openstack-tox-validate http://logs.openstack.org/98/662598/5/gate/openstack-tox-validate/9fd94a5/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/98/662598/5/gate/requirements-tox-babel/7291041/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/662598/5/gate/requirements-tox-py27-check-uc/9318e53/ : SUCCESS in 6m 32s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/662598/5/gate/requirements-tox-py36-check-uc/b386a1b/ : SUCCESS in 7m 31s\n- cross-cinder-py27 http://logs.openstack.org/98/662598/5/gate/cross-cinder-py27/63c1d00/ : SUCCESS in 9m 04s\n- cross-cinder-py36 http://logs.openstack.org/98/662598/5/gate/cross-cinder-py36/36c4613/ : SUCCESS in 8m 40s\n- cross-glance-py27 http://logs.openstack.org/98/662598/5/gate/cross-glance-py27/022144e/ : SUCCESS in 4m 38s\n- cross-glance-py36 http://logs.openstack.org/98/662598/5/gate/cross-glance-py36/88828d5/ : SUCCESS in 4m 10s\n- cross-horizon-py27 http://logs.openstack.org/98/662598/5/gate/cross-horizon-py27/65120f2/ : SUCCESS in 10m 42s\n- cross-horizon-py36 http://logs.openstack.org/98/662598/5/gate/cross-horizon-py36/76f5f91/ : SUCCESS in 9m 46s\n- cross-horizon-npm http://logs.openstack.org/98/662598/5/gate/cross-horizon-npm/7c5396a/ : SUCCESS in 6m 27s\n- cross-keystone-py27 http://logs.openstack.org/98/662598/5/gate/cross-keystone-py27/6bed640/ : SUCCESS in 33m 17s\n- cross-keystone-py36 http://logs.openstack.org/98/662598/5/gate/cross-keystone-py36/c4ec8db/ : SUCCESS in 20m 38s\n- cross-neutron-py27 http://logs.openstack.org/98/662598/5/gate/cross-neutron-py27/a6043f3/ : SUCCESS in 23m 30s\n- cross-neutron-py36 http://logs.openstack.org/98/662598/5/gate/cross-neutron-py36/e7526d6/ : SUCCESS in 22m 36s\n- cross-nova-py27 http://logs.openstack.org/98/662598/5/gate/cross-nova-py27/e0e1df0/ : SUCCESS in 15m 07s\n- cross-nova-py36 http://logs.openstack.org/98/662598/5/gate/cross-nova-py36/9b06efb/ : SUCCESS in 13m 01s\n- cross-nova-functional http://logs.openstack.org/98/662598/5/gate/cross-nova-functional/6e69dfe/ : SUCCESS in 19m 16s\n- cross-osc-tox-docs http://logs.openstack.org/98/662598/5/gate/cross-osc-tox-docs/20b3448/html/ : SUCCESS in 6m 18s\n- cross-swift-py27 http://logs.openstack.org/98/662598/5/gate/cross-swift-py27/245664d/ : SUCCESS in 11m 09s\n- cross-swift-py36 http://logs.openstack.org/98/662598/5/gate/cross-swift-py36/086ebe3/ : SUCCESS in 9m 13s\n- requirements-integration http://logs.openstack.org/98/662598/5/gate/requirements-integration/a932ff5/ : SUCCESS in 45m 35s\n- neutron-grenade-multinode http://logs.openstack.org/98/662598/5/gate/neutron-grenade-multinode/62fb438/ : SUCCESS in 1h 07m 27s\n- tempest-full http://logs.openstack.org/98/662598/5/gate/tempest-full/4ec5253/ : SUCCESS in 1h 45m 15s\n- neutron-grenade http://logs.openstack.org/98/662598/5/gate/neutron-grenade/cf6ee32/ : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":5},{"id":"37fe8c8c53c75edc653f826a88239e13745d02e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 22:48:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f6b15cf96cfb0ed2bf9a18c8803d4e980a537ee4","revisions":{"b355203a1b2964d8f69b546ce01d69586c4fc986":{"kind":"REWORK","_number":1,"created":"2019-06-01 06:13:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/662598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/662598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/662598/1"}}},"commit":{"parents":[{"commit":"6c09c23711db60a915f15772b2ae473bc8f24673","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c09c23711db60a915f15772b2ae473bc8f24673"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-01 06:13:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-01 06:13:20.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b2ae14f14326463c34bf227dc759dfa3c038bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b355203a1b2964d8f69b546ce01d69586c4fc986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b355203a1b2964d8f69b546ce01d69586c4fc986"}]},"branch":"refs/heads/master"},"8f1970aadab6f3fa5a1c8855ce91e25ea41a3dc4":{"kind":"REWORK","_number":2,"created":"2019-06-02 06:10:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/662598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/662598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/662598/2"}}},"commit":{"parents":[{"commit":"47979b8d9080b529152a985820d3615e754e9c4a","subject":"update constraint for openstacksdk to new release 0.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47979b8d9080b529152a985820d3615e754e9c4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-02 06:10:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-02 06:10:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b2ae14f14326463c34bf227dc759dfa3c038bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f1970aadab6f3fa5a1c8855ce91e25ea41a3dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f1970aadab6f3fa5a1c8855ce91e25ea41a3dc4"}]},"branch":"refs/heads/master"},"815510877d4feed7102e24ac56cb045fa85054bf":{"kind":"REWORK","_number":3,"created":"2019-06-03 06:31:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/662598/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/662598/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/662598/3"}}},"commit":{"parents":[{"commit":"47979b8d9080b529152a985820d3615e754e9c4a","subject":"update constraint for openstacksdk to new release 0.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47979b8d9080b529152a985820d3615e754e9c4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-03 06:31:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-03 06:31:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b2ae14f14326463c34bf227dc759dfa3c038bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/815510877d4feed7102e24ac56cb045fa85054bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/815510877d4feed7102e24ac56cb045fa85054bf"}]},"branch":"refs/heads/master"},"497b56a312d998f743a6c9b04b33f221389c7284":{"kind":"REWORK","_number":4,"created":"2019-06-03 15:06:04.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/98/662598/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/662598/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/662598/4"}}},"commit":{"parents":[{"commit":"47979b8d9080b529152a985820d3615e754e9c4a","subject":"update constraint for openstacksdk to new release 0.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47979b8d9080b529152a985820d3615e754e9c4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-03 06:31:13.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-06-03 15:05:44.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b2ae14f14326463c34bf227dc759dfa3c038bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/497b56a312d998f743a6c9b04b33f221389c7284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/497b56a312d998f743a6c9b04b33f221389c7284"}]},"branch":"refs/heads/master"},"f6b15cf96cfb0ed2bf9a18c8803d4e980a537ee4":{"kind":"REWORK","_number":5,"created":"2019-06-04 14:36:58.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/98/662598/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/662598/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/662598/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/662598/5"}}},"commit":{"parents":[{"commit":"47979b8d9080b529152a985820d3615e754e9c4a","subject":"update constraint for openstacksdk to new release 0.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47979b8d9080b529152a985820d3615e754e9c4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-03 06:31:13.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-06-04 14:36:42.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b2ae14f14326463c34bf227dc759dfa3c038bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6b15cf96cfb0ed2bf9a18c8803d4e980a537ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6b15cf96cfb0ed2bf9a18c8803d4e980a537ee4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
