)]}'
{"id":"openstack%2Frequirements~662608","triplet_id":"openstack%2Frequirements~master~I3edf9f9e58cb745597c2523d3e872347f14cada3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3edf9f9e58cb745597c2523d3e872347f14cada3","subject":"update constraint for openstacksdk to new release 0.29.0","status":"MERGED","created":"2019-06-01 14:55:37.000000000","updated":"2019-06-01 20:12:17.000000000","submitted":"2019-06-01 20:12:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662608-1559419937733-02bd17ee","meta_rev_id":"19f16dfb04cef9e5652676701bac494be432bc2e","_number":662608,"virtual_id_number":662608,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-01 20:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-01 18:22:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-01 18:22:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-01 18:22:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-01 20:12:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06bbfe553d16422e42d43d40d03efb7ea7a1c8f5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-01 14:55:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4747c70ac90228e00889d27e50ca35550cf70c57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 16:39:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/662608/1/check/openstack-tox-docs/e5596f3/html/ : SUCCESS in 3m 01s\n- openstack-tox-pypy http://logs.openstack.org/08/662608/1/check/openstack-tox-pypy/5b6a7c0/ : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/08/662608/1/check/openstack-tox-pep8/5264103/ : SUCCESS in 2m 35s\n- openstack-tox-py27 http://logs.openstack.org/08/662608/1/check/openstack-tox-py27/a0c74d2/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/08/662608/1/check/openstack-tox-py36/91f5887/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/08/662608/1/check/openstack-tox-validate/1255e57/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/08/662608/1/check/requirements-tox-babel/509f119/ : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/08/662608/1/check/requirements-tox-py27-check-uc/b5398f7/ : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc http://logs.openstack.org/08/662608/1/check/requirements-tox-py36-check-uc/764f5a1/ : SUCCESS in 5m 58s\n- cross-cinder-py27 http://logs.openstack.org/08/662608/1/check/cross-cinder-py27/88044d3/ : SUCCESS in 8m 41s\n- cross-cinder-py36 http://logs.openstack.org/08/662608/1/check/cross-cinder-py36/cb8bd25/ : SUCCESS in 9m 36s\n- cross-glance-py27 http://logs.openstack.org/08/662608/1/check/cross-glance-py27/ace19a4/ : SUCCESS in 3m 47s\n- cross-glance-py36 http://logs.openstack.org/08/662608/1/check/cross-glance-py36/ee5782f/ : SUCCESS in 3m 36s\n- cross-horizon-py27 http://logs.openstack.org/08/662608/1/check/cross-horizon-py27/b9cbef3/ : SUCCESS in 8m 19s\n- cross-horizon-py36 http://logs.openstack.org/08/662608/1/check/cross-horizon-py36/146bb15/ : SUCCESS in 7m 22s\n- cross-horizon-npm http://logs.openstack.org/08/662608/1/check/cross-horizon-npm/50a8d6d/ : SUCCESS in 5m 31s\n- cross-keystone-py27 http://logs.openstack.org/08/662608/1/check/cross-keystone-py27/c10d574/ : SUCCESS in 25m 43s\n- cross-keystone-py36 http://logs.openstack.org/08/662608/1/check/cross-keystone-py36/624a82c/ : SUCCESS in 26m 54s\n- cross-neutron-py27 http://logs.openstack.org/08/662608/1/check/cross-neutron-py27/b6d0825/ : SUCCESS in 20m 03s\n- cross-neutron-py36 http://logs.openstack.org/08/662608/1/check/cross-neutron-py36/411d7f9/ : SUCCESS in 23m 03s\n- cross-nova-py27 http://logs.openstack.org/08/662608/1/check/cross-nova-py27/7fa993d/ : SUCCESS in 11m 32s\n- cross-nova-py36 http://logs.openstack.org/08/662608/1/check/cross-nova-py36/167f062/ : SUCCESS in 11m 01s\n- cross-nova-functional http://logs.openstack.org/08/662608/1/check/cross-nova-functional/5532ddd/ : SUCCESS in 15m 50s\n- cross-osc-tox-docs http://logs.openstack.org/08/662608/1/check/cross-osc-tox-docs/2af2701/html/ : SUCCESS in 5m 34s\n- cross-swift-py27 http://logs.openstack.org/08/662608/1/check/cross-swift-py27/3533c3e/ : SUCCESS in 10m 28s\n- cross-swift-py36 http://logs.openstack.org/08/662608/1/check/cross-swift-py36/b9739fb/ : SUCCESS in 10m 15s\n- requirements-integration http://logs.openstack.org/08/662608/1/check/requirements-integration/217e58f/ : SUCCESS in 50m 52s\n- neutron-grenade-multinode http://logs.openstack.org/08/662608/1/check/neutron-grenade-multinode/fb8729c/ : SUCCESS in 1h 02m 25s\n- tempest-full http://logs.openstack.org/08/662608/1/check/tempest-full/39719cb/ : SUCCESS in 1h 36m 16s\n- neutron-grenade http://logs.openstack.org/08/662608/1/check/neutron-grenade/7f4074a/ : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6b29af86c8830091d75c96a9d0b1483502ec3be5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-01 18:22:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4091be887a839bae600e6d1bcd90e3d469ea40e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 18:22:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"add082613a6f17d1d08d8775045fb6295f2c892c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 20:12:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d0f04104858801f526589ae411a707fdcb279a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 20:12:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/662608/1/gate/openstack-tox-docs/4d27820/html/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/08/662608/1/gate/openstack-tox-pep8/127a07d/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/08/662608/1/gate/openstack-tox-py27/8cf9ab2/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/08/662608/1/gate/openstack-tox-py36/e31471e/ : SUCCESS in 3m 03s\n- openstack-tox-validate http://logs.openstack.org/08/662608/1/gate/openstack-tox-validate/297771b/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/08/662608/1/gate/requirements-tox-babel/eb9f1e2/ : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc http://logs.openstack.org/08/662608/1/gate/requirements-tox-py27-check-uc/106fe53/ : SUCCESS in 6m 00s\n- requirements-tox-py36-check-uc http://logs.openstack.org/08/662608/1/gate/requirements-tox-py36-check-uc/94d6a59/ : SUCCESS in 5m 24s\n- cross-cinder-py27 http://logs.openstack.org/08/662608/1/gate/cross-cinder-py27/d555d6c/ : SUCCESS in 8m 30s\n- cross-cinder-py36 http://logs.openstack.org/08/662608/1/gate/cross-cinder-py36/ded524b/ : SUCCESS in 7m 16s\n- cross-glance-py27 http://logs.openstack.org/08/662608/1/gate/cross-glance-py27/7e2c5d6/ : SUCCESS in 4m 01s\n- cross-glance-py36 http://logs.openstack.org/08/662608/1/gate/cross-glance-py36/e7becb8/ : SUCCESS in 3m 48s\n- cross-horizon-py27 http://logs.openstack.org/08/662608/1/gate/cross-horizon-py27/abe62cf/ : SUCCESS in 9m 00s\n- cross-horizon-py36 http://logs.openstack.org/08/662608/1/gate/cross-horizon-py36/0440941/ : SUCCESS in 9m 37s\n- cross-horizon-npm http://logs.openstack.org/08/662608/1/gate/cross-horizon-npm/1f21764/ : SUCCESS in 5m 42s\n- cross-keystone-py27 http://logs.openstack.org/08/662608/1/gate/cross-keystone-py27/f25e054/ : SUCCESS in 21m 24s\n- cross-keystone-py36 http://logs.openstack.org/08/662608/1/gate/cross-keystone-py36/2ccbc67/ : SUCCESS in 19m 50s\n- cross-neutron-py27 http://logs.openstack.org/08/662608/1/gate/cross-neutron-py27/1cbe9dc/ : SUCCESS in 22m 15s\n- cross-neutron-py36 http://logs.openstack.org/08/662608/1/gate/cross-neutron-py36/5c7c3ce/ : SUCCESS in 21m 45s\n- cross-nova-py27 http://logs.openstack.org/08/662608/1/gate/cross-nova-py27/76ea22c/ : SUCCESS in 15m 34s\n- cross-nova-py36 http://logs.openstack.org/08/662608/1/gate/cross-nova-py36/923b366/ : SUCCESS in 11m 32s\n- cross-nova-functional http://logs.openstack.org/08/662608/1/gate/cross-nova-functional/b7523a3/ : SUCCESS in 15m 55s\n- cross-osc-tox-docs http://logs.openstack.org/08/662608/1/gate/cross-osc-tox-docs/edab476/html/ : SUCCESS in 4m 48s\n- cross-swift-py27 http://logs.openstack.org/08/662608/1/gate/cross-swift-py27/dc2245b/ : SUCCESS in 9m 54s\n- cross-swift-py36 http://logs.openstack.org/08/662608/1/gate/cross-swift-py36/ba625e8/ : SUCCESS in 8m 44s\n- requirements-integration http://logs.openstack.org/08/662608/1/gate/requirements-integration/d0700db/ : SUCCESS in 41m 48s\n- neutron-grenade-multinode http://logs.openstack.org/08/662608/1/gate/neutron-grenade-multinode/d0ee18e/ : SUCCESS in 54m 32s\n- tempest-full http://logs.openstack.org/08/662608/1/gate/tempest-full/872b462/ : SUCCESS in 1h 43m 03s\n- neutron-grenade http://logs.openstack.org/08/662608/1/gate/neutron-grenade/e3b8582/ : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47979b8d9080b529152a985820d3615e754e9c4a","revisions":{"47979b8d9080b529152a985820d3615e754e9c4a":{"kind":"REWORK","_number":1,"created":"2019-06-01 14:55:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/662608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/662608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/662608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/662608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/662608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/662608/1"}}},"commit":{"parents":[{"commit":"6c09c23711db60a915f15772b2ae473bc8f24673","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c09c23711db60a915f15772b2ae473bc8f24673"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-01 14:55:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-01 14:55:35.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.29.0","message":"update constraint for openstacksdk to new release 0.29.0\n\nChange-Id: I3edf9f9e58cb745597c2523d3e872347f14cada3\nmeta:version: 0.29.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: Ie5634e1ec36cee418770c51b4b8c1efe40e3f104\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47979b8d9080b529152a985820d3615e754e9c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47979b8d9080b529152a985820d3615e754e9c4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
