)]}'
{"id":"openstack%2Frequirements~662805","triplet_id":"openstack%2Frequirements~master~I34b72e8b7d96b02c2e0143e58445d7247503694b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I34b72e8b7d96b02c2e0143e58445d7247503694b","subject":"update constraint for os-traits to new release 0.13.0","status":"MERGED","created":"2019-06-03 16:57:29.000000000","updated":"2019-06-05 04:04:25.000000000","submitted":"2019-06-05 04:04:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662805-1559707465344-03aab1a2","meta_rev_id":"ebf1d2246029ba9dbb91e799094e53e029e7f0f5","_number":662805,"virtual_id_number":662805,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-05 04:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-03 21:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-04 22:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-06-03 17:26:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-06-03 21:48:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-04 22:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"date":"2019-06-03 17:26:36.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 17:26:30.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-03 17:26:36.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-03 21:10:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-03 21:48:19.000000000","updated_by":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"reviewer":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"state":"REVIEWER"},{"updated":"2019-06-04 22:31:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-05 04:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dcce84ea1a3e91f5598b4f13b37c02e7db60b430","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-03 16:57:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecdf5293839a85f7926b29a79867a77ed61ce8e6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-03 17:26:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a60cd315d3d66c37627b24d07b985a0602a726da","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 21:10:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"085bc16f67108b85207df18eea83551dfead3e2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:42:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/662805/1/check/openstack-tox-docs/2068afe/html/ : SUCCESS in 2m 52s\n- openstack-tox-pypy http://logs.openstack.org/05/662805/1/check/openstack-tox-pypy/cd0a6f4/ : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/05/662805/1/check/openstack-tox-pep8/ac0f1d3/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/05/662805/1/check/openstack-tox-py27/3283131/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/05/662805/1/check/openstack-tox-py36/84f8ad5/ : SUCCESS in 3m 08s\n- openstack-tox-validate http://logs.openstack.org/05/662805/1/check/openstack-tox-validate/870c3d9/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/05/662805/1/check/requirements-tox-babel/4bdff5b/ : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc http://logs.openstack.org/05/662805/1/check/requirements-tox-py27-check-uc/fbe8f7b/ : SUCCESS in 7m 22s\n- requirements-tox-py36-check-uc http://logs.openstack.org/05/662805/1/check/requirements-tox-py36-check-uc/bcb8d1d/ : SUCCESS in 8m 16s\n- cross-cinder-py27 http://logs.openstack.org/05/662805/1/check/cross-cinder-py27/7c6bc2f/ : SUCCESS in 8m 38s\n- cross-cinder-py36 http://logs.openstack.org/05/662805/1/check/cross-cinder-py36/9355bee/ : SUCCESS in 8m 18s\n- cross-glance-py27 http://logs.openstack.org/05/662805/1/check/cross-glance-py27/a9f208e/ : SUCCESS in 5m 31s\n- cross-glance-py36 http://logs.openstack.org/05/662805/1/check/cross-glance-py36/1bf7933/ : SUCCESS in 4m 58s\n- cross-horizon-py27 http://logs.openstack.org/05/662805/1/check/cross-horizon-py27/a46b585/ : SUCCESS in 10m 27s\n- cross-horizon-py36 http://logs.openstack.org/05/662805/1/check/cross-horizon-py36/23dcd5f/ : SUCCESS in 7m 55s\n- cross-horizon-npm http://logs.openstack.org/05/662805/1/check/cross-horizon-npm/145dd9f/ : SUCCESS in 9m 43s\n- cross-keystone-py27 http://logs.openstack.org/05/662805/1/check/cross-keystone-py27/384ec11/ : SUCCESS in 23m 27s\n- cross-keystone-py36 http://logs.openstack.org/05/662805/1/check/cross-keystone-py36/a1347be/ : SUCCESS in 23m 49s\n- cross-neutron-py27 http://logs.openstack.org/05/662805/1/check/cross-neutron-py27/b561c0a/ : SUCCESS in 22m 09s\n- cross-neutron-py36 http://logs.openstack.org/05/662805/1/check/cross-neutron-py36/e9c23ff/ : SUCCESS in 22m 14s\n- cross-nova-py27 http://logs.openstack.org/05/662805/1/check/cross-nova-py27/9788417/ : SUCCESS in 11m 32s\n- cross-nova-py36 http://logs.openstack.org/05/662805/1/check/cross-nova-py36/1fa2cc5/ : SUCCESS in 10m 29s\n- cross-nova-functional http://logs.openstack.org/05/662805/1/check/cross-nova-functional/5005586/ : SUCCESS in 17m 31s\n- cross-osc-tox-docs http://logs.openstack.org/05/662805/1/check/cross-osc-tox-docs/d323273/html/ : SUCCESS in 4m 48s\n- cross-swift-py27 http://logs.openstack.org/05/662805/1/check/cross-swift-py27/312415f/ : SUCCESS in 7m 15s\n- cross-swift-py36 http://logs.openstack.org/05/662805/1/check/cross-swift-py36/0b960ec/ : SUCCESS in 10m 39s\n- requirements-integration http://logs.openstack.org/05/662805/1/check/requirements-integration/f74b088/ : SUCCESS in 53m 29s\n- neutron-grenade-multinode http://logs.openstack.org/05/662805/1/check/neutron-grenade-multinode/323b3f6/ : SUCCESS in 1h 14m 42s\n- tempest-full http://logs.openstack.org/05/662805/1/check/tempest-full/52091c3/ : TIMED_OUT in 2h 07m 05s\n- neutron-grenade http://logs.openstack.org/05/662805/1/check/neutron-grenade/9e02ade/ : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f572c2a7c5f1abee6cd8f52a5a24a607e706b537","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-06-03 21:47:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1e2a4e21f3af554baf067ee134f5c708dd03c8d8","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-06-03 21:48:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"56ffc7c6fe79acaa293932b8a891ab64b0ab7ee0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 02:57:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/662805/1/check/openstack-tox-docs/f223b89/html/ : SUCCESS in 3m 02s\n- openstack-tox-pypy http://logs.openstack.org/05/662805/1/check/openstack-tox-pypy/03d6094/ : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/05/662805/1/check/openstack-tox-pep8/7f2dd7f/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/05/662805/1/check/openstack-tox-py27/c4b7bec/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/05/662805/1/check/openstack-tox-py36/64385e5/ : SUCCESS in 3m 29s\n- openstack-tox-validate http://logs.openstack.org/05/662805/1/check/openstack-tox-validate/e03ef65/ : SUCCESS in 4m 59s\n- requirements-tox-babel http://logs.openstack.org/05/662805/1/check/requirements-tox-babel/284418a/ : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/05/662805/1/check/requirements-tox-py27-check-uc/9fa1b31/ : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc http://logs.openstack.org/05/662805/1/check/requirements-tox-py36-check-uc/1b29c04/ : SUCCESS in 6m 02s\n- cross-cinder-py27 http://logs.openstack.org/05/662805/1/check/cross-cinder-py27/b9fd5a6/ : SUCCESS in 8m 44s\n- cross-cinder-py36 http://logs.openstack.org/05/662805/1/check/cross-cinder-py36/3730cff/ : SUCCESS in 7m 56s\n- cross-glance-py27 http://logs.openstack.org/05/662805/1/check/cross-glance-py27/aa0f2bc/ : SUCCESS in 4m 52s\n- cross-glance-py36 http://logs.openstack.org/05/662805/1/check/cross-glance-py36/b5fb3bc/ : SUCCESS in 4m 05s\n- cross-horizon-py27 http://logs.openstack.org/05/662805/1/check/cross-horizon-py27/16a1cb9/ : SUCCESS in 9m 34s\n- cross-horizon-py36 http://logs.openstack.org/05/662805/1/check/cross-horizon-py36/8ece315/ : SUCCESS in 9m 20s\n- cross-horizon-npm http://logs.openstack.org/05/662805/1/check/cross-horizon-npm/0fd2c4a/ : SUCCESS in 6m 02s\n- cross-keystone-py27 http://logs.openstack.org/05/662805/1/check/cross-keystone-py27/f358a72/ : SUCCESS in 21m 31s\n- cross-keystone-py36 http://logs.openstack.org/05/662805/1/check/cross-keystone-py36/bba56e8/ : SUCCESS in 24m 15s\n- cross-neutron-py27 http://logs.openstack.org/05/662805/1/check/cross-neutron-py27/d3b78c7/ : SUCCESS in 27m 11s\n- cross-neutron-py36 http://logs.openstack.org/05/662805/1/check/cross-neutron-py36/12f58d2/ : SUCCESS in 22m 34s\n- cross-nova-py27 http://logs.openstack.org/05/662805/1/check/cross-nova-py27/98e48c1/ : SUCCESS in 11m 59s\n- cross-nova-py36 http://logs.openstack.org/05/662805/1/check/cross-nova-py36/971e4fd/ : SUCCESS in 12m 37s\n- cross-nova-functional http://logs.openstack.org/05/662805/1/check/cross-nova-functional/10fe677/ : SUCCESS in 19m 31s\n- cross-osc-tox-docs http://logs.openstack.org/05/662805/1/check/cross-osc-tox-docs/d795470/html/ : SUCCESS in 6m 02s\n- cross-swift-py27 http://logs.openstack.org/05/662805/1/check/cross-swift-py27/38b2ef8/ : SUCCESS in 7m 45s\n- cross-swift-py36 http://logs.openstack.org/05/662805/1/check/cross-swift-py36/6e5fe01/ : SUCCESS in 8m 28s\n- requirements-integration http://logs.openstack.org/05/662805/1/check/requirements-integration/78fd667/ : SUCCESS in 52m 46s\n- neutron-grenade-multinode http://logs.openstack.org/05/662805/1/check/neutron-grenade-multinode/cb1c164/ : SUCCESS in 1h 07m 46s\n- tempest-full http://logs.openstack.org/05/662805/1/check/tempest-full/04d41d8/ : SUCCESS in 1h 57m 19s\n- neutron-grenade http://logs.openstack.org/05/662805/1/check/neutron-grenade/7790dd1/ : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ad7cf1e181956773ddc28b55a46f9569bf414dea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-04 22:31:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"040836d2d73a9dd055ad78f645d609680e443bc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 22:31:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6a6ff4d68831b6c24ce330b4af06625f548a0ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 04:04:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/662805/1/gate/openstack-tox-docs/d0b245d/html/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/05/662805/1/gate/openstack-tox-pep8/a5c17c0/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/05/662805/1/gate/openstack-tox-py27/314090f/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/05/662805/1/gate/openstack-tox-py36/d178e80/ : SUCCESS in 5m 01s\n- openstack-tox-validate http://logs.openstack.org/05/662805/1/gate/openstack-tox-validate/c7d8c04/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/05/662805/1/gate/requirements-tox-babel/8b780d8/ : SUCCESS in 4m 39s\n- requirements-tox-py27-check-uc http://logs.openstack.org/05/662805/1/gate/requirements-tox-py27-check-uc/7dc5f22/ : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/05/662805/1/gate/requirements-tox-py36-check-uc/6c5685b/ : SUCCESS in 6m 02s\n- cross-cinder-py27 http://logs.openstack.org/05/662805/1/gate/cross-cinder-py27/fcc49c2/ : SUCCESS in 8m 35s\n- cross-cinder-py36 http://logs.openstack.org/05/662805/1/gate/cross-cinder-py36/ba99136/ : SUCCESS in 8m 16s\n- cross-glance-py27 http://logs.openstack.org/05/662805/1/gate/cross-glance-py27/d58d75f/ : SUCCESS in 4m 50s\n- cross-glance-py36 http://logs.openstack.org/05/662805/1/gate/cross-glance-py36/58b5003/ : SUCCESS in 4m 23s\n- cross-horizon-py27 http://logs.openstack.org/05/662805/1/gate/cross-horizon-py27/8619e30/ : SUCCESS in 9m 30s\n- cross-horizon-py36 http://logs.openstack.org/05/662805/1/gate/cross-horizon-py36/c4505c6/ : SUCCESS in 9m 17s\n- cross-horizon-npm http://logs.openstack.org/05/662805/1/gate/cross-horizon-npm/25a54f5/ : SUCCESS in 5m 49s\n- cross-keystone-py27 http://logs.openstack.org/05/662805/1/gate/cross-keystone-py27/311527a/ : SUCCESS in 20m 03s\n- cross-keystone-py36 http://logs.openstack.org/05/662805/1/gate/cross-keystone-py36/219fa94/ : SUCCESS in 25m 57s\n- cross-neutron-py27 http://logs.openstack.org/05/662805/1/gate/cross-neutron-py27/bba8850/ : SUCCESS in 23m 19s\n- cross-neutron-py36 http://logs.openstack.org/05/662805/1/gate/cross-neutron-py36/9dcc717/ : SUCCESS in 22m 49s\n- cross-nova-py27 http://logs.openstack.org/05/662805/1/gate/cross-nova-py27/bbcec4f/ : SUCCESS in 12m 43s\n- cross-nova-py36 http://logs.openstack.org/05/662805/1/gate/cross-nova-py36/67ec397/ : SUCCESS in 12m 32s\n- cross-nova-functional http://logs.openstack.org/05/662805/1/gate/cross-nova-functional/095fa9d/ : SUCCESS in 16m 54s\n- cross-osc-tox-docs http://logs.openstack.org/05/662805/1/gate/cross-osc-tox-docs/9349ab5/html/ : SUCCESS in 4m 53s\n- cross-swift-py27 http://logs.openstack.org/05/662805/1/gate/cross-swift-py27/bfac3ab/ : SUCCESS in 8m 50s\n- cross-swift-py36 http://logs.openstack.org/05/662805/1/gate/cross-swift-py36/1e8eaab/ : SUCCESS in 12m 25s\n- requirements-integration http://logs.openstack.org/05/662805/1/gate/requirements-integration/259312e/ : SUCCESS in 49m 15s\n- neutron-grenade-multinode http://logs.openstack.org/05/662805/1/gate/neutron-grenade-multinode/dd7c563/ : SUCCESS in 1h 07m 40s\n- tempest-full http://logs.openstack.org/05/662805/1/gate/tempest-full/df51924/ : SUCCESS in 2h 06m 06s\n- neutron-grenade http://logs.openstack.org/05/662805/1/gate/neutron-grenade/f7f4317/ : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d4f1745bbf76f6c76b9a08f996a50fbbcc046707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 04:04:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1e8d322f367088712b436665d4e8cee063fde68","revisions":{"b1e8d322f367088712b436665d4e8cee063fde68":{"kind":"REWORK","_number":1,"created":"2019-06-03 16:57:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/662805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/662805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/662805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/662805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/662805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/662805/1"}}},"commit":{"parents":[{"commit":"47979b8d9080b529152a985820d3615e754e9c4a","subject":"update constraint for openstacksdk to new release 0.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47979b8d9080b529152a985820d3615e754e9c4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-03 16:57:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-03 16:57:27.000000000","tz":0},"subject":"update constraint for os-traits to new release 0.13.0","message":"update constraint for os-traits to new release 0.13.0\n\nChange-Id: I34b72e8b7d96b02c2e0143e58445d7247503694b\nmeta:version: 0.13.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Chris Dent \u003ccdent@anticdent.org\u003e\nmeta:release:Commit: Chris Dent \u003ccdent@anticdent.org\u003e\nmeta:release:Change-Id: Ib9aa0dc28b901cd7941a696d0093cb0b10ce4522\nmeta:release:Code-Review+1: Adam Spiers \u003caspiers@suse.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1e8d322f367088712b436665d4e8cee063fde68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1e8d322f367088712b436665d4e8cee063fde68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
