)]}'
{"id":"openstack%2Frequirements~663516","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I2c51fc8ffa45bc31eabbed5b5bf5dc4fdd2a4022","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I2c51fc8ffa45bc31eabbed5b5bf5dc4fdd2a4022","subject":"update constraint for ironic-lib to new release 2.12.3","status":"MERGED","created":"2019-06-06 06:45:29.000000000","updated":"2019-06-10 19:48:16.000000000","submitted":"2019-06-10 19:48:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663516-1560196096104-d9eb55b5","meta_rev_id":"cebdd234eff96862e3b9cf57e50934661f13c02a","_number":663516,"virtual_id_number":663516,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-10 19:48:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-06 20:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-10 15:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-06-10 15:16:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 20:56:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-10 15:16:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-10 19:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff4fca4a118a2ab8eb46f501ff2feffcc7b48767","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-06 06:45:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3bed219cc817484c977e416c479c231d435a162","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:15:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/16/663516/1/check/openstack-tox-pypy/888598a/ : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/16/663516/1/check/openstack-tox-pep8/9e76e2f/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/16/663516/1/check/openstack-tox-py27/1e51a08/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/16/663516/1/check/openstack-tox-py35/8a07f6c/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/16/663516/1/check/openstack-tox-validate/3b31e7d/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/16/663516/1/check/requirements-tox-babel/43c41e9/ : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/16/663516/1/check/requirements-tox-py27-check-uc/14055dc/ : SUCCESS in 6m 45s\n- requirements-tox-py35-check-uc http://logs.openstack.org/16/663516/1/check/requirements-tox-py35-check-uc/1b05532/ : SUCCESS in 6m 52s\n- cross-cinder-py27 http://logs.openstack.org/16/663516/1/check/cross-cinder-py27/f494a5d/ : SUCCESS in 8m 12s\n- cross-cinder-py35 http://logs.openstack.org/16/663516/1/check/cross-cinder-py35/97db78c/ : FAILURE in 11m 31s\n- cross-glance-py27 http://logs.openstack.org/16/663516/1/check/cross-glance-py27/e1b4795/ : SUCCESS in 4m 47s\n- cross-glance-py35 http://logs.openstack.org/16/663516/1/check/cross-glance-py35/65d8bbe/ : SUCCESS in 4m 56s\n- cross-horizon-py27 http://logs.openstack.org/16/663516/1/check/cross-horizon-py27/5c24c28/ : SUCCESS in 11m 51s\n- cross-horizon-py35 http://logs.openstack.org/16/663516/1/check/cross-horizon-py35/09d3839/ : SUCCESS in 10m 44s\n- cross-keystone-py27 http://logs.openstack.org/16/663516/1/check/cross-keystone-py27/d9a87df/ : TIMED_OUT in 40m 49s\n- cross-keystone-py35 http://logs.openstack.org/16/663516/1/check/cross-keystone-py35/fea3603/ : SUCCESS in 37m 57s\n- cross-neutron-py27 http://logs.openstack.org/16/663516/1/check/cross-neutron-py27/ec693a9/ : SUCCESS in 23m 35s\n- cross-neutron-py35 http://logs.openstack.org/16/663516/1/check/cross-neutron-py35/30f3cab/ : SUCCESS in 26m 26s\n- cross-nova-py27 http://logs.openstack.org/16/663516/1/check/cross-nova-py27/432663d/ : SUCCESS in 12m 42s\n- cross-nova-py35 http://logs.openstack.org/16/663516/1/check/cross-nova-py35/6b00bf3/ : SUCCESS in 17m 59s\n- cross-nova-functional http://logs.openstack.org/16/663516/1/check/cross-nova-functional/03a6f7b/ : SUCCESS in 17m 27s\n- cross-swift-py27 http://logs.openstack.org/16/663516/1/check/cross-swift-py27/1cee2f5/ : SUCCESS in 7m 10s\n- cross-swift-py35 http://logs.openstack.org/16/663516/1/check/cross-swift-py35/45514ae/ : SUCCESS in 3m 58s\n- requirements-integration http://logs.openstack.org/16/663516/1/check/requirements-integration/fd3860a/ : SUCCESS in 51m 07s\n- neutron-grenade-multinode http://logs.openstack.org/16/663516/1/check/neutron-grenade-multinode/f9183fc/ : SUCCESS in 1h 05m 03s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/16/663516/1/check/legacy-tempest-dsvm-neutron-full/2945698/ : SUCCESS in 1h 33m 18s\n- neutron-grenade http://logs.openstack.org/16/663516/1/check/neutron-grenade/baac0da/ : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5d82cc5a8ab1f5c0a1ecdc1cc2b724dbeb1d7d48","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-06 14:32:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be9808f944a759fd13f10e08c6234c1be07432e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 19:55:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/16/663516/1/check/openstack-tox-pypy/42d067d/ : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/16/663516/1/check/openstack-tox-pep8/3ffb210/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/16/663516/1/check/openstack-tox-py27/2892ada/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/16/663516/1/check/openstack-tox-py35/a5aa962/ : SUCCESS in 3m 39s\n- openstack-tox-validate http://logs.openstack.org/16/663516/1/check/openstack-tox-validate/3ab8127/ : SUCCESS in 2m 55s\n- requirements-tox-babel http://logs.openstack.org/16/663516/1/check/requirements-tox-babel/8144acf/ : SUCCESS in 3m 32s\n- requirements-tox-py27-check-uc http://logs.openstack.org/16/663516/1/check/requirements-tox-py27-check-uc/5ddc24b/ : SUCCESS in 6m 00s\n- requirements-tox-py35-check-uc http://logs.openstack.org/16/663516/1/check/requirements-tox-py35-check-uc/4670960/ : SUCCESS in 5m 55s\n- cross-cinder-py27 http://logs.openstack.org/16/663516/1/check/cross-cinder-py27/8b83011/ : SUCCESS in 9m 17s\n- cross-cinder-py35 http://logs.openstack.org/16/663516/1/check/cross-cinder-py35/f78efb7/ : SUCCESS in 8m 49s\n- cross-glance-py27 http://logs.openstack.org/16/663516/1/check/cross-glance-py27/3b14c7b/ : SUCCESS in 4m 55s\n- cross-glance-py35 http://logs.openstack.org/16/663516/1/check/cross-glance-py35/bbf3372/ : SUCCESS in 5m 16s\n- cross-horizon-py27 http://logs.openstack.org/16/663516/1/check/cross-horizon-py27/7189384/ : SUCCESS in 12m 11s\n- cross-horizon-py35 http://logs.openstack.org/16/663516/1/check/cross-horizon-py35/5a2dcfc/ : SUCCESS in 9m 50s\n- cross-keystone-py27 http://logs.openstack.org/16/663516/1/check/cross-keystone-py27/ba4ba53/ : SUCCESS in 22m 52s\n- cross-keystone-py35 http://logs.openstack.org/16/663516/1/check/cross-keystone-py35/71a763f/ : TIMED_OUT in 41m 10s\n- cross-neutron-py27 http://logs.openstack.org/16/663516/1/check/cross-neutron-py27/2a3f14e/ : SUCCESS in 20m 52s\n- cross-neutron-py35 http://logs.openstack.org/16/663516/1/check/cross-neutron-py35/2d1b75e/ : SUCCESS in 31m 51s\n- cross-nova-py27 http://logs.openstack.org/16/663516/1/check/cross-nova-py27/28e7a0f/ : SUCCESS in 11m 16s\n- cross-nova-py35 http://logs.openstack.org/16/663516/1/check/cross-nova-py35/df235bc/ : SUCCESS in 12m 28s\n- cross-nova-functional http://logs.openstack.org/16/663516/1/check/cross-nova-functional/14783c6/ : SUCCESS in 18m 44s\n- cross-swift-py27 http://logs.openstack.org/16/663516/1/check/cross-swift-py27/7ae976b/ : SUCCESS in 5m 38s\n- cross-swift-py35 http://logs.openstack.org/16/663516/1/check/cross-swift-py35/500d7c6/ : SUCCESS in 3m 08s\n- requirements-integration http://logs.openstack.org/16/663516/1/check/requirements-integration/18269fb/ : SUCCESS in 56m 35s\n- neutron-grenade-multinode http://logs.openstack.org/16/663516/1/check/neutron-grenade-multinode/bd6bf83/ : SUCCESS in 1h 04m 25s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/16/663516/1/check/legacy-tempest-dsvm-neutron-full/2bc7745/ : SUCCESS in 1h 20m 07s\n- neutron-grenade http://logs.openstack.org/16/663516/1/check/neutron-grenade/7df9112/ : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":1},{"id":"854dedec98538edfa7a98e8d7c49d304ef9bdc80","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-06 20:56:09.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2d552e0514b204417af47a8bb3bcf0985695e76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/16/663516/1/check/openstack-tox-pypy/f051514/ : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/16/663516/1/check/openstack-tox-pep8/50daf8f/ : SUCCESS in 2m 38s\n- openstack-tox-py27 http://logs.openstack.org/16/663516/1/check/openstack-tox-py27/7a8c2e6/ : SUCCESS in 6m 15s\n- openstack-tox-py35 http://logs.openstack.org/16/663516/1/check/openstack-tox-py35/8ecc7e8/ : SUCCESS in 4m 29s\n- openstack-tox-validate http://logs.openstack.org/16/663516/1/check/openstack-tox-validate/e11e1d5/ : SUCCESS in 6m 37s\n- requirements-tox-babel http://logs.openstack.org/16/663516/1/check/requirements-tox-babel/62c1163/ : SUCCESS in 5m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/16/663516/1/check/requirements-tox-py27-check-uc/602ff9e/ : SUCCESS in 5m 46s\n- requirements-tox-py35-check-uc http://logs.openstack.org/16/663516/1/check/requirements-tox-py35-check-uc/fc864dd/ : SUCCESS in 6m 17s\n- cross-cinder-py27 http://logs.openstack.org/16/663516/1/check/cross-cinder-py27/c5af9b5/ : SUCCESS in 8m 25s\n- cross-cinder-py35 http://logs.openstack.org/16/663516/1/check/cross-cinder-py35/bd4efe9/ : SUCCESS in 8m 37s\n- cross-glance-py27 http://logs.openstack.org/16/663516/1/check/cross-glance-py27/36b7b18/ : SUCCESS in 5m 11s\n- cross-glance-py35 http://logs.openstack.org/16/663516/1/check/cross-glance-py35/c0f435f/ : SUCCESS in 5m 37s\n- cross-horizon-py27 http://logs.openstack.org/16/663516/1/check/cross-horizon-py27/3d34438/ : SUCCESS in 12m 46s\n- cross-horizon-py35 http://logs.openstack.org/16/663516/1/check/cross-horizon-py35/d3b9498/ : SUCCESS in 9m 18s\n- cross-keystone-py27 http://logs.openstack.org/16/663516/1/check/cross-keystone-py27/705657c/ : SUCCESS in 12m 06s\n- cross-keystone-py35 http://logs.openstack.org/16/663516/1/check/cross-keystone-py35/0ddbcc8/ : SUCCESS in 12m 10s\n- cross-neutron-py27 http://logs.openstack.org/16/663516/1/check/cross-neutron-py27/623c54b/ : SUCCESS in 23m 17s\n- cross-neutron-py35 http://logs.openstack.org/16/663516/1/check/cross-neutron-py35/2854f55/ : SUCCESS in 39m 34s\n- cross-nova-py27 http://logs.openstack.org/16/663516/1/check/cross-nova-py27/2bd456a/ : SUCCESS in 12m 30s\n- cross-nova-py35 http://logs.openstack.org/16/663516/1/check/cross-nova-py35/05c1390/ : SUCCESS in 12m 57s\n- cross-nova-functional http://logs.openstack.org/16/663516/1/check/cross-nova-functional/9deac9c/ : SUCCESS in 19m 18s\n- cross-swift-py27 http://logs.openstack.org/16/663516/1/check/cross-swift-py27/a3a791a/ : SUCCESS in 5m 48s\n- cross-swift-py35 http://logs.openstack.org/16/663516/1/check/cross-swift-py35/6bea79c/ : SUCCESS in 3m 13s\n- requirements-integration http://logs.openstack.org/16/663516/1/check/requirements-integration/7d82438/ : SUCCESS in 51m 04s\n- neutron-grenade-multinode http://logs.openstack.org/16/663516/1/check/neutron-grenade-multinode/91cd5a7/ : SUCCESS in 1h 10m 48s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/16/663516/1/check/legacy-tempest-dsvm-neutron-full/3e1fc20/ : SUCCESS in 1h 23m 33s\n- neutron-grenade http://logs.openstack.org/16/663516/1/check/neutron-grenade/f4d6748/ : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2442f4d34eefc533ae0b732b9751adb51bd30ed1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-10 15:16:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f822ed69a58e230668f6761341695805053ab89a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 15:17:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c92f3cb429e0e7d5b45bc7be8dc95c21ba43132","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:48:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/16/663516/1/gate/openstack-tox-pep8/891b03d/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/16/663516/1/gate/openstack-tox-py27/4c16531/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/16/663516/1/gate/openstack-tox-py35/5a05ad3/ : SUCCESS in 4m 51s\n- openstack-tox-validate http://logs.openstack.org/16/663516/1/gate/openstack-tox-validate/4653500/ : SUCCESS in 3m 37s\n- requirements-tox-babel http://logs.openstack.org/16/663516/1/gate/requirements-tox-babel/17013cc/ : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc http://logs.openstack.org/16/663516/1/gate/requirements-tox-py27-check-uc/f351341/ : SUCCESS in 6m 26s\n- requirements-tox-py35-check-uc http://logs.openstack.org/16/663516/1/gate/requirements-tox-py35-check-uc/57ad346/ : SUCCESS in 7m 01s\n- cross-cinder-py27 http://logs.openstack.org/16/663516/1/gate/cross-cinder-py27/39c0f15/ : SUCCESS in 9m 13s\n- cross-cinder-py35 http://logs.openstack.org/16/663516/1/gate/cross-cinder-py35/da67350/ : SUCCESS in 9m 33s\n- cross-glance-py27 http://logs.openstack.org/16/663516/1/gate/cross-glance-py27/1b0d90d/ : SUCCESS in 5m 14s\n- cross-glance-py35 http://logs.openstack.org/16/663516/1/gate/cross-glance-py35/27636ef/ : SUCCESS in 5m 19s\n- cross-horizon-py27 http://logs.openstack.org/16/663516/1/gate/cross-horizon-py27/8849779/ : SUCCESS in 11m 52s\n- cross-horizon-py35 http://logs.openstack.org/16/663516/1/gate/cross-horizon-py35/2ebdbc9/ : SUCCESS in 12m 30s\n- cross-keystone-py27 http://logs.openstack.org/16/663516/1/gate/cross-keystone-py27/ab01e62/ : SUCCESS in 12m 32s\n- cross-keystone-py35 http://logs.openstack.org/16/663516/1/gate/cross-keystone-py35/e121fa4/ : SUCCESS in 11m 57s\n- cross-neutron-py27 http://logs.openstack.org/16/663516/1/gate/cross-neutron-py27/0734c0d/ : SUCCESS in 24m 49s\n- cross-neutron-py35 http://logs.openstack.org/16/663516/1/gate/cross-neutron-py35/0b45ac4/ : SUCCESS in 27m 28s\n- cross-nova-py27 http://logs.openstack.org/16/663516/1/gate/cross-nova-py27/d8aa9f5/ : SUCCESS in 13m 54s\n- cross-nova-py35 http://logs.openstack.org/16/663516/1/gate/cross-nova-py35/ddbebdd/ : SUCCESS in 14m 17s\n- cross-nova-functional http://logs.openstack.org/16/663516/1/gate/cross-nova-functional/4a6bfbb/ : SUCCESS in 19m 07s\n- cross-swift-py27 http://logs.openstack.org/16/663516/1/gate/cross-swift-py27/2c529e7/ : SUCCESS in 6m 31s\n- cross-swift-py35 http://logs.openstack.org/16/663516/1/gate/cross-swift-py35/26336ce/ : SUCCESS in 5m 12s\n- requirements-integration http://logs.openstack.org/16/663516/1/gate/requirements-integration/f2f0e5a/ : SUCCESS in 55m 06s\n- neutron-grenade-multinode http://logs.openstack.org/16/663516/1/gate/neutron-grenade-multinode/3011b6b/ : SUCCESS in 1h 03m 33s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/16/663516/1/gate/legacy-tempest-dsvm-neutron-full/e4b3f53/ : SUCCESS in 1h 25m 50s\n- neutron-grenade http://logs.openstack.org/16/663516/1/gate/neutron-grenade/a5005b9/ : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fdc08f54cf1fc724e827d3e61bfb0467f8bd0475","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:48:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1927144f3cd47da1fd5f86d71cabffb9ac30368e","revisions":{"1927144f3cd47da1fd5f86d71cabffb9ac30368e":{"kind":"REWORK","_number":1,"created":"2019-06-06 06:45:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/663516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/663516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/663516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/663516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/663516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/663516/1"}}},"commit":{"parents":[{"commit":"6f9bb94a7cf2a85b78d860683040e5c4799a0818","subject":"Merge \"update constraint for oslo.messaging to new release 5.35.5\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f9bb94a7cf2a85b78d860683040e5c4799a0818"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-06 06:45:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-06 06:45:27.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.12.3","message":"update constraint for ironic-lib to new release 2.12.3\n\nChange-Id: I2c51fc8ffa45bc31eabbed5b5bf5dc4fdd2a4022\nmeta:version: 2.12.3\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: Ib7b9545c6b529e18e73868b053f129df5e83ec45\nmeta:release:Code-Review+1: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1927144f3cd47da1fd5f86d71cabffb9ac30368e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1927144f3cd47da1fd5f86d71cabffb9ac30368e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
