)]}'
{"id":"openstack%2Frequirements~663740","triplet_id":"openstack%2Frequirements~master~If7b45a4994784aaf3f1f789c729c9d970b72e538","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If7b45a4994784aaf3f1f789c729c9d970b72e538","subject":"update constraint for glance_store to new release 0.29.0","status":"ABANDONED","created":"2019-06-06 18:22:02.000000000","updated":"2019-06-08 00:09:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c0dd3ffd66f476a9e7d3fd4f1051195472af50b","_number":663740,"virtual_id_number":663740,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2019-06-07 06:31:10.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2019-06-08 00:05:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-06-08 00:09:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 06:31:10.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-06-08 00:05:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-08 00:09:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"42a16d763e560155a1783b99981cb4278ac50641","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-06 18:22:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f85484455c2eb42354d3597de07c2b20c16cf7e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:35:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/40/663740/1/check/openstack-tox-docs/88eac83/html/ : SUCCESS in 4m 27s\n- openstack-tox-pypy http://logs.openstack.org/40/663740/1/check/openstack-tox-pypy/f5e3775/ : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/663740/1/check/openstack-tox-pep8/4cc9d8b/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/40/663740/1/check/openstack-tox-py27/893ae53/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/40/663740/1/check/openstack-tox-py36/de1ea26/ : SUCCESS in 3m 23s\n- openstack-tox-validate http://logs.openstack.org/40/663740/1/check/openstack-tox-validate/19b661a/ : SUCCESS in 3m 37s\n- requirements-tox-babel http://logs.openstack.org/40/663740/1/check/requirements-tox-babel/6e0569c/ : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/663740/1/check/requirements-tox-py27-check-uc/5f854d6/ : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc http://logs.openstack.org/40/663740/1/check/requirements-tox-py36-check-uc/a4e6a47/ : SUCCESS in 6m 22s\n- cross-cinder-py27 http://logs.openstack.org/40/663740/1/check/cross-cinder-py27/570434f/ : SUCCESS in 11m 33s\n- cross-cinder-py36 http://logs.openstack.org/40/663740/1/check/cross-cinder-py36/0d6f0cc/ : SUCCESS in 8m 55s\n- cross-glance-py27 http://logs.openstack.org/40/663740/1/check/cross-glance-py27/50a9a9c/ : FAILURE in 4m 29s\n- cross-glance-py36 http://logs.openstack.org/40/663740/1/check/cross-glance-py36/e42aad1/ : FAILURE in 4m 29s\n- cross-horizon-py27 http://logs.openstack.org/40/663740/1/check/cross-horizon-py27/20ea9b6/ : SUCCESS in 10m 46s\n- cross-horizon-py36 http://logs.openstack.org/40/663740/1/check/cross-horizon-py36/e581a52/ : SUCCESS in 9m 21s\n- cross-horizon-npm http://logs.openstack.org/40/663740/1/check/cross-horizon-npm/7572f22/ : SUCCESS in 5m 44s\n- cross-keystone-py27 http://logs.openstack.org/40/663740/1/check/cross-keystone-py27/7f95c70/ : SUCCESS in 30m 08s\n- cross-keystone-py36 http://logs.openstack.org/40/663740/1/check/cross-keystone-py36/3a19b0f/ : SUCCESS in 25m 21s\n- cross-neutron-py27 http://logs.openstack.org/40/663740/1/check/cross-neutron-py27/2cfaef2/ : SUCCESS in 23m 44s\n- cross-neutron-py36 http://logs.openstack.org/40/663740/1/check/cross-neutron-py36/568b948/ : SUCCESS in 22m 45s\n- cross-nova-py27 http://logs.openstack.org/40/663740/1/check/cross-nova-py27/556d765/ : SUCCESS in 13m 04s\n- cross-nova-py36 http://logs.openstack.org/40/663740/1/check/cross-nova-py36/13fa673/ : SUCCESS in 11m 41s\n- cross-nova-functional http://logs.openstack.org/40/663740/1/check/cross-nova-functional/ac602d2/ : SUCCESS in 16m 00s\n- cross-osc-tox-docs http://logs.openstack.org/40/663740/1/check/cross-osc-tox-docs/52efec0/html/ : SUCCESS in 5m 04s\n- cross-swift-py27 http://logs.openstack.org/40/663740/1/check/cross-swift-py27/9edf7aa/ : SUCCESS in 13m 28s\n- cross-swift-py36 http://logs.openstack.org/40/663740/1/check/cross-swift-py36/3665f64/ : SUCCESS in 12m 06s\n- requirements-integration http://logs.openstack.org/40/663740/1/check/requirements-integration/cbf8317/ : SUCCESS in 53m 19s\n- neutron-grenade-multinode http://logs.openstack.org/40/663740/1/check/neutron-grenade-multinode/406b83d/ : SUCCESS in 1h 16m 09s\n- tempest-full http://logs.openstack.org/40/663740/1/check/tempest-full/10a29f3/ : SUCCESS in 1h 41m 08s\n- neutron-grenade http://logs.openstack.org/40/663740/1/check/neutron-grenade/c4a94ec/ : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cfee5c69495eb804bd46766bea8f51447390b0e8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-07 05:01:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa229a6fbab1f9df8d56b46fa87e94a2605c4974","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 06:07:26.000000000","message":"Patch Set 1:\n\nReported bug https://bugs.launchpad.net/glance/+bug/1831963","accounts_in_message":[],"_revision_number":1},{"id":"68f1b9258e51d79cfbe55c033e727a8f70dd1b83","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 06:31:10.000000000","message":"Patch Set 1:\n\nNeed to hold until https://review.opendev.org/#/c/663840/ merges.","accounts_in_message":[],"_revision_number":1},{"id":"6da0411891e0e901a0b410ad5eef5268c6165332","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 06:43:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/40/663740/1/check/openstack-tox-docs/9521ea0/html/ : SUCCESS in 3m 01s\n- openstack-tox-pypy http://logs.openstack.org/40/663740/1/check/openstack-tox-pypy/54da372/ : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/663740/1/check/openstack-tox-pep8/3bfdfec/ : SUCCESS in 2m 32s\n- openstack-tox-py27 http://logs.openstack.org/40/663740/1/check/openstack-tox-py27/5bd7d15/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/40/663740/1/check/openstack-tox-py36/76d3706/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/40/663740/1/check/openstack-tox-validate/bd30566/ : SUCCESS in 3m 00s\n- requirements-tox-babel http://logs.openstack.org/40/663740/1/check/requirements-tox-babel/d73d182/ : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/663740/1/check/requirements-tox-py27-check-uc/d26e6bc/ : SUCCESS in 6m 48s\n- requirements-tox-py36-check-uc http://logs.openstack.org/40/663740/1/check/requirements-tox-py36-check-uc/6c26b3f/ : SUCCESS in 9m 43s\n- cross-cinder-py27 http://logs.openstack.org/40/663740/1/check/cross-cinder-py27/bd7e754/ : SUCCESS in 14m 14s\n- cross-cinder-py36 http://logs.openstack.org/40/663740/1/check/cross-cinder-py36/e5e746b/ : SUCCESS in 13m 20s\n- cross-glance-py27 http://logs.openstack.org/40/663740/1/check/cross-glance-py27/b74d9af/ : FAILURE in 6m 08s\n- cross-glance-py36 http://logs.openstack.org/40/663740/1/check/cross-glance-py36/245cfa8/ : FAILURE in 3m 57s\n- cross-horizon-py27 http://logs.openstack.org/40/663740/1/check/cross-horizon-py27/99ba2f8/ : SUCCESS in 9m 29s\n- cross-horizon-py36 http://logs.openstack.org/40/663740/1/check/cross-horizon-py36/157d62b/ : SUCCESS in 10m 00s\n- cross-horizon-npm http://logs.openstack.org/40/663740/1/check/cross-horizon-npm/2173d41/ : SUCCESS in 5m 41s\n- cross-keystone-py27 http://logs.openstack.org/40/663740/1/check/cross-keystone-py27/8f7c371/ : SUCCESS in 20m 50s\n- cross-keystone-py36 http://logs.openstack.org/40/663740/1/check/cross-keystone-py36/35505e9/ : SUCCESS in 22m 48s\n- cross-neutron-py27 http://logs.openstack.org/40/663740/1/check/cross-neutron-py27/6cd9a84/ : SUCCESS in 24m 11s\n- cross-neutron-py36 http://logs.openstack.org/40/663740/1/check/cross-neutron-py36/83b73cf/ : SUCCESS in 22m 15s\n- cross-nova-py27 http://logs.openstack.org/40/663740/1/check/cross-nova-py27/a34951e/ : SUCCESS in 15m 20s\n- cross-nova-py36 http://logs.openstack.org/40/663740/1/check/cross-nova-py36/8e244e5/ : SUCCESS in 14m 22s\n- cross-nova-functional http://logs.openstack.org/40/663740/1/check/cross-nova-functional/3300573/ : SUCCESS in 21m 14s\n- cross-osc-tox-docs http://logs.openstack.org/40/663740/1/check/cross-osc-tox-docs/0b8fb7c/html/ : SUCCESS in 4m 48s\n- cross-swift-py27 http://logs.openstack.org/40/663740/1/check/cross-swift-py27/1d5391e/ : SUCCESS in 7m 42s\n- cross-swift-py36 http://logs.openstack.org/40/663740/1/check/cross-swift-py36/1bf733f/ : SUCCESS in 7m 58s\n- requirements-integration http://logs.openstack.org/40/663740/1/check/requirements-integration/7b619d1/ : SUCCESS in 55m 43s\n- neutron-grenade-multinode http://logs.openstack.org/40/663740/1/check/neutron-grenade-multinode/787f4e7/ : SUCCESS in 1h 14m 18s\n- tempest-full http://logs.openstack.org/40/663740/1/check/tempest-full/1660c09/ : SUCCESS in 1h 35m 49s\n- neutron-grenade http://logs.openstack.org/40/663740/1/check/neutron-grenade/b0153b0/ : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":1},{"id":"291e1bcb861d9e7dfa48d0b06c12dec0dc5915b4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-07 14:08:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae51b554c1177732760cf6994bac3685891420fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:05:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/40/663740/1/check/openstack-tox-docs/836b23c/html/ : SUCCESS in 3m 03s\n- openstack-tox-pypy http://logs.openstack.org/40/663740/1/check/openstack-tox-pypy/79623e8/ : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/663740/1/check/openstack-tox-pep8/8456a20/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/40/663740/1/check/openstack-tox-py27/4784e09/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/40/663740/1/check/openstack-tox-py36/0b38e66/ : SUCCESS in 3m 14s\n- openstack-tox-validate http://logs.openstack.org/40/663740/1/check/openstack-tox-validate/dfc593a/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/40/663740/1/check/requirements-tox-babel/6ba7a15/ : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/663740/1/check/requirements-tox-py27-check-uc/2e8d81c/ : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/40/663740/1/check/requirements-tox-py36-check-uc/9e012c0/ : SUCCESS in 6m 07s\n- cross-cinder-py27 http://logs.openstack.org/40/663740/1/check/cross-cinder-py27/a6911f5/ : SUCCESS in 12m 35s\n- cross-cinder-py36 http://logs.openstack.org/40/663740/1/check/cross-cinder-py36/0f10e0a/ : SUCCESS in 8m 07s\n- cross-glance-py27 http://logs.openstack.org/40/663740/1/check/cross-glance-py27/e1d9b79/ : SUCCESS in 4m 31s\n- cross-glance-py36 http://logs.openstack.org/40/663740/1/check/cross-glance-py36/e7e75b6/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/40/663740/1/check/cross-horizon-py27/488b7a3/ : SUCCESS in 12m 32s\n- cross-horizon-py36 http://logs.openstack.org/40/663740/1/check/cross-horizon-py36/6501bff/ : SUCCESS in 10m 06s\n- cross-horizon-npm http://logs.openstack.org/40/663740/1/check/cross-horizon-npm/8229a05/ : SUCCESS in 10m 24s\n- cross-keystone-py27 http://logs.openstack.org/40/663740/1/check/cross-keystone-py27/c774ffd/ : SUCCESS in 23m 24s\n- cross-keystone-py36 http://logs.openstack.org/40/663740/1/check/cross-keystone-py36/e87b7d9/ : SUCCESS in 27m 43s\n- cross-neutron-py27 http://logs.openstack.org/40/663740/1/check/cross-neutron-py27/4bee35a/ : SUCCESS in 25m 38s\n- cross-neutron-py36 http://logs.openstack.org/40/663740/1/check/cross-neutron-py36/f6ae964/ : SUCCESS in 19m 20s\n- cross-nova-py27 http://logs.openstack.org/40/663740/1/check/cross-nova-py27/f298fa2/ : SUCCESS in 28m 08s\n- cross-nova-py36 http://logs.openstack.org/40/663740/1/check/cross-nova-py36/8157498/ : SUCCESS in 11m 55s\n- cross-nova-functional http://logs.openstack.org/40/663740/1/check/cross-nova-functional/ba5354e/ : SUCCESS in 17m 49s\n- cross-osc-tox-docs http://logs.openstack.org/40/663740/1/check/cross-osc-tox-docs/759cac0/html/ : SUCCESS in 5m 30s\n- cross-swift-py27 http://logs.openstack.org/40/663740/1/check/cross-swift-py27/499e833/ : SUCCESS in 8m 22s\n- cross-swift-py36 http://logs.openstack.org/40/663740/1/check/cross-swift-py36/d09f5ba/ : SUCCESS in 7m 40s\n- requirements-integration http://logs.openstack.org/40/663740/1/check/requirements-integration/e58b1d3/ : SUCCESS in 52m 09s\n- neutron-grenade-multinode http://logs.openstack.org/40/663740/1/check/neutron-grenade-multinode/ac08541/ : SUCCESS in 1h 07m 40s\n- tempest-full http://logs.openstack.org/40/663740/1/check/tempest-full/4132679/ : SUCCESS in 1h 54m 01s\n- neutron-grenade http://logs.openstack.org/40/663740/1/check/neutron-grenade/c1e52eb/ : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8a36e546e178801b72ea9b4fe3f9b8c693e60d38","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-08 00:09:36.000000000","message":"Patch Set 1: Workflow-1\n\nknown bad release","accounts_in_message":[],"_revision_number":1},{"id":"8a19721e59ed2aa14d3f931d4ce7a2a20d02506c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-08 00:09:39.000000000","message":"Abandoned\n\nknown bad release","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b06b3fffe468d653b31502f2cab40b5d502087f","revisions":{"5b06b3fffe468d653b31502f2cab40b5d502087f":{"kind":"REWORK","_number":1,"created":"2019-06-06 18:22:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/663740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/663740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/663740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/663740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/663740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/663740/1"}}},"commit":{"parents":[{"commit":"dba4e52baed5f7f8fee9e65170c8f0b57e0ec6c1","subject":"Merge \"update constraint for openstacksdk to new release 0.30.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dba4e52baed5f7f8fee9e65170c8f0b57e0ec6c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-06 18:21:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-06 18:21:59.000000000","tz":0},"subject":"update constraint for glance_store to new release 0.29.0","message":"update constraint for glance_store to new release 0.29.0\n\nChange-Id: If7b45a4994784aaf3f1f789c729c9d970b72e538\nmeta:version: 0.29.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Change-Id: Ie7a7f85bdc0133146d132e70c47f4b5298121928\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b06b3fffe468d653b31502f2cab40b5d502087f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b06b3fffe468d653b31502f2cab40b5d502087f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
