)]}'
{"id":"openstack%2Frequirements~664052","triplet_id":"openstack%2Frequirements~master~Ib6816ab78d58910e280d1be5ff8ab6b985e5872a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib6816ab78d58910e280d1be5ff8ab6b985e5872a","subject":"update constraint for tripleo-common to new release 11.0.0","status":"MERGED","created":"2019-06-07 19:59:09.000000000","updated":"2019-06-08 17:46:26.000000000","submitted":"2019-06-08 17:46:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664052-1560015986009-f0d02f45","meta_rev_id":"4e75c0f2d8bce40f49fc94aa3b8bd451a7a42315","_number":664052,"virtual_id_number":664052,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-08 17:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-08 15:37:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-08 15:37:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 15:37:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-08 17:46:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cad8ce4cb120556595f0a1d5e86cefacaef5e2bb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 19:59:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f01922f3fb718f646981868ebab674a81bf813d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 06:59:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/52/664052/1/check/openstack-tox-docs/9cbaf36/html/ : SUCCESS in 3m 22s\n- openstack-tox-pypy http://logs.openstack.org/52/664052/1/check/openstack-tox-pypy/8139a83/ : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/52/664052/1/check/openstack-tox-pep8/e345a21/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/52/664052/1/check/openstack-tox-py27/14ff0bd/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/52/664052/1/check/openstack-tox-py36/b36eb55/ : SUCCESS in 3m 23s\n- openstack-tox-validate http://logs.openstack.org/52/664052/1/check/openstack-tox-validate/8cdee0a/ : SUCCESS in 3m 17s\n- requirements-tox-babel http://logs.openstack.org/52/664052/1/check/requirements-tox-babel/fefbd33/ : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc http://logs.openstack.org/52/664052/1/check/requirements-tox-py27-check-uc/a2774e6/ : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc http://logs.openstack.org/52/664052/1/check/requirements-tox-py36-check-uc/a834c32/ : SUCCESS in 6m 46s\n- cross-cinder-py27 http://logs.openstack.org/52/664052/1/check/cross-cinder-py27/c1a1036/ : SUCCESS in 10m 23s\n- cross-cinder-py36 http://logs.openstack.org/52/664052/1/check/cross-cinder-py36/3c8dd8e/ : SUCCESS in 8m 27s\n- cross-glance-py27 http://logs.openstack.org/52/664052/1/check/cross-glance-py27/8bbe6b6/ : SUCCESS in 4m 43s\n- cross-glance-py36 http://logs.openstack.org/52/664052/1/check/cross-glance-py36/ebd31ed/ : SUCCESS in 4m 01s\n- cross-horizon-py27 http://logs.openstack.org/52/664052/1/check/cross-horizon-py27/badd468/ : SUCCESS in 10m 44s\n- cross-horizon-py36 http://logs.openstack.org/52/664052/1/check/cross-horizon-py36/4e7100e/ : SUCCESS in 9m 08s\n- cross-horizon-npm http://logs.openstack.org/52/664052/1/check/cross-horizon-npm/c7c0a3f/ : SUCCESS in 7m 26s\n- cross-keystone-py27 http://logs.openstack.org/52/664052/1/check/cross-keystone-py27/7561e26/ : SUCCESS in 22m 03s\n- cross-keystone-py36 http://logs.openstack.org/52/664052/1/check/cross-keystone-py36/257aca9/ : SUCCESS in 26m 05s\n- cross-neutron-py27 http://logs.openstack.org/52/664052/1/check/cross-neutron-py27/d68b334/ : SUCCESS in 21m 03s\n- cross-neutron-py36 http://logs.openstack.org/52/664052/1/check/cross-neutron-py36/fe01d24/ : SUCCESS in 22m 10s\n- cross-nova-py27 http://logs.openstack.org/52/664052/1/check/cross-nova-py27/711c5a7/ : SUCCESS in 13m 26s\n- cross-nova-py36 http://logs.openstack.org/52/664052/1/check/cross-nova-py36/2256392/ : SUCCESS in 11m 31s\n- cross-nova-functional http://logs.openstack.org/52/664052/1/check/cross-nova-functional/b36d876/ : SUCCESS in 18m 41s\n- cross-osc-tox-docs http://logs.openstack.org/52/664052/1/check/cross-osc-tox-docs/3e9913d/html/ : SUCCESS in 5m 16s\n- cross-swift-py27 http://logs.openstack.org/52/664052/1/check/cross-swift-py27/9c43576/ : SUCCESS in 8m 03s\n- cross-swift-py36 http://logs.openstack.org/52/664052/1/check/cross-swift-py36/a416711/ : SUCCESS in 13m 01s\n- requirements-integration http://logs.openstack.org/52/664052/1/check/requirements-integration/cd2f49b/ : SUCCESS in 1h 00m 14s\n- neutron-grenade-multinode http://logs.openstack.org/52/664052/1/check/neutron-grenade-multinode/54bdb32/ : SUCCESS in 1h 18m 22s\n- tempest-full http://logs.openstack.org/52/664052/1/check/tempest-full/6156306/ : SUCCESS in 1h 43m 16s\n- neutron-grenade http://logs.openstack.org/52/664052/1/check/neutron-grenade/56ddef9/ : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f5178a06c86d5491c061db80ef607ae6cd824480","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-08 15:37:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35a0b022827c8193554a1d33d5bd913377a1764d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 15:37:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43f7c2f344c418cb215c4ce35c085c621f1e0ee1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:46:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/52/664052/1/gate/openstack-tox-docs/deed604/html/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 http://logs.openstack.org/52/664052/1/gate/openstack-tox-pep8/dc7c75c/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/52/664052/1/gate/openstack-tox-py27/824bee5/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/52/664052/1/gate/openstack-tox-py36/d020d20/ : SUCCESS in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/52/664052/1/gate/openstack-tox-validate/bdc5014/ : SUCCESS in 3m 14s\n- requirements-tox-babel http://logs.openstack.org/52/664052/1/gate/requirements-tox-babel/fdbca0e/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/52/664052/1/gate/requirements-tox-py27-check-uc/68ef674/ : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc http://logs.openstack.org/52/664052/1/gate/requirements-tox-py36-check-uc/769a78e/ : SUCCESS in 6m 48s\n- cross-cinder-py27 http://logs.openstack.org/52/664052/1/gate/cross-cinder-py27/168b338/ : SUCCESS in 10m 41s\n- cross-cinder-py36 http://logs.openstack.org/52/664052/1/gate/cross-cinder-py36/e00bd96/ : SUCCESS in 8m 43s\n- cross-glance-py27 http://logs.openstack.org/52/664052/1/gate/cross-glance-py27/27bc250/ : SUCCESS in 4m 18s\n- cross-glance-py36 http://logs.openstack.org/52/664052/1/gate/cross-glance-py36/315b492/ : SUCCESS in 4m 18s\n- cross-horizon-py27 http://logs.openstack.org/52/664052/1/gate/cross-horizon-py27/b070415/ : SUCCESS in 10m 56s\n- cross-horizon-py36 http://logs.openstack.org/52/664052/1/gate/cross-horizon-py36/ebb425f/ : SUCCESS in 8m 48s\n- cross-horizon-npm http://logs.openstack.org/52/664052/1/gate/cross-horizon-npm/de01c7c/ : SUCCESS in 6m 44s\n- cross-keystone-py27 http://logs.openstack.org/52/664052/1/gate/cross-keystone-py27/6b02be2/ : SUCCESS in 21m 19s\n- cross-keystone-py36 http://logs.openstack.org/52/664052/1/gate/cross-keystone-py36/5b4a498/ : SUCCESS in 21m 16s\n- cross-neutron-py27 http://logs.openstack.org/52/664052/1/gate/cross-neutron-py27/0e1dfd5/ : SUCCESS in 23m 48s\n- cross-neutron-py36 http://logs.openstack.org/52/664052/1/gate/cross-neutron-py36/0a50c6e/ : SUCCESS in 22m 22s\n- cross-nova-py27 http://logs.openstack.org/52/664052/1/gate/cross-nova-py27/a0a87e8/ : SUCCESS in 13m 32s\n- cross-nova-py36 http://logs.openstack.org/52/664052/1/gate/cross-nova-py36/53ec88e/ : SUCCESS in 13m 15s\n- cross-nova-functional http://logs.openstack.org/52/664052/1/gate/cross-nova-functional/d8c837b/ : SUCCESS in 18m 43s\n- cross-osc-tox-docs http://logs.openstack.org/52/664052/1/gate/cross-osc-tox-docs/589fd34/html/ : SUCCESS in 5m 43s\n- cross-swift-py27 http://logs.openstack.org/52/664052/1/gate/cross-swift-py27/f0f9823/ : SUCCESS in 8m 03s\n- cross-swift-py36 http://logs.openstack.org/52/664052/1/gate/cross-swift-py36/a1be7bd/ : SUCCESS in 9m 33s\n- requirements-integration http://logs.openstack.org/52/664052/1/gate/requirements-integration/e0c807a/ : SUCCESS in 47m 43s\n- neutron-grenade-multinode http://logs.openstack.org/52/664052/1/gate/neutron-grenade-multinode/cbe4521/ : SUCCESS in 1h 11m 41s\n- tempest-full http://logs.openstack.org/52/664052/1/gate/tempest-full/9df705f/ : SUCCESS in 1h 47m 44s\n- neutron-grenade http://logs.openstack.org/52/664052/1/gate/neutron-grenade/341b507/ : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":1},{"id":"bd55bc4844eaaf7e2d9064ff3f39c83b2821c547","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:46:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55d666487c35cada80f2c410724a917c301ac0c0","revisions":{"55d666487c35cada80f2c410724a917c301ac0c0":{"kind":"REWORK","_number":1,"created":"2019-06-07 19:59:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/664052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/664052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/664052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/664052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/664052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/664052/1"}}},"commit":{"parents":[{"commit":"0cb1facc8f7d94e19ec829f5150908a0d0ced6ea","subject":"update constraint for os-win to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cb1facc8f7d94e19ec829f5150908a0d0ced6ea"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 19:59:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 19:59:07.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 11.0.0","message":"update constraint for tripleo-common to new release 11.0.0\n\nChange-Id: Ib6816ab78d58910e280d1be5ff8ab6b985e5872a\nmeta:version: 11.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: If8a2c2d5160f2d75ae2d67bf662f8735851b99c0\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55d666487c35cada80f2c410724a917c301ac0c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55d666487c35cada80f2c410724a917c301ac0c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
